)]}'
{"id":"openstack%2Fglance~719023","triplet_id":"openstack%2Fglance~master~I1c2c14c3e1997b1adbc28022ea03fd541554bdb6","project":"openstack/glance","branch":"master","topic":"M3/config-refresh","hashtags":[],"change_id":"I1c2c14c3e1997b1adbc28022ea03fd541554bdb6","subject":"Refresh Glance example configs for Ussuri milestone 3","status":"MERGED","created":"2020-04-10 14:16:26.000000000","updated":"2020-04-13 19:48:09.000000000","submitted":"2020-04-13 19:46:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719023-1586807180656-170bf913","meta_rev_id":"1149b4571139eecebcd94b868fca02b446682710","_number":719023,"virtual_id_number":719023,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-04-13 17:03:05.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 19:46:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-04-10 14:18:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-04-13 13:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-04-13 13:48:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-10 14:17:32.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-04-10 14:17:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-10 14:17:32.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-04-10 14:18:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-13 13:48:33.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-04-13 17:03:05.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-13 19:46:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8395785b654f427cafaad7fb44259b3fc4e0c51a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-10 14:16:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99b1180adafa0e1463d2dcc28faf3d3c8655f56b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-10 14:17:05.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"f629baeda5b07d46cee55f81cc0443179e797220","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-10 14:17:19.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"f7bb843e726ba68d401d37bb03d5067a8a2fdf60","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 14:18:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da3822ac78c8c6b997b866f5f11f319da7264c59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 15:43:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4aca83997a6342b59bba7ef390347d75 : SUCCESS in 1h 10m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d740c53cc4d445bc88fede3a502a3071 : SUCCESS in 1h 19m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27adb3ba279d4c50ba3a056313f77f09 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/502d88fbf9b54bbda5a1de88ebb4aac9 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f076101bd8f451fa220a4e31486dcf5 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b67033de2574ed98ba009e7ee7e1e34 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4fa2763f895483cabc025134e42fd7f : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080b1229af1249629d74e7849b6acb29 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ccd92b32721d42528036a890bd970513 : TIMED_OUT in 31m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b80f0609e2c0431895ccb87e765b42cf : SUCCESS in 8m 04s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3721de716d484713ab18057a4fd6a9ce : SUCCESS in 55m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d89ba70ea00f470ba03492c651a39598 : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":1},{"id":"479e875b9be076b54e4112a2ecc02dedb54964df","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-10 16:05:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5eca2a9aeed84426a5a848df3db24eb1049b1b50","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-10 16:05:24.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"5b67d4fbe337ba240e2f3b83306283640842626c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 17:47:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bffdec5945994f04b06f1818aa8cfd06 : SUCCESS in 1h 30m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/50cc5b5f247746838c5dc41f01c747b8 : SUCCESS in 1h 40m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d074c9a2cc242a3b67b4f953cdffca0 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8671f528ecc434dad6638320cf5b406 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/178137c38ef14a708dbcf90a05d1257d : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f5c89c138a864ea994ae43d7749a2a85 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1013d74041134b07ac818b272986a1fc : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb5b7a41d23f43b4b34b1250f7f55550 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/81dd978979124f30a53e7735ec2ba8d5 : SUCCESS in 9m 02s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/96f8a92028f7416aa55419b9df95914b : SUCCESS in 9m 44s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6593b3c5720d443b858dffb51ee87b96 : SUCCESS in 57m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3f489bbf87c4f4286ca9f4f24a7e45d : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":1},{"id":"cc558d72339e8df435e28489f3ecd503d7630423","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-04-13 13:48:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"8d8b10cd59050445d291c4988c66bb9d39e68c14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 13:48:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b9dd65dc5380ad3b2b9700e2ca6ab63150f794f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 15:21:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1986f3960bc64c0ea67197dfaa10780d : SUCCESS in 1h 22m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/751418e0f4ee4d1594e608e6251040f6 : SUCCESS in 1h 13m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0858de5756b47bb9047675c4aa633f6 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35816b132bb1430cabb4d7a056c309d5 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a4f7b11acfc44e9bc91ed980b5817eb : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82365c5331b74e618e35edd22b323551 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/027631a8f05e493cb4e20d76da15ce00 : FAILURE in 8m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7aa55560cec94300956ec50389214050 : SUCCESS in 13m 14s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0eb91cf7c2d64fb5b7223e5cd95f9067 : SUCCESS in 7m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f7960ae03f44a25b1dea2b6cb926070 : SUCCESS in 59m 07s\n\nWarning:\n  Comments left for invalid file doc/source/glossary.rst","accounts_in_message":[],"_revision_number":1},{"id":"cd6fbeda68d41f76c996d2cbff146ffb0adccc35","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-13 17:02:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45f50990863c448d33fef63484f48a09b192a457","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-13 17:03:05.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"ff0b0b8015333b301ee9e74a7403b742f08f822e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 18:24:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fdd49c0db83a49a39e984c68c42537d9 : SUCCESS in 1h 15m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cea942be94784518b067ca02ce48ab31 : SUCCESS in 1h 16m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa25fb2ad09b45f98eb0b620d9e07816 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b785d65b0f2c44d0be187120d6176957 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83d4e700716f4513958dd163ecf5ee9a : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e105dc6c28c14b299e506b7b5e2abf31 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3b6b3e3cf274555b81ec0c5c488be45 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a526a5ea523a4017b8245f7fed313e30 : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97d68026e5dc49a7a2db2e554a130a2d : SUCCESS in 7m 10s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3e5dc852cbb44be3a8722d0377625451 : SUCCESS in 7m 56s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/eeab2f383fad4b5f97fe0de3ca78524e : SUCCESS in 58m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0148636d233a4c33a18f429e5e29fa57 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ff31371e645eb513978a70fbd5df3a3fb4c1c343","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 18:24:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"465af808e6068d23e1931baa55b7576c0f798e84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 19:46:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/89c03dd368e042f8912dc95ba76f41fa : SUCCESS in 1h 10m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/134776d4cea54411a783878ce0acec36 : SUCCESS in 1h 16m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/564b1c59591746948a7a004854592137 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bacbb6456a92482097321e4cc1126767 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7ae2b5562cf48e9a405062b8a743bd7 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/957c2c5c5de340b7bc2aa1947d8ba7f0 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/122e4304898a4fa783abde160c5e3e85 : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa5ceab63fc645e9821b2a3ae04af694 : SUCCESS in 8m 19s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1f2950d3fc14430eaedf06b216fddce2 : SUCCESS in 9m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/813faab506b746118b71f792c73a3c78 : SUCCESS in 1h 07m 55s","accounts_in_message":[],"_revision_number":1},{"id":"11430f241a98c78731244dc101101bb83684d9d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 19:46:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c0ffcd4c65e485da1ca31171e15446f79d506046","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 19:48:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca41e13d5de24e438ed5a36fb0daa8a0 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68e5fa35cbf392a94f21f10361bc09c4002cfc64","revisions":{"68e5fa35cbf392a94f21f10361bc09c4002cfc64":{"kind":"REWORK","_number":1,"created":"2020-04-10 14:16:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/23/719023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/23/719023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/719023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/719023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/719023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/23/719023/1"}}},"commit":{"parents":[{"commit":"28afc9fc6fc49532be6d5dd51eff8b61a9d1b4ca","subject":"Deprecate admin_role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28afc9fc6fc49532be6d5dd51eff8b61a9d1b4ca"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-04-10 14:15:57.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-04-10 14:15:57.000000000","tz":0},"subject":"Refresh Glance example configs for Ussuri milestone 3","message":"Refresh Glance example configs for Ussuri milestone 3\n\nChange-Id: I1c2c14c3e1997b1adbc28022ea03fd541554bdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/68e5fa35cbf392a94f21f10361bc09c4002cfc64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/68e5fa35cbf392a94f21f10361bc09c4002cfc64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
