)]}'
{"id":"openstack%2Fglance~720914","triplet_id":"openstack%2Fglance~master~I44e7b6f76e2d12f620ec602afc77ce11ba6b9d9a","project":"openstack/glance","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"I44e7b6f76e2d12f620ec602afc77ce11ba6b9d9a","subject":"Use unittest.mock instead of third party mock","status":"MERGED","created":"2020-04-18 16:52:53.000000000","updated":"2020-04-21 05:33:32.000000000","submitted":"2020-04-21 05:31:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720914-1587447117655-276f548f","meta_rev_id":"397e9180e5284bf0392a462313f01c5d499de8ba","_number":720914,"virtual_id_number":720914,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-20 15:07:50.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-21 05:31:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-20 17:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-21 05:19:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 15:07:50.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-21 05:19:50.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-04-21 05:31:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1cd368b1c5889e4fa2c4503d85f49858da9ac60a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-18 16:52:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"724f38ec3ec61f4da5c96bf5b41eb2988ec89e30","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-18 16:53:18.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"4b4fe2a88017f7339d34eb0492c0c82c106a8ded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 17:08:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1aefb5aad6d54842952f9b25520d5160 : SUCCESS in 5m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/482cdc3f4a8b47abbb0b6a37348bf3ed : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bec3b5cf7b734d11a0b69ffb97efa9ad : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0601880a6a4f4cfa8567e0c5063a4701 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48ed3d424cd04379b6f3f8745360e657 : SUCCESS in 7m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57f583ef8cc944b08bdafee57111c65c : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3213b8805dbd48ed83068df88c1b3cc2 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f0d645eee194bd68a8b4af99f2afcae : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9fed734dc6f472bbba67fe00e4254b8 : SUCCESS in 9m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b30471233b84fb8b23fa028935daa24 : SUCCESS in 9m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6c2b94640b9a41cb830e039d19849f13 : FAILURE in 10m 15s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/50c3c9f570644169b819a59e01de5ec6 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1eac0932968fd1bcff4e5de688fb2a32312109ac","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-20 15:07:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e2099239edcc8d7493f68a2f13e2117d063f65ff","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-20 15:07:50.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"4400417c4d074c9fc2103faa4f18ec4edb844baa","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-20 17:51:46.000000000","message":"Patch Set 2: Code-Review+2\n\nThank you Sean!!!","accounts_in_message":[],"_revision_number":2},{"id":"0150676d37e9f16a70a73c71d9a2e3876a06172b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 22:46:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d0b3d1043b4c4896a101436e58f5bf2c : SUCCESS in 7m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/80136dd6f77f4b0a85888450d151de0f : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d70fc929dd143c3ad6549663e7976ba : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3900c723c24476be854849ad3ad4f1 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff83534de56c40dcbc61842d42cfcb44 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b439b972394a4ef9b3f68ede5b28ce47 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02a965c8ec194b3b8794d29cc3ba2230 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84059e6fa19646f994ac627d8522f25d : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/442dd0ff544b49bfbc5e858732b3c36e : SUCCESS in 10m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a68ee7f79dca4dffb9cc58a9e248425e : SUCCESS in 19m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8e49168a5cbe46ff9795de09158bd205 : SUCCESS in 10m 40s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ed3051a09c624d65807fb76f5eb2ffe6 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"39131b276a2f4a92f1852c5381526c1be054823d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-21 05:19:50.000000000","message":"Patch Set 2: Workflow+1\n\nSimple change, Ninja approving as just touches tests.\nThank you Sean!!!","accounts_in_message":[],"_revision_number":2},{"id":"f1adc2c8f2fd0a93b5ab5bf8c1b5c698626687eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 05:20:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aabbd2e3aeff3c6fa76966f07c4497f0acae0f68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 05:31:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ead8a9d53f9d517e1830127ad3b1bb6b3d6f1516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 05:31:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/736252200e7843e598cceecf2b22725c : SUCCESS in 4m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f0a53dd357fd4b3799f00749c39a614c : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c817327d98b742b8a09daafa373768b0 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36356582dd154f3fa51f760c727b3a4a : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1bfa66691a24060bb6cf29a4b2d52f4 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dccac8a4b38b48aba436bc91a6e76ed0 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a383c4a5c3ab4c9dab110b801c983f9a : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/625a29c620844faa91ec4124992095dd : SUCCESS in 8m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d2b9a8edeb640c180ff1f1df3739212 : SUCCESS in 10m 07s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9533216227f04fd1b43d5fcc4751af0a : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b61c5e6e6920fc8cbc623e7fc3cf918f9367fc23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 05:33:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c90739abb756412aae23556ab2861639 : SUCCESS in 1m 12s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e19542861d9c4b0c9e7cf1e0e26db94c : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8f63137139442799497eb2d2d39ef38 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"94b0876429a4234738a9481a749d2dd5458be298","revisions":{"3743a7ac8aa42ce19683e31b6c5f6a93e9309a46":{"kind":"REWORK","_number":1,"created":"2020-04-18 16:52:53.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/14/720914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/720914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/720914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/720914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/720914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/720914/1"}}},"commit":{"parents":[{"commit":"bd16ab256e6bf6b423f3bdbd366ba0b56d5bc619","subject":"Add warning and note on image schema customization docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd16ab256e6bf6b423f3bdbd366ba0b56d5bc619"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:52:42.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:52:42.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I44e7b6f76e2d12f620ec602afc77ce11ba6b9d9a\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3743a7ac8aa42ce19683e31b6c5f6a93e9309a46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3743a7ac8aa42ce19683e31b6c5f6a93e9309a46"}]},"branch":"refs/heads/master"},"94b0876429a4234738a9481a749d2dd5458be298":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-20 15:07:00.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/14/720914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/720914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/720914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/720914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/720914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/720914/2"}}},"commit":{"parents":[{"commit":"7af0b68b79281248e9f62e9444413989fa988bf7","subject":"Skip \u0027test_image_member_lifecycle_for_multiple_stores\u0027 on failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7af0b68b79281248e9f62e9444413989fa988bf7"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-18 16:52:42.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-20 15:07:00.000000000","tz":0},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: I44e7b6f76e2d12f620ec602afc77ce11ba6b9d9a\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94b0876429a4234738a9481a749d2dd5458be298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94b0876429a4234738a9481a749d2dd5458be298"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
