)]}'
{"id":"openstack%2Fglance~721155","triplet_id":"openstack%2Fglance~master~Iad92971425c383010c0f7df0a4c32c327d4ea336","project":"openstack/glance","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"Iad92971425c383010c0f7df0a4c32c327d4ea336","subject":"Use unittest.mock instead of third party mock","status":"ABANDONED","created":"2020-04-20 03:04:37.000000000","updated":"2020-04-20 06:21:55.000000000","total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"58e887cdb384c831fa0157c917b3325b7556fa59","_number":721155,"virtual_id_number":721155,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-20 03:04:58.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-20 03:15:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":-2,"date":"2020-04-20 06:15:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 03:04:58.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-20 03:15:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-20 06:15:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"5bbd5a70755d675be2ee8c4c336d16e241a9bd84","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 03:04:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a42c8261d901cc9e568b5c81ca3917ad7ce998aa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-20 03:04:58.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"94d6d599e6417785bdde68b0ab9c2af2ee8c7171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 03:15:47.000000000","message":"Patch Set 1: Verified-1\n\n(28 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b08b2fcd373c45428f79290437e4d7fc : SUCCESS in 6m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f5a1d747d97848189863231334fb9074 : SUCCESS in 4m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d9d71e007e346b5a09227c74928d311 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a412e8dcb73a4f64a4dd109b473ad2be : FAILURE in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/127cce192d8243b99c153d7387475117 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4322702ba71a44c4853cff712ed4ddd8 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc092274482346a884deda016259d66c : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bec51a0090941b5923912af01272584 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e16bd8e72ba94247aa56246bde07d23c : SUCCESS in 9m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3a137b43daea4c11abd3715d4c12c30e : SUCCESS in 7m 20s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1f77494fd42e4fd6b60cd2d61c480d51 : FAILURE in 9m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ce2b2643dda14ab79c245aa5064338a8 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d2beb64c4fa25ef494955ecafbed25c74ff7271d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-20 06:15:45.000000000","message":"Patch Set 1: Code-Review-2\n\nDuplicate of https://review.opendev.org/#/c/720914/\n\nThank you for your efforts though!!!","accounts_in_message":[],"_revision_number":1},{"id":"519787308b4a8c53f1ea53aa8a2a89ea83c1fab2","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-20 06:21:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"108cefdb7b6643efaa2a27ce544521db7e8c3d53","revisions":{"108cefdb7b6643efaa2a27ce544521db7e8c3d53":{"kind":"REWORK","_number":1,"created":"2020-04-20 03:04:37.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/55/721155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/721155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/721155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/721155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/721155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/721155/1"}}},"commit":{"parents":[{"commit":"bd16ab256e6bf6b423f3bdbd366ba0b56d5bc619","subject":"Add warning and note on image schema customization docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd16ab256e6bf6b423f3bdbd366ba0b56d5bc619"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 02:47:57.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-20 03:04:34.000000000","tz":480},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: Iad92971425c383010c0f7df0a4c32c327d4ea336\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/108cefdb7b6643efaa2a27ce544521db7e8c3d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/108cefdb7b6643efaa2a27ce544521db7e8c3d53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
