)]}'
{"id":"openstack%2Fglance~722695","triplet_id":"openstack%2Fglance~master~I961af5693debacf7e29021a96e2c2b7d510da991","project":"openstack/glance","branch":"master","topic":"add-py38-meta","hashtags":[],"change_id":"I961af5693debacf7e29021a96e2c2b7d510da991","subject":"Add py38 package metadata","status":"MERGED","created":"2020-04-24 13:24:47.000000000","updated":"2020-04-27 08:30:29.000000000","submitted":"2020-04-27 08:28:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722695-1587976106636-8eb4b95b","meta_rev_id":"e308e7356ce6dadf04b36ef077b7e2671c66d84f","_number":722695,"virtual_id_number":722695,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-24 13:25:20.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 08:28:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-27 06:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-27 06:51:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 13:25:20.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-27 06:51:56.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-04-27 08:28:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c352ee755d8d53efbc75138f574555ccc737369a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 13:24:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a2407399f5d48a2430110e7d389d61f412b8efa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-24 13:25:20.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"b3793e88265ea7ff39e7c966392efbd07ca8633f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:52:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e9ad8b5767ce4681bb2c152aad30d655 : SUCCESS in 4m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/45e83c2ce3c04f19a48e4bd706994867 : SUCCESS in 1h 08m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea02939b2976433985a0e9a4468d5284 : SUCCESS in 1h 14m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b309761f12c432e984ad686ec720242 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a4717673a1f49b58dfee3f593fa7ec8 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f249a94acb2242d58a59afbe2c91474e : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deedc88affa04668a65dab49e0d23d4d : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/584f45b585f14f9f945b54a3233d1a74 : SUCCESS in 9m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6c050413c74d4fdca4fc80435fbdb2d9 : SUCCESS in 11m 37s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d2616d9c63a34920b888c8d42f92fe86 : SUCCESS in 10m 45s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2619c66b5e73465383b4c43a4760648b : SUCCESS in 1h 04m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f8ae5cd66c94eaf8b76a949305344ec : SUCCESS in 58m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e72e01dd2d52f49cd2550a1d0b03ad1fedb913ff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-04-27 06:51:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87f046134f14894a30a29f7d6012c77de9585407","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 06:52:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f4e4eb1973c42b32645542f7559bf06b58d300f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:28:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e659e7495d44aaf91b9847f5a968db7 : SUCCESS in 5m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4e6fb0f8d0f247229fb044413fb93f50 : SUCCESS in 1h 32m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/207ff3a84df74d6fb23c699a92fc6102 : SUCCESS in 1h 28m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f756ac4936a4c628639138ab471c26f : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe29fbdcf8724a80889d5d41fd85f8dc : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a98eb91491d148fe8ce78d86b277ee19 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93cb9c27177b4730998741b4e092fd30 : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b35794145674f899053047c7acb05ce : SUCCESS in 9m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ed50d03d131f4b08be43f52b8970afbb : SUCCESS in 10m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3cc9427640d5440ba3206f67f173efa3 : SUCCESS in 12m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47f8a6558adc478d8bbd2b12f939fdda : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":1},{"id":"2df90a1e8088eb22a6a19d65358720aa8a1b4f90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:28:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b5bef2792f74d894d33852ba829acc7642d621b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 08:30:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59fc7cb73e23402c907aed470114b741 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4505f5d69bf9f189868f4fceb67093302c37dcdd","revisions":{"4505f5d69bf9f189868f4fceb67093302c37dcdd":{"kind":"REWORK","_number":1,"created":"2020-04-24 13:24:47.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/95/722695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/722695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/722695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/722695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/722695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/722695/1"}}},"commit":{"parents":[{"commit":"f0d77a4e074a8862ca72071a92b04a1389aa179c","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0d77a4e074a8862ca72071a92b04a1389aa179c"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:22:34.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:22:34.000000000","tz":-300},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nNow that we are running the Victoria tests that include a\nvoting py38, we can now add the Python 3.8 metadata to the\npackage information to reflect that support.\n\nChange-Id: I961af5693debacf7e29021a96e2c2b7d510da991\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4505f5d69bf9f189868f4fceb67093302c37dcdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4505f5d69bf9f189868f4fceb67093302c37dcdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
