)]}'
{"id":"openstack%2Fglance~724957","triplet_id":"openstack%2Fglance~master~I53454c05710ebc45381315a20e10e627cbb65650","project":"openstack/glance","branch":"master","topic":"bug/1876419","hashtags":[],"change_id":"I53454c05710ebc45381315a20e10e627cbb65650","subject":"Fix a failure to parse json file","status":"MERGED","created":"2020-05-02 05:39:45.000000000","updated":"2020-09-17 08:36:14.000000000","submitted":"2020-05-04 16:26:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724957-1588609571441-d77804d9","meta_rev_id":"6d299b4f9a8f7c93ab2a20eac7faced1568881ee","_number":724957,"virtual_id_number":724957,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-02 05:40:07.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 16:26:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-04 14:31:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-04 14:31:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-02 05:40:07.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-04 14:31:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-05-04 16:26:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"885de589d45f605b09adb9d5b0ebabae001cbed8","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2020-05-02 05:39:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10708710e10bbcf0d00a091b58619a8919253c2d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-05-02 05:40:07.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"fbc4fecba785c9f90c9d0cec257657c3ea974261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 06:56:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5506acc175354198bf0920478eb01f6a : SUCCESS in 1h 02m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d6619b5af7f46ad908a79424a676580 : SUCCESS in 1h 12m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ff43c646c8545929aa3629e0f9a1645 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e74e8900d9b4c2eb16fec9539fe5c68 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da1ddd4e65b24771b689d24ca321b584 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e742c84c5f543d5adf332c1e350f8cf : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1adce4c6c7dc48d881903d19fdfc7d9e : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e5c96b0ed0224e369cf01018ba598e1d : SUCCESS in 7m 10s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5b67f6dc62e24fa08474c694e78389c5 : SUCCESS in 8m 40s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/130944528f744d94abaf360da55d61b0 : SUCCESS in 1h 01m 36s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a4bbf087eebd47b88cd5f98a778e8e5f : SUCCESS in 1h 10m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/814cc1f28224466dbaf4e24e8361320e : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a74bbcb8d3babd2572289431000793e6f1957491","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-05-04 14:31:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ba0e0d5b7f01d61849f26db3471844ca1531299","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 14:31:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"775f0a0385bfe481d22fb8f7d79c6348bd683749","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:26:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e059ed431c934d9c2a3c1a6db9cb836259178cc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:26:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce6834d81b274f86907bd55826303cd9 : SUCCESS in 1h 18m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86610cb95bb94c4f8e38ffd53c9218e7 : SUCCESS in 1h 35m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20b6b7bba5804e9cb88e01c313bafcd7 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778587e87bfa45eda579ab4ee4f17f48 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3735ca7135a94925ac0ebcdcd7f91e00 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18cfe507c2944040a0d472304699f5c9 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c822f6c7b44e158515c6d4a9ebe5d3 : SUCCESS in 10m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35e0326691a147698a18462a5b78571a : SUCCESS in 9m 59s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/28ff30713dc74b648e21eeea35866d2b : SUCCESS in 11m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5f9e5920e7ae42d68c1b28838e1e1095 : SUCCESS in 1h 20m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5e551fb3ba74d9e9408ff9fa32b0b99 : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":1},{"id":"95bd745e5196ca2a6973656002c1795f4ab8062d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:28:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4281ab7d4314444b88e4617d5057e9e : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2234dbbe1b9a951e0405937bab57f416f675f500","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2020-09-17 08:36:14.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 977ce4d79cb8e704aa6b5b1d2d983a9509168e8d","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed3bb4e1fb07c2817b4b7ad2e3e069f78bd839ba","revisions":{"ed3bb4e1fb07c2817b4b7ad2e3e069f78bd839ba":{"kind":"REWORK","_number":1,"created":"2020-05-02 05:39:45.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/57/724957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/724957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/724957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/724957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/724957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/724957/1"}}},"commit":{"parents":[{"commit":"10f7130bcf9c66944b6deb0fcebf57bf8e3bf596","subject":"fix typo in gerrit doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/10f7130bcf9c66944b6deb0fcebf57bf8e3bf596"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2020-05-02 05:35:12.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2020-05-02 05:36:42.000000000","tz":540},"subject":"Fix a failure to parse json file","message":"Fix a failure to parse json file\n\nChange-Id: I53454c05710ebc45381315a20e10e627cbb65650\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nCloses-Bug: #1876419\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed3bb4e1fb07c2817b4b7ad2e3e069f78bd839ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed3bb4e1fb07c2817b4b7ad2e3e069f78bd839ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
