)]}'
{"id":"openstack%2Fglance~725188","triplet_id":"openstack%2Fglance~master~I5fe73e09dc12018fc64226fd22d06fedd5b15d42","project":"openstack/glance","branch":"master","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I5fe73e09dc12018fc64226fd22d06fedd5b15d42","subject":"Switch to new grenade job name","status":"MERGED","created":"2020-05-04 06:39:57.000000000","updated":"2020-05-05 06:40:23.000000000","submitted":"2020-05-04 21:47:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725188-1588628879361-e143bcec","meta_rev_id":"0306a61e382faaef94393d74b817e4570cdf5660","_number":725188,"virtual_id_number":725188,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 21:47:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-04 11:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-04 14:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2020-05-04 11:55:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-05-04 11:14:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-04 14:29:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 11:14:31.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-05-04 11:55:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-04 11:56:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-04 14:29:32.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-05-04 21:47:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfbe22f48ab67d3d6ee9cf807a2382d0e66daa93","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-04 06:39:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f459db0537e7bb09077609013c0ae0bf15adb6c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 07:33:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f471571b1794b88ac5b2cb344307023 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbccfc9113c34448aa46efcd80aa4f48 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3495009f002a4bbfa768afccf32f655e : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8f533305a964e32b49dabbfc5b17b4e : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/452e7ac50f334480bbf6c562570cba85 : SUCCESS in 8m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9bf1f764f3ab465f805db14df61cab08 : SUCCESS in 7m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/99d3c10dde8f4a2c9e71087f2db59519 : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1772b39643007bfb7634e8b25740c9b2359be278","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-04 11:14:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f5e05ca9bdb014db752ca1a4ad29721a491b5725","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-04 11:55:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"82a39e0eadebceec339beb39a40996bb395256b6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-04 11:56:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d5cd7863f2eb7d26bf7f195767dd3cb15db70719","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-05-04 14:29:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e622b5412ef93611a0731bb35f174f09dea3efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 14:29:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5206cc8181305806f73c0e8adbc7130005c1d66d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 15:57:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/934a9ba8a87a454c9b16f1060cbb4a11 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ab233b771d441b79a592b6d633e53d0 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98129ca18e9044cda80296227c31b5f1 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/816dbecbcfd6436d89c4b8ef3c4bbfcd : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec008b8b9e9c4e30bf5d517b5c36d04a : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ba347e3214014abda3a3c6db674b6aad : SUCCESS in 10m 01s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ee7f3afb5e4249e9a9f2b10866ed3205 : FAILURE in 11m 58s","accounts_in_message":[],"_revision_number":1},{"id":"55b0e77beb198adaaeebd01f3fb5eee7ddfe7983","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-04 15:58:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b4dcca791223304e545119324e51f701d2bd4704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:10:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f96f724decc4f249bbd6921b1969003 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/776e8e703c534e66a0961fb8d11545f4 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68e2b3b067334d6d9416d5160b037a49 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/335c1ea027554a5897933162ef5e93b1 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d291c2399e64005b123fdb10beac050 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86a04cf2467d42b7a3e8c84cc36ed9c4 : SUCCESS in 7m 17s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/651d33c63f7c481d8689de9e7a1378e8 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a376c9ec86f66b9d3249104231e4c14eb2479104","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 16:11:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5044f4714333852a111854601a49d7676bf5c98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:06:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/252d691550d9440ca10c55ab27ea2759 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe2ec949fc4414dabca6c76a4ea1ec8 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b959e424d74a46a79ed9dd309db73e95 : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/045b8c3e403340e2bd39129df8b79f4b : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d0750149834629833840acf402f3bd : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/320d87e4d685479db2ae24d9c5ff9f57 : FAILURE in 14m 18s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8fed69667b0640d59546bc3817b68dc6 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5f1c038510119b6b9ca7d3982a4b67a65eb4d260","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-04 18:12:07.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nInteresting enough, the grenade job is not executed. I guess the reason is that exclusion rule in the irrelevant-files:","accounts_in_message":[],"_revision_number":1},{"id":"681878235ff78b37f3d6bbefb68a01110e46bb22","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-05-04 18:12:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"674e55fb497af090dec39610c5a2defb5c4c1e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:30:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/789e8c5eb55449d995775646ebfdfa9b : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc8c80d48639430e9e5a331bb1f7a2d7 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8939e16b2914c71af518463646ee5de : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0970584709e4a71a94a0ad38624d944 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d7b004b10244b12afbc038472cafcd4 : SUCCESS in 11m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1dacff3213824442be56630c94393ec9 : SUCCESS in 8m 30s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/68b7462391614800873e11ccb500bf02 : SUCCESS in 15m 55s","accounts_in_message":[],"_revision_number":1},{"id":"df414613b4d001967db93615e3dfb1d6cdd1367e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 18:30:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3027c6600643026ff90978ebd31b5bd491789c8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 21:47:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44d1886f7cae019fa14948427ecda48e8ad34c57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 21:47:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35d58f042109478bb26c163b332075fb : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/047a6c76a2fe48e9b00dbbd1450986db : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edcc22cfb1a645098111919e51d2d83f : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a9fd4020eb34d7eb7245081b1ed2bfc : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57141bfb9811494e86b209eca9e04031 : SUCCESS in 8m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b1896fc0faa0471f9cf334bae65d2c13 : SUCCESS in 12m 26s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f056e90086774e5685c00c38503f47a8 : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9a8186a2904aa3be779bb5e38b93e2508909cc62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 21:54:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18679d1c565e45b4811e513563759aa4 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":1},{"id":"990e798c63ca0e1130e930c973ed96f33459576f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-05 06:40:23.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit f78463ed32f67e551662c77c6bd9beff80a2121f","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06eb617bec73f4313d9d3b27eea91d57143e19d5","revisions":{"06eb617bec73f4313d9d3b27eea91d57143e19d5":{"kind":"REWORK","_number":1,"created":"2020-05-04 06:39:57.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/88/725188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/725188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/725188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/725188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/725188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/725188/1"}}},"commit":{"parents":[{"commit":"10f7130bcf9c66944b6deb0fcebf57bf8e3bf596","subject":"fix typo in gerrit doc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/10f7130bcf9c66944b6deb0fcebf57bf8e3bf596"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 06:39:47.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-04 06:39:47.000000000","tz":120},"subject":"Switch to new grenade job name","message":"Switch to new grenade job name\n\nThe integrated gate templates has switched to the new grenade name\n(grenade-py3 -\u003e grenade). This repo uses an integrated gate template\nbut also listed the job for irrelevant files an extra entry.\n\nRename the job following the template change to avoid duplicate\ngrenade runs.\n\nDetails:\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014602.html\n\nChange-Id: I5fe73e09dc12018fc64226fd22d06fedd5b15d42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06eb617bec73f4313d9d3b27eea91d57143e19d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06eb617bec73f4313d9d3b27eea91d57143e19d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
