)]}'
{"id":"openstack%2Fglance~727492","triplet_id":"openstack%2Fglance~master~I8ce6e3aa6e8d9dbb04261238c37744b053f72eca","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I8ce6e3aa6e8d9dbb04261238c37744b053f72eca","subject":"Fix hacking min version to 3.0.1","status":"MERGED","created":"2020-05-13 00:24:19.000000000","updated":"2020-06-25 16:31:15.000000000","submitted":"2020-05-14 19:39:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727492-1589485178766-3f177800","meta_rev_id":"2304ff5f0e1cfa3db1b4dff902b11d067548a92b","_number":727492,"virtual_id_number":727492,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 19:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-13 08:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-14 19:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-14 19:04:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 08:46:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-14 19:04:59.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-05-14 19:39:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21a350ad873893979c6bdb59308cd1edd7adc8d3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-13 00:24:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba1c9e693b1774cad479f768c580a1a60427cbad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 02:38:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d292ba46926940b3902512f1ba8abd6e : SUCCESS in 5m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e81540457d024e968e9f4df70b999ed2 : SUCCESS in 8m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d79635d3f69a4623a2ef50aa6998888f : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8ed43261f3343bdb406ab16ffd8182a : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faf903f915d745e9a84de75089d6da1c : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/195914286b8445ce90b4a1585ff77b36 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7da859184f384e56a774d72ec833b7a1 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f94f250e1804fd9b98618fb62fbc15d : SUCCESS in 11m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7829ba3efd86451e962a543c790feb82 : SUCCESS in 8m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e733ee25780a4e13a551a00b2916c216 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1da4e72cb9b26b6ecc90792594ca14e1aad07a7f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-13 08:46:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8235fcfa87a5a714cbe9fb7b1d188487a298478e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-05-14 19:04:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"032185fb2e176d4ff755e2f386ce7c9ad8b3d766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:05:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b50c4a430bb28662956ef5448b5e995d0da0ef6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:39:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b749a3023bbb43aab1a49f77c34160d2 : SUCCESS in 5m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4af1f100b4b24fdcbe149413095f19c8 : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad82b12f6dae44d6b3d394d3ad2a62d0 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67d44548a4434309a5adf7f06ab3cb0e : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bad97741a3414f1395bf6c36e07d88fe : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ebbeb298cb541bba1d9c4f1bb31679a : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c6b3ceb86f40678962e7383cf80386 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f95c614465294c909418f343d2a6e469 : SUCCESS in 8m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a394a88ed434431ca8a6c1712a61cd07 : SUCCESS in 8m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e18ac4e580224cd98f6c3c95b397a335 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5667610e46b6f0048a4643ce3ed36d4f549c2e7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:39:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"188cf574562dd9d95eba309fce2bf30822778209","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:42:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/8f4cb9baf9714bba9afceed8db06104c : SUCCESS in 2m 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68a3b6efbb794d6886e41ca54026f10a : SUCCESS in 2m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea3b4a0387d443d4bdfdf04ccbd045a5 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2304ff5f0e1cfa3db1b4dff902b11d067548a92b","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-06-25 16:31:15.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit ea195853780f260a9b5e1485096f999d546afb18","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf637a2a0c803d636c1b61d0a3cac08219bc2ff5","revisions":{"bf637a2a0c803d636c1b61d0a3cac08219bc2ff5":{"kind":"REWORK","_number":1,"created":"2020-05-13 00:24:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/92/727492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/727492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/727492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/727492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/727492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/727492/1"}}},"commit":{"parents":[{"commit":"d114ed5a7e4e46421c460db706443a2c341f5ba7","subject":"Merge \"Use py38 instead of py37 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d114ed5a7e4e46421c460db706443a2c341f5ba7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-12 23:03:09.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-13 00:24:18.000000000","tz":-300},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nflake8 new release 3.8.0 added new checks and gate pep8\njob start failing. hacking 3.0.1 fix the pinning of flake8 to\navoid bringing in a new version with new checks.\n\nThough it is fixed in latest hacking but 2.0 and 3.0 has cap for\nflake8 as \u003c4.0.0 which mean flake8 new version 3.9.0 can also\nbreak the pep8 job if new check are added.\n\nTo avoid similar gate break in future, we need to bump the hacking min\nversion.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2020-May/014828.html\n\nChange-Id: I8ce6e3aa6e8d9dbb04261238c37744b053f72eca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf637a2a0c803d636c1b61d0a3cac08219bc2ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf637a2a0c803d636c1b61d0a3cac08219bc2ff5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
