)]}'
{"id":"openstack%2Fglance~729468","triplet_id":"openstack%2Fglance~master~I045d12749e952165f6379f5f664ae97c7fd918fd","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I045d12749e952165f6379f5f664ae97c7fd918fd","subject":"Fix pygments style","status":"ABANDONED","created":"2020-05-20 01:42:24.000000000","updated":"2020-07-02 00:54:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"41977a8e1db36bf5757e8221c46c5227a55ab418","_number":729468,"virtual_id_number":729468,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-20 01:57:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 01:57:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc63fb45926acccafb1c32d5638def8667b2b32c","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-20 01:42:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cffed0fd49008c88fe2c06a5420f7bef28eb136c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 01:57:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/73b0019504144f9bb0e2ee97cc759b00 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/022d26ccae0b4652a1f433f0d38104a5 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a726fb3e7e4507865669b582dd06ac : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47720a769d8642478611b52a7d4e0393 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":1},{"id":"41977a8e1db36bf5757e8221c46c5227a55ab418","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-07-02 00:54:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed1f99215b0606a2e816c4c9db9b67b1e46bf5ec","revisions":{"ed1f99215b0606a2e816c4c9db9b67b1e46bf5ec":{"kind":"REWORK","_number":1,"created":"2020-05-20 01:42:24.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/68/729468/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/729468/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/729468/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/729468/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/729468/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/729468/1"}}},"commit":{"parents":[{"commit":"c9c98727690702914784898f999a449f89100aba","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c9c98727690702914784898f999a449f89100aba"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-20 01:40:39.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-20 01:41:51.000000000","tz":480},"subject":"Fix pygments style","message":"Fix pygments style\n\nNew theme of docs (Victoria+) respects pygments_style.\nSince we starts using Victoria reqs while being on Ussuri,\nthis patch ensures proper rendering both in Ussuri and Victoria.\n\nChange-Id: I045d12749e952165f6379f5f664ae97c7fd918fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed1f99215b0606a2e816c4c9db9b67b1e46bf5ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed1f99215b0606a2e816c4c9db9b67b1e46bf5ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
