)]}'
{"id":"openstack%2Fglance~730946","triplet_id":"openstack%2Fglance~master~Ia2435620da6a91a8bd1c08b8ae2e3b231f3c1371","project":"openstack/glance","branch":"master","topic":"cap_jsonschema","hashtags":[],"change_id":"Ia2435620da6a91a8bd1c08b8ae2e3b231f3c1371","subject":"Cap jsonschema 3.2.0 as the minimal version","status":"MERGED","created":"2020-05-26 19:49:10.000000000","updated":"2020-07-10 19:24:21.000000000","submitted":"2020-07-10 19:22:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730946-1594408955991-55706493","meta_rev_id":"869d53876ddf99c1f76943112053305164f894c3","_number":730946,"virtual_id_number":730946,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-10 15:18:18.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2020-07-09 07:50:23.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 19:22:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-26 20:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-10 13:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-10 13:48:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 20:18:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-10 13:48:21.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-10 15:18:18.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-10 19:22:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac0ee022296af6b5735f0a0c1d6ca557b9068b12","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-26 19:49:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93d59e9418fda64b3f6bb951598935ec80394b36","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-26 20:01:37.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027m not sure we should move the minimums to the 3.8 versions yet. We still support py36, so I think we should allow the minimum to be a version that works with 3.6.","accounts_in_message":[],"_revision_number":1},{"id":"a7d6193e764670da965e0baa6c9729ca19dbd618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 20:05:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf008b79b90442ef8e1e826c631418b8 : SUCCESS in 3m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b290289f26904da1bf72592a7bed4a1d : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8dade6713e945af9e6d6cc649a2409e : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bbae83da871438ebb851f54d823e713 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5f909ed1976466e8b355f83ef869771 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49dae60435724160a83be36f5f165994 : SUCCESS in 9m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b19e5c0060c4a57a85ccaa2ddd1521a : SUCCESS in 10m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f9742aea6a224161ba48c2575515fd15 : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b4d7d03ec6729297b50d98b7c1f7caa62bc8b47e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-26 20:11:58.000000000","message":"Patch Set 1:\n\n3.2.0 should work with 3.x so with python 3.6 too","accounts_in_message":[],"_revision_number":1},{"id":"03e719949b82927b56d91c7ddbb24be6b5d9c452","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-26 20:18:46.000000000","message":"Patch Set 1: Code-Review+2\n\nOK, that\u0027s good then. That\u0027s not the impression I got from the commit message.","accounts_in_message":[],"_revision_number":1},{"id":"5b98327b77ead922a704aa9cf75d7b1d2a74b332","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-23 11:44:53.000000000","message":"Patch Set 1:\n\nAny update?","accounts_in_message":[],"_revision_number":1},{"id":"319c42ef81dfb172b2efd37e0902d457a2ef757c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-09 07:50:23.000000000","message":"Patch Set 1:\n\nHello, do you think is it possible to ask for someone to merge these changes to avoid possible future conflicts and rebases.","accounts_in_message":[],"_revision_number":1},{"id":"bb4e5cfb53f38827349c5195c1d6f5a6df0b0bc3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-10 13:48:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM thanks!","accounts_in_message":[],"_revision_number":1},{"id":"ffae6c0aa86a184e119ec08e0ed0ed52b826c8b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 13:48:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"344f38c1cc45bd2c1ab9f49aa36a14fbefe549df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:05:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4ae71b88be134c8486d041cadc76da9d : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/175b52593adb42afa742fbd996370851 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d5b33d5cfe4ac1a4ea64ba4cda5acf : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c744ae5a4a8441f8a55b8ee13ee0eee : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edd3c137d85d41488deb986f8f565416 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85c900f76b4242829185d999e69b6f28 : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/69ea4196bcbe4f7bbd0a478e9596d8f3 : TIMED_OUT in 31m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/910086581913455790c0494e4815f26b : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fcc60d57d95442077b6aa0c7391cacf98faa37d0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 15:06:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d6195b5d44216f539b35e0338d3a887fc3461d0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-10 15:18:18.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"94b8acd2329843b379ccd929bef19d44e17dc73f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 16:04:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b7324e016b24935b0fba06a7ac4fe17 : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbf402abec514e7fa6c7fcd25cd4d63a : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/816dc4f361e3418c9b47b9518012ba25 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4908b4af3ef49c3b475c97e8f874610 : SUCCESS in 11m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cbbada02ef045a5a59f0a2b997e448e : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb729efde8d548a09e53f3dffbb83845 : SUCCESS in 13m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2d97479f1a364def8845d66326e2ff88 : SUCCESS in 9m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c1b5bc80c6c649148459f7c251a17dde : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f9d894424f2c288377fa82b070afac203dc12af7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 16:05:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04ed06672113c5e29b8696cf56c18d77f283892f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 19:22:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e54c815bf1a6f99fd9c20c015f7d72afe7a6e5df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 19:22:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fdaedb5a3ee841568b6270f0bbfd5e1a : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d028d07204ad431ba82a36f64c290a09 : SUCCESS in 13m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15a889a201b246d1bb1983562238a5bc : SUCCESS in 11m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/658d0ff7ed3541db8d9201329599d814 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fba3527958384082aa0511112de87d1b : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69e7e830ccd48129d1003a1444922ce : SUCCESS in 8m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c2504ae7e2fd499c9a4fa5eb8674d456 : SUCCESS in 10m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4ef3ff8ca67640999c1fab7ff324cbd8 : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":1},{"id":"598246f2657477ca53ba30b2fa7bd2483f0e1349","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 19:24:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e10ae1453f904128a7b93e8a514b30d6 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4cd56a49d3a8e39391479ee5abaf8debfc65a892","revisions":{"4cd56a49d3a8e39391479ee5abaf8debfc65a892":{"kind":"REWORK","_number":1,"created":"2020-05-26 19:49:10.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/730946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/730946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/730946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/730946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/730946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/730946/1"}}},"commit":{"parents":[{"commit":"c9c98727690702914784898f999a449f89100aba","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c9c98727690702914784898f999a449f89100aba"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 19:49:00.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 19:49:00.000000000","tz":120},"subject":"Cap jsonschema 3.2.0 as the minimal version","message":"Cap jsonschema 3.2.0 as the minimal version\n\nPrevious versions of jsonschema (\u003c3.2.0) doesn\u0027t support python 3.8 [1].\nPython 3.8 is part of the victoria supported runtimes [2] so we now force\nto use jsonschema version 3.2.0 to avoid issues, remove ambiguity and ensure\nthat everything works with python 3 in general.\n\n[1] https://github.com/Julian/jsonschema/pull/627\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: Ia2435620da6a91a8bd1c08b8ae2e3b231f3c1371\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4cd56a49d3a8e39391479ee5abaf8debfc65a892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4cd56a49d3a8e39391479ee5abaf8debfc65a892"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
