)]}'
{"id":"openstack%2Fglance~732843","triplet_id":"openstack%2Fglance~master~I82b61a2d7687ff109051815e01619eb612ea9073","project":"openstack/glance","branch":"master","topic":"drop_future_imports","hashtags":[],"change_id":"I82b61a2d7687ff109051815e01619eb612ea9073","subject":"Stop to use the __future__ module.","status":"MERGED","created":"2020-06-02 18:14:39.000000000","updated":"2020-07-16 00:14:10.000000000","submitted":"2020-07-16 00:12:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732843-1594858349131-dfe663bb","meta_rev_id":"cf80a524d0c63b4714ab3c80c263916ae173e6cf","_number":732843,"virtual_id_number":732843,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-15 21:39:17.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2020-07-15 12:59:21.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-16 00:12:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-02 18:17:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-15 00:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-15 00:19:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 18:17:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-15 00:19:06.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-15 21:39:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-16 00:12:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"941b7d8a4633339dbfb51034a275cd31e8cec497","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 18:14:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a78ec42127b735cd3b937d3337737465c9bf4de","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-02 18:17:07.000000000","message":"Patch Set 1: Code-Review+2\n\nNone of these are needed anymore.","accounts_in_message":[],"_revision_number":1},{"id":"cb42f7dcd1027dafe5c14fe0013f8bcc768c3557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 19:41:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/eac8526419504e8b965e101278bc9e49 : SUCCESS in 1h 05m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3820f69b5c5146c4a0d64346ff6dddeb : SUCCESS in 1h 20m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee8e14f4648348c983079e8974be14ce : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110da6d44fb84377b58d2fd2e24d65ae : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2db910c46ad8492c85ad319d11b6eef5 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6957deb6f6243f8890bc9a4eb34da67 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77778d3832b94b699bd7b802b0f70473 : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9010a689cf40418783b686196d558b03 : SUCCESS in 7m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e10b7eef73984e2893a5aed780531e5d : SUCCESS in 8m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1144e9d3999d4ee09421f497ed9debf0 : SUCCESS in 6m 05s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e9370b018b3444889edca1d11f673734 : SUCCESS in 1h 04m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ade33664724149f7acf839e59af81a57 : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1eb205922719c9bbc94508325e838f8574dbd818","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-09 09:48:24.000000000","message":"Patch Set 1:\n\nHello, do you think is it possible to ask for someone to approve these changes to avoid possible future conflicts and rebases.","accounts_in_message":[],"_revision_number":1},{"id":"cced829915708adb7f6baeda39e8f44fda301df7","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-15 00:19:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3cf35dd2f137390f20f71f3725a0689b11b2e8e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 00:19:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c91158e10d6933fd72fda57fc5b877d7becda544","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 02:11:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bfed692ccd934c2ca962ad8727d879f6 : SUCCESS in 1h 10m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/10bc6ddc847d4230aa5d819fb5045422 : FAILURE in 1h 20m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2db3fbd5a37443c6b9d924a26964bce5 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c044fd3b31e49119d20f6a2440bbdd5 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5af915b151d0484d80a72ded448d3a10 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf1d1941273244818c59fa2a95f8e005 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129a2b00da22465da81066696c91f4e1 : SUCCESS in 5m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/94c4c94b53484964b5637b995cd4e8b5 : SUCCESS in 7m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fce2359e1bb94f44b24890c7f57136d6 : SUCCESS in 13m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54a98c9c925842e6a7959c203bbfc5e0 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":1},{"id":"36d92eaf3f35f2449543d0b416566db35a1ad9c0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-15 12:59:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a2647b9491782802b6618b8560061755181f98b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-15 13:05:52.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"7853c8e185d7016cf73e7ed4f7558afdfbc6d183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 18:57:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/893fe8cd899b485abc8adedd798cf319 : SUCCESS in 1h 05m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f9089df4adf34bf4b32f863537d1cdbe : SUCCESS in 1h 09m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4490e29ebab47dbbb1496b54fe98069 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c2081d647f34579af883fcac95935c2 : SUCCESS in 9m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/445ab48158b74c26bc42ed3688729911 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/581836fb1b4c45e6888e8869771ee0c2 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5003e067b842f2925029931331a97b : SUCCESS in 7m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dfdd2d0f39854f8bb3c9b15dd6878454 : SUCCESS in 11m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3d717a0ae9fd4bf48073f980ac318ba8 : SUCCESS in 13m 10s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7bcebdfd1e124913bca7f7887ddff22f : SUCCESS in 6m 06s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6575974d5cec4ac083513c4d361982c7 : SUCCESS in 1h 11m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/751d3f4f154b4621888dbd47b9b5bb39 : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1fc879f5342e3cdbee6ea89fdc49f64aedd3803e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 18:57:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13fd0c3353d2f717aeb0d8cd25bd381bffe13162","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 20:26:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1d6df5c29a424b6fb662573dbaf8ead1 : SUCCESS in 57m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7aaa962bd8f14d2585d952bc191ef342 : FAILURE in 1h 26m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48d5178991ef43abac8bf0032a83205b : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeab306e3ac2432397332606e5ea3933 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76481cac1cfb4bcb9dfe9a54fb50ff09 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9644fdb05e644915841a85f6eb3444e1 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eafcb3ea41e4da1b700d3b7730e8e5f : SUCCESS in 7m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/21f621c0a0e94747957ed865c32cdc13 : SUCCESS in 8m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a7890241d4134981814781dde9f758c3 : SUCCESS in 10m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1fdf224206a4a04842ca18d735d90cb : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":1},{"id":"552797d612f6784a45a2274f10f3db04a2afc733","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 20:31:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77ce748cf6fb0d83ef4267a287d3300f6efc4338","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-15 21:39:17.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"79bac98c670cebb7fa3b8f3a17182d47ad14944a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 22:51:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/62864554151142c9aa0394423322b20c : SUCCESS in 1h 02m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1328b8bc650e4cd1b867e50f638d6c33 : SUCCESS in 1h 22m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e586cf0ad344cedac0f92a5a0157e18 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9326430ea56e486d81304d816e265a2c : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/289f7532c3684c27beebe29326e52dd0 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adbf5291f7374c9ea73c771fc246e546 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fb31316ad104d40b24dbe00454eb7d2 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1fa9aab50ea44bb88326c4b0f19d9d78 : SUCCESS in 7m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6e1e508880fe40e19bc408ad37f33fa5 : SUCCESS in 8m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0208e56367e746958e6962787273e770 : SUCCESS in 5m 07s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/86f289776f1a46db971f656ae4837497 : SUCCESS in 1h 04m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/673f1d913fd3435c82fc6b17cf7b20ec : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0d721a4a1138b9d3a0aed3d9eedea2ca9cf54bf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 22:51:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"293232100487bfd8024d7726ec3c5106c1706892","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 00:12:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d087174ed820452a8362455a87d11964 : SUCCESS in 1h 02m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5d4e5edc4e742c0bc53f036b7765499 : SUCCESS in 1h 01m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f04d3a0b00374275858b5c5f97127321 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f0298a1662a48ce8d414d80c35afe6a : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ba5bfedbe1447a0a76865711b52b574 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0d1652509bf461aa6f56a69ca838167 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f15a64faa71a4107a13f67b22f24a647 : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26b1061ae1ed4378ac6a764bf6dd4fed : SUCCESS in 8m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a8cc666cdc4c485485bbe55b464b371b : SUCCESS in 11m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f56d2bdee584796b98ee297890e7ad3 : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":1},{"id":"abf9f2237a62bad5115f8129969a840eb63f3e7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 00:12:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8ba9112f9675d2c3d124e016a29e606c58306030","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 00:14:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd75a095b3094c70a25d50b2399d6c20 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64e3c4116720531d69a522d1b056636a0b0065b5","revisions":{"64e3c4116720531d69a522d1b056636a0b0065b5":{"kind":"REWORK","_number":1,"created":"2020-06-02 18:14:39.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/43/732843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/43/732843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/732843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/732843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/43/732843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/43/732843/1"}}},"commit":{"parents":[{"commit":"03cb88a4376cab5e80697800591db47ca801cc56","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03cb88a4376cab5e80697800591db47ca801cc56"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 18:14:31.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 18:14:31.000000000","tz":120},"subject":"Stop to use the __future__ module.","message":"Stop to use the __future__ module.\n\nThe __future__ module [1] was used in this context to ensure compatibility\nbetween python 2 and python 3.\n\nWe previously dropped the support of python 2.7 [2] and now we only support\npython 3 so we don\u0027t need to continue to use this module and the imports\nlisted below.\n\nImports commonly used and their related PEPs:\n- `division` is related to PEP 238 [3]\n- `print_function` is related to PEP 3105 [4]\n- `unicode_literals` is related to PEP 3112 [5]\n- `with_statement` is related to PEP 343 [6]\n- `absolute_import` is related to PEP 328 [7]\n\n[1] https://docs.python.org/3/library/__future__.html\n[2] https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n[3] https://www.python.org/dev/peps/pep-0238\n[4] https://www.python.org/dev/peps/pep-3105\n[5] https://www.python.org/dev/peps/pep-3112\n[6] https://www.python.org/dev/peps/pep-0343\n[7] https://www.python.org/dev/peps/pep-0328\n\nChange-Id: I82b61a2d7687ff109051815e01619eb612ea9073\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/64e3c4116720531d69a522d1b056636a0b0065b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/64e3c4116720531d69a522d1b056636a0b0065b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
