)]}'
{"id":"openstack%2Fglance~733683","triplet_id":"openstack%2Fglance~stable%2Fussuri~I2ad41022aa709491881e78129708da0ccc25f4f6","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1881958-stable/ussuri","hashtags":[],"change_id":"I2ad41022aa709491881e78129708da0ccc25f4f6","subject":"Exclude http store if --all-stores specified for import/copy operation","status":"MERGED","created":"2020-06-04 19:12:27.000000000","updated":"2020-06-05 10:12:39.000000000","submitted":"2020-06-05 10:09:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733683-1591351793202-0bd3b327","meta_rev_id":"2f5a3dc67477defee8728786b6ce6d3ff48feaaa","_number":733683,"virtual_id_number":733683,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2020-06-05 07:15:39.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 10:09:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 19:14:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 23:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 23:24:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 19:14:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 23:24:11.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-06-05 10:09:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54f3223d30a76089b406c12dabc8a22eb07816f4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-04 19:12:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6b7f7a5863141b86b14932b5bbfb822a9625e255","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 19:14:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"54e6cf20101ad57f81d782c4a5deddfdb2822d56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 20:21:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/540b4f9cd10843efbbb6c41a1be9ab42 : SUCCESS in 1h 06m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/873577d6ea974de198f67b721a977eca : SUCCESS in 1h 01m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e2f76599beb4cce947717519ddb709b : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7075a6d4d544e939b2083ea3f69af31 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/065137ac768a48e6a47b7c58db619e77 : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d6d5e2ad9d2464a8f5d52e59f85679c : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f2fe98bf4a4acca810f6e202120628 : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa8083aff30e49c697a2a6820304474b : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63b35968fb9f4c6ba179d71cd4ff1564 : SUCCESS in 8m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/01a2dadb3058477abbc5bda50ea5fa7c : SUCCESS in 9m 38s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8d3ef1da066148a48f99171b70fd6fd5 : SUCCESS in 6m 02s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1c6960ff1a37412590e3a1ad2ff1234e : SUCCESS in 1h 03m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0412fdb2f9c54f1d9d40548c319285fd : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":1},{"id":"295204b51c2ff53491561ba7a29705d2fd0940a3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-06-04 23:24:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ff0fd490f6e180269d04e9cc23cdcc63ec2a4ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 23:24:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56908c47ac350cf11d01e931c8e2dd7d334aae0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 00:43:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a63c8a7dd1f14ef0b2ef7206025046a2 : SUCCESS in 1h 13m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a16d2c094204ec0b297993638d65b80 : SUCCESS in 1h 04m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/064ca8ddbe88463d8988e8e011f63383 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72bf91febf904c5893fb83c99d81cfa9 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/701e33b231d3498ea73810833350490a : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04817e6b2c6d42afbb257e99b68aa24f : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1930d14d28f24098b3bfc1ef2d387e47 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93db00b53cee4d8b8338ff9f2069c07e : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/44df2f26bb354875bed862968a2a26d1 : SUCCESS in 7m 04s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e22fa030f70b46a384c7e95808586a8b : TIMED_OUT in 31m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc3f0b2c8b1747fcbd115531754a53cd : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":1},{"id":"46436178b161ae277d1a7f58062fb4b8f6b7dc2d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-05 07:15:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3c07f3ee102f1530e2defea1f62e92d254a26ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 08:50:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5cdba794a2f847158d1295e06a2d97e0 : SUCCESS in 1h 29m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4fe0101b0b91407996fdd9b35a2ac8f9 : SUCCESS in 1h 16m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f00c0c0cec784058b36c2fae5c7445ec : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dee4a82ac34648ac870bac9e68749c4e : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78d7fce601c44539be8bca7b6b7df51f : SUCCESS in 7m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cd879f2e0244dd59f5af668f681eb35 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8571a05201c0494880d87f116fd2c3d0 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3accea37e5cf4bc3bd0a2a2972d9a588 : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b4087beef414b998ad6c21ae62c930a : SUCCESS in 7m 20s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3872f87f098941219af9560468a0620e : SUCCESS in 7m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ea9693ff90974144a4bd34c35dbf32aa : SUCCESS in 5m 49s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ce3ce43c0dea4d60804d405b54d672f4 : SUCCESS in 1h 07m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e095d28ec06b4dc0a65fa9d6a3b994ee : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ae008401813f31272d1c3bb1d4b663435e440194","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 08:50:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6246e0edb5e96f405e1f202506da9119df75ee18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 10:09:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/fd119cf8ff1541ea8cee390588d44518 : SUCCESS in 1h 18m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/73bb21165f59427296cd815f4250dcf9 : SUCCESS in 1h 11m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bab121e7537e47f7a6f58c071d979bf6 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f524d1aec324417b4628ea066237258 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5045adfd6514098a967116e9252b1f6 : SUCCESS in 9m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b3d08d15f6b54681a60a408dd2997469 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78473e8c38cd440a8d86a237352d4419 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4da9b792ccc4f4f90030e760005ddfa : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3da4c0b80d244d7e86b2bfbfcced5b87 : SUCCESS in 9m 34s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/244a16eb1f2c4d7d8ca686477279d5aa : SUCCESS in 8m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/210997ea0aed401487b920749c7caa33 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"fb018d05de10128bb11ffd145ed13d702c00f9df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 10:09:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fae6bc444b2bbaf54ca643e8d0bbcae5bc36d8c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 10:12:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62943746be74c808a3f02e2af32b64f : SUCCESS in 1m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95ebc4ef646447a18389deefdc502743 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3fd60f70b07ddf7d2b79c5fc2704bf1619f4c96","revisions":{"c3fd60f70b07ddf7d2b79c5fc2704bf1619f4c96":{"kind":"REWORK","_number":1,"created":"2020-06-04 19:12:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/83/733683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/733683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/733683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/733683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/733683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/733683/1"}}},"commit":{"parents":[{"commit":"f78463ed32f67e551662c77c6bd9beff80a2121f","subject":"Switch to new grenade job name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f78463ed32f67e551662c77c6bd9beff80a2121f"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-06-03 20:25:06.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-06-04 19:12:27.000000000","tz":0},"subject":"Exclude http store if --all-stores specified for import/copy operation","message":"Exclude http store if --all-stores specified for import/copy operation\n\nIf operator has configured read-only \u0027http\u0027 store in multiple stores and while\nimporting or copying existing image in multiple stores if user has specified\n--all-stores true then depending on `allow-failure`; if it is True then image\nis imported/copied to other stores than http or if it is false then image can\nnot be imported/copied to other stores as well. Aslo, new image property\n`os_glance_failed_imports` shows `http` as failed store.\n\nExcluded `http` store from the list of stores for import/copying workflow\nif user specifies --all-stores as True.\n\nChange-Id: I2ad41022aa709491881e78129708da0ccc25f4f6\nCloses-Bug: #1881958\n(cherry picked from commit f7b163119052b96e3d1577c61982ef3b18ecdc97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c3fd60f70b07ddf7d2b79c5fc2704bf1619f4c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c3fd60f70b07ddf7d2b79c5fc2704bf1619f4c96"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
