)]}'
{"id":"openstack%2Fglance~734215","triplet_id":"openstack%2Fglance~master~I7bf3284cba4c38605fb50b3c458e53f896f34086","project":"openstack/glance","branch":"master","topic":"remove-reg-entry","hashtags":[],"change_id":"I7bf3284cba4c38605fb50b3c458e53f896f34086","subject":"Remove configs and entries for deprecated registry","status":"MERGED","created":"2020-06-08 20:03:06.000000000","updated":"2020-06-09 14:37:17.000000000","submitted":"2020-06-09 14:35:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734215-1591713326330-f90b450b","meta_rev_id":"d57d16a97dcf75701e60ff9e47b3b2e02643c452","_number":734215,"virtual_id_number":734215,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-09 11:23:15.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-09 11:22:52.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 14:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-08 20:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-06-09 09:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-06-09 09:20:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 20:13:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-09 09:20:40.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-06-09 11:23:15.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-09 14:35:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"264857c72e7bab0e062b51ae3b8462ad1f06b374","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-06-08 20:03:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"daaea67da9a3f541d38f6257523ca16fd0b7b190","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-08 20:08:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)\n\nLooks like everything is covered. A couple minor things inline, but nothing I would consider critical.","accounts_in_message":[],"_revision_number":1},{"id":"686d47e3ad54e6478b8ded598a21fadfb348c1b6","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-06-08 20:12:19.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3fd4b9f90b2d4692532d703bff651bc22b05adb0","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-06-08 20:12:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f649a6192a40841cd195be7bff7b6dcdd47b257","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-08 20:13:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"08553e974d92aa3ed202af33045fd141a030ae69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 22:32:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6aac260d9d424bd18ed25902ece87f2e : SUCCESS in 1h 08m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ed136d07b3f490aa2b50bee28e68708 : SUCCESS in 1h 20m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c7e5421d7fb4b2098ad59ba1e26e41c : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9c0bb6e231b4a248b1f5ca3ffe1c3ec : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d49aeda8c1e448ae83e426ff836cac9f : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bea8bdd2dbc3442c9aaf3b6aa5c29fc3 : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de974f7aa9d499b86c5c4da0e11e9f4 : SUCCESS in 8m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c89903b5816e49498b6161e8f2704b75 : SUCCESS in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a5ff88a7c6364225a2ef23430e937f5a : SUCCESS in 9m 03s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/abdb6f606aea41a4a7d57ebdfc88bef2 : SUCCESS in 7m 33s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/264a6fc9c2b347b1baa23c8be39f47bc : SUCCESS in 1h 05m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bfae6628a814d87beb10580922dc0e3 : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":2},{"id":"9e9bf956fb27740c22fddf4bb2087c18fc503066","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-09 09:20:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThank you!!!","accounts_in_message":[],"_revision_number":2},{"id":"c476b4372045813748efbcbbeb92194d6bfb1d90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 09:20:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fc1a9fa755405a5a5f1b71ce2b4a8d3aaf4ccb56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 10:38:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6dcbcf7a056b407cae40bad350b4c7de : SUCCESS in 1h 02m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70695e6300724228a5d5cb1a34ae692a : SUCCESS in 1h 12m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ee76252263f41a9a10e9bf6a8bd55c4 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6865fbc8eb314de7ad6fe9313bc0d7eb : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0f28c0a0e4d48d3942faf7f52981102 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3e18f882d9142289c327c403880f9c0 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4476e3c4446b4b5da99796a5ecac4747 : SUCCESS in 7m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8702f48052014555a5548408103bb155 : SUCCESS in 15m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a11a9623aaf141bb8c21573919237d04 : FAILURE in 17m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/835d333180344d2bb11f1379a104e758 : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":2},{"id":"15dca8a8c26c75de81d71538b2752ac88263da5c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-06-09 11:22:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fb74dc7f74c2b147b34b9d4b28ae097597eb8668","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-09 11:23:15.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"e6a92bff0f64530fc54a2343d7c31d2e9c37a7f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 13:25:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5fb74d4106604e6e9b099f7b02283797 : SUCCESS in 1h 15m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/291525dddd3b482fbd7ae27b461efe41 : SUCCESS in 1h 04m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9ad690723884fa2b70d8f78bfb4e37e : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e61cf1477d57406db346afd7c2d8a9fe : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74a59022b5d747ebb04c1b7c828f20f2 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98eebc2745c2408fbbcf67626b1cecd3 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02bc9f17bfc9460dac190b010653a3d6 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4860ae4a4bd04bb3bc7d2bfd44b82efd : SUCCESS in 7m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e5a4c42158141ca9e7594a9c2d7e52a : SUCCESS in 8m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f97165171570466d9d85a22ba9a191bf : SUCCESS in 4m 23s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/369467799a984d3cb90e62077c606062 : SUCCESS in 1h 00m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aaf177f758db4099a9860920b38903e2 : SUCCESS in 1h 05m 11s","accounts_in_message":[],"_revision_number":2},{"id":"7a8fb80a05290a53bf41996185290d87745b302e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 13:25:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cdd590ce541844ffc2a9b5f8185bc6e9683f827a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:35:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4261d3cb4b9e47debaf96399aeb88d65 : SUCCESS in 59m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/426d35e95ffa430084874e174ddc943c : SUCCESS in 1h 07m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13051a2a1f124ed7b582cf9b2aac56d2 : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b73dd63f03ba4a9fbb84a8087f3634fd : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc90c836731348f7b53f624ea2a226a5 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48ccc3522d41416b99d78c5d0b3c94dc : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5252b5c9a6ae49bd95d16c8e0b62e9b4 : SUCCESS in 7m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/08a7855bec544e288577bed4974f1103 : SUCCESS in 11m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c11ee2174ab94c77a544b32827900481 : SUCCESS in 10m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3588b26a5da941e9b15a83088d792707 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":2},{"id":"44a0122a201d8e390a7d27db06efc22fb49508fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:35:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"16bafd26e848299f3c136629581f8f5b1d0aa08e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 14:37:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06637bb5a40044dc99ccfe1e3d38346b : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"781da9354ff0bcd3c22cbe0dc5beb5f204d91769","revisions":{"3502a62bc0b75ab08bad3f8590015c07dba5451f":{"kind":"REWORK","_number":1,"created":"2020-06-08 20:03:06.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/15/734215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/734215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/734215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/734215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/734215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/734215/1"}}},"commit":{"parents":[{"commit":"f7b163119052b96e3d1577c61982ef3b18ecdc97","subject":"Exclude http store if --all-stores specified for import/copy operation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f7b163119052b96e3d1577c61982ef3b18ecdc97"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-02-13 12:30:50.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-06-08 20:02:44.000000000","tz":60},"subject":"Remove configs and entries for deprecated registry","message":"Remove configs and entries for deprecated registry\n\nIncluding removal of the example conf in docs.\nIncluding glance-registry command entrypoint to eventlet.\nIncluding rpc_controller from wsgi, changed to reject.\nNot including the files devstack is depending on.\n\nThis change means that glance-registry starts and announces\nthat it has been removed.\nThis change means that when ran as wsgi app, all calls to\n/rpc endpoint will be rejected.\n\nThis will allow devstack to make a graceful transition to\nlife without the glance registry, which was deprecated in\nNewton release.\n\nChange-Id: I7bf3284cba4c38605fb50b3c458e53f896f34086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3502a62bc0b75ab08bad3f8590015c07dba5451f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3502a62bc0b75ab08bad3f8590015c07dba5451f"}]},"branch":"refs/heads/master"},"781da9354ff0bcd3c22cbe0dc5beb5f204d91769":{"kind":"REWORK","_number":2,"created":"2020-06-08 20:12:34.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/15/734215/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/734215/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/734215/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/734215/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/734215/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/734215/2"}}},"commit":{"parents":[{"commit":"f7b163119052b96e3d1577c61982ef3b18ecdc97","subject":"Exclude http store if --all-stores specified for import/copy operation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f7b163119052b96e3d1577c61982ef3b18ecdc97"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-02-13 12:30:50.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-06-08 20:11:45.000000000","tz":60},"subject":"Remove configs and entries for deprecated registry","message":"Remove configs and entries for deprecated registry\n\nIncluding removal of the example conf in docs.\nIncluding glance-registry command entrypoint to eventlet.\nIncluding rpc_controller from wsgi, changed to reject.\nNot including the files devstack is depending on.\n\nThis change means that glance-registry starts and announces\nthat it has been removed.\nThis change means that when ran as wsgi app, all calls to\n/rpc endpoint will be rejected.\n\nThis will allow devstack to make a graceful transition to\nlife without the glance registry, which was deprecated in\nNewton release.\n\nChange-Id: I7bf3284cba4c38605fb50b3c458e53f896f34086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/781da9354ff0bcd3c22cbe0dc5beb5f204d91769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/781da9354ff0bcd3c22cbe0dc5beb5f204d91769"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
