)]}'
{"id":"openstack%2Fglance~734665","triplet_id":"openstack%2Fglance~master~Ie3698c7334b31c2291b41fd3f7e21add0dd6a19b","project":"openstack/glance","branch":"master","topic":"lower-constraints","hashtags":[],"change_id":"Ie3698c7334b31c2291b41fd3f7e21add0dd6a19b","subject":"Update lower-constraints versions","status":"MERGED","created":"2020-06-09 18:29:59.000000000","updated":"2020-07-06 16:41:59.000000000","submitted":"2020-07-06 16:37:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734665-1594053479414-1a0f072d","meta_rev_id":"2790861b079e9a2f8020a6d830e2da206d764089","_number":734665,"virtual_id_number":734665,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-03 18:32:55.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-06 15:20:29.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 16:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-06 15:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-07-06 13:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-06 15:22:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-03 18:32:55.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-06 13:50:15.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-06 15:22:48.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-06 16:37:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5237b62562d722244977a236b6177a689b198cb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-09 18:29:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99e51b6fdc125d4d960fae48f769e19b5f00d41a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 18:45:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4bbef8e55654acbbe57dd4eac3c2c59 : SUCCESS in 3m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf305f5356ea419999c05991ef683158 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95027a6e6ba84148b39dacf773e699f3 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1082d58864d541c2b9458fe6e4cc22d3 : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22612b3f802f43cca5881f5c098eaa45 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6598a006df8b4445a3e7cc199f311848 : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d0595d856b944cc0a91abb0d0e83bd73 : SUCCESS in 8m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5c8d6c53c9754794bfb5d90df02f100c : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":1},{"id":"17c2e2c06cf68b3f5590219d12369b272e14852d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-03 18:32:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ac90b7bf4c988c6be1970c2539b455992930482","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-03 18:32:55.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"26cb3a73c2a7fbbfb2f6ee43f7771a7be5ce7cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 18:46:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/88700f532ed84523afabbca7d6802cb9 : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01c9f0a2bd2340018458f2adc3511d49 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69829baffc2c47908485661291cd4316 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fbc84439fbd446a9e733d3473b2e95e : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c68e5571448e43d9a059cfd878bbff69 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbfb0442936a4117bd123739d940907d : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6470504d538b4a76a63014bc43609bef : SUCCESS in 7m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4d9ec2d7fa8d457188e07bd5ae577595 : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":2},{"id":"da9ff565af50d13bb0ba931b2b153804a28f9449","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-06 13:50:15.000000000","message":"Patch Set 2: Code-Review+2\n\nThank you sean!!","accounts_in_message":[],"_revision_number":2},{"id":"fa3276485b06ff2744190751b25f0967cc9d34fc","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-06 15:13:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nQuestion about the mock, otherwise looks ok.","accounts_in_message":[],"_revision_number":2},{"id":"8f619f76a7ca4cedbaf09a34008f14de6d22ddba","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-06 15:20:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee4f366bfc97ff4a73703033842287c76bbac72c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-06 15:22:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nOk, thanks for clarification and the refreshing the list Sean!","accounts_in_message":[],"_revision_number":2},{"id":"afc5d9fd4cac322c82b7b73081d946cf97541a1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:23:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"60c979d3f234fafa5abf87dafa6b05c00a538e3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:37:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"38c06ad6c283ff012d7343bc97f2f98ca224aee4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:37:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/318914886de74dd58b237ada30c17f2e : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e715975afc784271b94aec5b4f81ef87 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e929aa35e0bf4def818ad8de7219af26 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e45268a2899e4ac28078ddefdb805c95 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7108c91584c4b29b0794407a8656d6e : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fc9ccdd0ad74326bfbaec964fe72ace : SUCCESS in 7m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fff38ec8aefd429f82960e3367fdf9d0 : SUCCESS in 8m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/494cebd967d641cda6877506f21fafbe : SUCCESS in 10m 34s","accounts_in_message":[],"_revision_number":2},{"id":"dcfeadb7414edce84451242b07daf87c2f160d6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:41:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2ca038e1ba542b39b1f311909710f2e : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9eb562ebfa68bc419a82c05d1f96c8477322e25a","revisions":{"3e914fe06940469b02e797ecb0dd4c3dbc3c28e0":{"kind":"REWORK","_number":1,"created":"2020-06-09 18:29:59.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/65/734665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/734665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/734665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/734665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/734665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/734665/1"}}},"commit":{"parents":[{"commit":"781da9354ff0bcd3c22cbe0dc5beb5f204d91769","subject":"Remove configs and entries for deprecated registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/781da9354ff0bcd3c22cbe0dc5beb5f204d91769"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-09 18:27:10.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-09 18:29:54.000000000","tz":-300},"subject":"Update lower-constraints versions","message":"Update lower-constraints versions\n\nThis updates lower constraints to versions that will work with py38 so\nthat when we move to running on focal nodes, which has py38 as its\ndefault py3 runtime, the lower-constraints job will continue to pass.\n\nIt also cleans out some secondary requirements that are no longer needed\ndue to our direct dependencies being updated.\n\nLinters are removed that are kept in the global requirements blacklist\nas those are not version tracked and are not relevant for our\nlower-constraints unit test runs.\n\nChange-Id: Ie3698c7334b31c2291b41fd3f7e21add0dd6a19b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e914fe06940469b02e797ecb0dd4c3dbc3c28e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e914fe06940469b02e797ecb0dd4c3dbc3c28e0"}]},"branch":"refs/heads/master"},"9eb562ebfa68bc419a82c05d1f96c8477322e25a":{"kind":"REWORK","_number":2,"created":"2020-07-03 18:32:35.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/65/734665/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/734665/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/734665/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/734665/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/734665/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/734665/2"}}},"commit":{"parents":[{"commit":"e1a2fb9eccd577d6eef048906ce300228e262d16","subject":"Merge \"Use unittest.mock instead of mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1a2fb9eccd577d6eef048906ce300228e262d16"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-09 18:27:10.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-03 18:32:23.000000000","tz":-300},"subject":"Update lower-constraints versions","message":"Update lower-constraints versions\n\nThis updates lower constraints to versions that will work with py38 so\nthat when we move to running on focal nodes, which has py38 as its\ndefault py3 runtime, the lower-constraints job will continue to pass.\n\nIt also cleans out some secondary requirements that are no longer needed\ndue to our direct dependencies being updated.\n\nLinters are removed that are kept in the global requirements blacklist\nas those are not version tracked and are not relevant for our\nlower-constraints unit test runs.\n\nChange-Id: Ie3698c7334b31c2291b41fd3f7e21add0dd6a19b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9eb562ebfa68bc419a82c05d1f96c8477322e25a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9eb562ebfa68bc419a82c05d1f96c8477322e25a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
