)]}'
{"id":"openstack%2Fglance~737382","triplet_id":"openstack%2Fglance~master~Ie141779d9df63dce9a4107f045506a3a43d95944","project":"openstack/glance","branch":"master","topic":"bug/1884587","hashtags":[],"change_id":"Ie141779d9df63dce9a4107f045506a3a43d95944","subject":"WIP: Create image import factory with admin context","status":"ABANDONED","created":"2020-06-22 21:29:19.000000000","updated":"2020-06-23 13:58:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b066c2804d07513ac89cce2e96ce478564576ebe","_number":737382,"virtual_id_number":737382,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-23 13:33:43.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-22 23:41:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-22 21:29:37.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2020-06-23 13:33:24.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 21:29:37.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-06-22 23:41:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-23 13:33:24.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-06-23 13:33:43.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8c84b6eba64e5301a60d9d3639700bd50030437","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-22 21:29:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"926ce19d8a3b63955ef2e971e9d113fd0f8e7a23","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-22 21:29:42.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"5affde3a02632ba1dc03ce3a91025dc170068336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 23:41:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2db94c0613ac4df9bd07094cb1d3dd10 : SUCCESS in 1h 10m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0cf18992935c4fc89a39c27b0aaebba5 : TIMED_OUT in 2h 09m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d771c989d8ec4fefad465eeafd078ff4 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c37a03157db42b88b906fe6bcb58095 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d037a77cc84c4b42aa2f562d4a07d731 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8b20806b304469ba77f5263463f4449 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf879febb02d4b6aa93afeaea505b234 : SUCCESS in 8m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7cb2958964864bb1a9530e7d6feeeaf3 : SUCCESS in 12m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b48fb085851f4da6ba32d9992e4fe582 : SUCCESS in 13m 57s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/69a2883681304818bee553f4c0793403 : SUCCESS in 7m 32s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/51fc22c31be2481e93359efca768edf7 : SUCCESS in 1h 27m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db0bd1924ca5461ba2bdb14f7cbc5a2a : SUCCESS in 1h 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"39d751fde328c58e684921544bcb67331e1aaca5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-23 08:59:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fcc5584268a9710112fe4803ddcca639d04c222a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-23 09:00:16.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"11baca99a6c96a69dfdccc14439646f72533ddc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 10:19:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/77bda0f3f2eb448db6c010d0f9c7b55b : SUCCESS in 1h 04m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e6ae6a807ec740ae85616a184032db9c : SUCCESS in 1h 08m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66b0e6d91133482f9984a655776640f1 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb912c8d089b47cb8b61b2a2d66ba903 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc4d7abf81ac46c89830d3f97eb82ea6 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67204db69dbc430986bd2deec7450131 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eac4711001d4983921e25220ca70d6f : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19cc598283024ea8b4dad25995b084fb : FAILURE in 15m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/26615251c33d43f29c59c35f419a1022 : SUCCESS in 9m 26s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a0af300cb0a14d75bd5530858faec2dc : SUCCESS in 5m 32s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/21d0851683364ef79326c2b235d7bc64 : SUCCESS in 1h 02m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fdc83a458c142169659796546287e13 : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6108bb07c325bf84bf04024c13c1fb12dfae37ed","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-23 13:33:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb857e20f93e8a460031c232e4ee27ece9d50891","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-23 13:33:43.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"c457dd4c5077d0a28dce5628d43ad1ee93d467eb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-23 13:58:28.000000000","message":"Abandoned\n\nSounds like we\u0027re expecting only the owner to ever be able to copy images, or admin, so this is not a solution.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e5f0c3be25d2414f56b829fd8cc73dcf93d02f9","revisions":{"8e5f0c3be25d2414f56b829fd8cc73dcf93d02f9":{"kind":"REWORK","_number":1,"created":"2020-06-22 21:29:19.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/82/737382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/82/737382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/737382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/737382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/82/737382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/82/737382/1"}}},"commit":{"parents":[{"commit":"c9c98727690702914784898f999a449f89100aba","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c9c98727690702914784898f999a449f89100aba"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-22 21:27:05.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-22 21:29:28.000000000","tz":-420},"subject":"WIP: Create image import factory with admin context","message":"WIP: Create image import factory with admin context\n\nI have no idea if this is right or legit. It seems to me that the API intends\nfor users to be able to call copy-to-store on images that they are able to\n*see*, yet the task itself implicitly disallows this on images they do not\n*own*. The justification for this would be that they could just as easily\ndownload and re-upload the image in order to be able to copy it. Assuming that\nis the correct intent of the API, I don\u0027t know if this is the best way to go\nabout it or not, but this can serve as a placeholder for it to hopefully let\nme proceed with my nova-side test.\n\nChange-Id: Ie141779d9df63dce9a4107f045506a3a43d95944\nRelated-Bug: #1884587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e5f0c3be25d2414f56b829fd8cc73dcf93d02f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e5f0c3be25d2414f56b829fd8cc73dcf93d02f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
