)]}'
{"id":"openstack%2Fglance~737417","triplet_id":"openstack%2Fglance~stable%2Fstein~I5ef940c6093aa1ccc660b7d82ba136f261086a4a","project":"openstack/glance","branch":"stable/stein","topic":"grenade-em-nv","hashtags":[],"change_id":"I5ef940c6093aa1ccc660b7d82ba136f261086a4a","subject":"Make greande jobs n-v for EM and oldest stable","status":"ABANDONED","created":"2020-06-23 02:06:19.000000000","updated":"2020-11-12 16:20:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5093a5d7ac3fb781cf37757b0cb69c5a08be5731","_number":737417,"virtual_id_number":737417,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"date":"2020-11-12 14:27:53.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"date":"2020-11-12 16:20:08.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2020-07-13 22:18:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2020-07-13 20:47:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-06-23 13:23:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":-1,"date":"2020-07-09 13:14:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2020-07-13 20:47:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 13:23:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-09 13:14:00.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-07-13 20:47:41.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-13 22:18:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-12 14:27:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"57e1e3ba046dd08bf818443b5c371de879034340","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-23 02:06:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8100b318fe95e4433719002d57e0747b7eba07c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 03:59:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/db2a0f7ec25c4944a9e2bf88924add86 : FAILURE in 18m 37s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eded9addf7f7424989eb234bb480dfe5 : SUCCESS in 1h 51m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc300f23ebba404387e5a93bb33f45a5 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6a0dfd1848e449292cd295340954552 : SUCCESS in 7m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fa2f3c8da8248f8916c2e038ce42d34 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cd800dc75c64634a37b09ad7baef48b : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b50c468e97724fe9b335e6f3825d0940 : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/006b54ac7edf435388ab6b8d6315c8be : SUCCESS in 8m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ee72199394864e3e90c044b5ba921b71 : SUCCESS in 9m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/afa6c077efe8427ea51baae8b40aae69 : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":1},{"id":"7db33dbb757f9733ae6af0e759dde08588bd53ef","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-23 13:23:37.000000000","message":"Patch Set 1: Code-Review+2\n\nMakes sense to me.","accounts_in_message":[],"_revision_number":1},{"id":"18ea8115ae34c7e08d7445a21a630948a621406a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-09 13:14:00.000000000","message":"Patch Set 1: Code-Review-1\n\nrecheck\n\nGrenade jobs should work now as devstack @ Rocky is fixed. So this patch is not needed I think.","accounts_in_message":[],"_revision_number":1},{"id":"a0e635e91244c36ba02786a1cc4b735ef586a754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 19:21:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e4ced8e470894ea29dfcb53935d4626a : SUCCESS in 1h 13m 06s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c619bdbd5f114d6ba02972e83fb39881 : SUCCESS in 1h 20m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4815bdd6c0f94c3f9ee53555c0fbcd3f : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bf70c6eb64e44378986c1fc4fc075c3 : SUCCESS in 7m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2a9004c75214a45951265c85fe42dfb : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/baf7b52380cc46abbf7775b31db30936 : SUCCESS in 9m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7f8e7bc6cd74f909b8243702d2b34a9 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb61b77177f48fdaa05c602b868170b : SUCCESS in 9m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d2b59d52e49e46b9af69a9edd958a2f6 : SUCCESS in 13m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/efee2461f1ed4bf38b96fb7ac8b0e6cb : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1b3131e167d6967c31fae0e0471cdd343815b48a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-13 20:47:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks Ghanshyam,\n\nElod, thanks for checking into this. While I understand that the jobs might work right now, I think this makes sense.","accounts_in_message":[],"_revision_number":1},{"id":"5318dadc3dfe9946bbc775bb472f603247dcfc03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 20:48:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc0e8d55bab261f1972a00e4198f1a127f95bf42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 22:18:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10873391f7a348198756b579f585a0de : SUCCESS in 1h 29m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d069ff71e9504a94bd130bb8dc59d985 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/486aff587f364407a3923b0cb9f50e9d : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7aa8e76e3834afab3a4b39b8fdaaef1 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef4879b90f554f968bc841ae3a88b6a9 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec4f68223047465ca0c66c8a2a0408d7 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f68ceb6397924e16b6565f4092fda287 : SUCCESS in 7m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9c23dd49338d4740a49b622e3904f0cf : SUCCESS in 12m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66e059038fe94f18bb9bb73019e30a3d : TIMED_OUT in 31m 14s","accounts_in_message":[],"_revision_number":1},{"id":"441defb242faf387fad62e3ab0b920e0c281ad24","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-14 06:12:58.000000000","message":"Patch Set 1:\n\n\u003e Thanks Ghanshyam,\n \u003e \n \u003e Elod, thanks for checking into this. While I understand that the\n \u003e jobs might work right now, I think this makes sense.\n\n@Erno the devstack and tempest jobs are fixed back till Pike, I\u0027m working on to revert all the similar changes to get back proper testing for Stein (maintained) and Rocky, Queens and Pike (extended maintenance).\nHowever I also understand your decision, I\u0027m a bit sad, as I think even older branches could benefit from proper testing. :(","accounts_in_message":[],"_revision_number":1},{"id":"1730aebe67d29c4dd0590536c5ef4d78af6b0343","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-11-12 14:27:53.000000000","message":"Patch Set 1:\n\nHello,\n\nDo we want to continue with this patch?\n\nStein is EM and glance will be moved to EM soon  https://review.opendev.org/#/c/762382/","accounts_in_message":[],"_revision_number":1},{"id":"425cba75df9d826528daf7d14bdcb5543c87b65a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-12 16:20:08.000000000","message":"Patch Set 1:\n\n\u003e Hello,\n \u003e \n \u003e Do we want to continue with this patch?\n \u003e \n \u003e Stein is EM and glance will be moved to EM soon  https://review.opendev.org/#/c/762382/\n\nwe can keep it voting as it is passing. If it start failing then we can make it n-v until we have maintainer for EM. \n\nabandoning it for now","accounts_in_message":[],"_revision_number":1},{"id":"82b60ae36e9eae94391e1a727878cbbc1c56dbdb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-11-12 16:20:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b768d9e657ac280945fc8b1b00e8e61eb8691d7f","revisions":{"b768d9e657ac280945fc8b1b00e8e61eb8691d7f":{"kind":"REWORK","_number":1,"created":"2020-06-23 02:06:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/17/737417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/17/737417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/737417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/737417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/737417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/17/737417/1"}}},"commit":{"parents":[{"commit":"149ea050cc58f39eaf9b4660bb8f0271b99d03da","subject":"Staging area not cleared if image is deleted while importing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/149ea050cc58f39eaf9b4660bb8f0271b99d03da"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-23 02:06:04.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-23 02:06:04.000000000","tz":-300},"subject":"Make greande jobs n-v for EM and oldest stable","message":"Make greande jobs n-v for EM and oldest stable\n\nAs discussed in ML thread[1], we are going to\nmake grenade jobs as non voting for all EM stable and\noldest stable. grenade jobs are failing not and it might take\ntime to fix those if we are able to fix. Once it jobs are\nworking depends on project team, they can bring them back to\nvoting or keep non-voting.\n\nIf those jobs are failing consistently and no one is fixing them\nthen removing those n-v jobs in future also fine.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015499.html\n\nStableOnly\n\nChange-Id: I5ef940c6093aa1ccc660b7d82ba136f261086a4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b768d9e657ac280945fc8b1b00e8e61eb8691d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b768d9e657ac280945fc8b1b00e8e61eb8691d7f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
