)]}'
{"id":"openstack%2Fglance~737867","triplet_id":"openstack%2Fglance~master~I44bfefb6eee421e18e5e95a0dafaef0ea4e170da","project":"openstack/glance","branch":"master","topic":"bug/1885003","hashtags":[],"change_id":"I44bfefb6eee421e18e5e95a0dafaef0ea4e170da","subject":"Fix: Interrupted copy-image leaking data on subsequent operation","status":"MERGED","created":"2020-06-24 20:06:58.000000000","updated":"2020-07-06 05:43:06.000000000","submitted":"2020-07-06 05:39:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737867-1594013940992-334cab2b","meta_rev_id":"b43bf60e22480af6c344b91fef5fa4d0a27fa350","_number":737867,"virtual_id_number":737867,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-06 03:14:45.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 05:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-03 18:35:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-03 15:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-03 18:35:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-06-30 15:08:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 07:17:50.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-25 15:18:05.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-07-03 15:15:38.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-03 18:35:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-06 03:14:45.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-06 05:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32010787b425c304f13d3160a52f473a56ab90ff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-24 20:06:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2141ea26d1ad8061331beb8bab0a4acb6070e77f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-24 20:07:19.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"fc27bec4c3708b37e48a297529388415add5d24b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-24 20:07:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e91b6963f78ea449f6dfb252b67e8b542f6204f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-24 20:18:57.000000000","message":"Patch Set 1:\n\n(9 comments)\n\nDefinitely an improvement, thanks for turning this around so quick. I think you\u0027re missing a test case, and I offered my usual annoying list of grammar nits if you\u0027re interested :)","accounts_in_message":[],"_revision_number":1},{"id":"1ab4c1199949b0e46669be8e891038429d48abdb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-24 20:28:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d82649b826642255c8379f17cf3d11c95265c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 23:24:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/371a401583e44dafa3dd29610e0799d9 : SUCCESS in 1h 01m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16f16470d68742da91f330b48a8f698b : SUCCESS in 1h 20m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/359a7645bd934afbacb4d14513aaac1d : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0f729b9039f4395ae1e3f0babd62305 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b632bc0d7fbb4ac2bf5ccd170025447e : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bc29e8f91014befa62c343ca43cdb1f : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f70efadd277149c2ae9770914569f380 : SUCCESS in 10m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e258a76159ba4fccb634ad2af31f1275 : SUCCESS in 8m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d1c9789f439f40a5b146ebd104328c51 : SUCCESS in 14m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8fa519b516e04d43a19863a8f1146655 : SUCCESS in 5m 50s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1fcb5f0d61ee462480ea7cf8e607ac2e : SUCCESS in 1h 29m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6efab45f12c34f7699fab3be3c8e5754 : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4b2ca7015369ad9efbbd502bba112bb7d8895855","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-25 07:16:40.000000000","message":"Patch Set 1:\n\n(9 comments)\n\nDan, thank you for the review!!","accounts_in_message":[],"_revision_number":1},{"id":"6c602ae03eae10f09e2e6ad8ef6f7719a56d630d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-25 07:16:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d715101682207e71684228b10a560d71dcdfc405","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-25 07:17:20.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"f1dcca585ef112bd28cc40e141c2bb5fa8c9536c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 09:16:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/088fde0157194934a1e9b12738f9f81b : SUCCESS in 1h 04m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f6cadec171a48ffa29e3aa56cdb59aa : SUCCESS in 1h 05m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1021f40e442490b956ee4170d39a290 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e601de09596c4c9ca60f211ec355bcba : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90fbf131d9fd49e796233e32529fd791 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4b0afe186234f4ca5979914d7716a4c : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29dddab8873643f28ac4c3ed95b919c5 : SUCCESS in 8m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/da4b36e01a4a417f81e44d411ea04aa2 : SUCCESS in 11m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0b80f874b6234c28ad350ac899b782b6 : SUCCESS in 12m 01s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/98df885ce6fb4193886b3572dc169223 : SUCCESS in 6m 04s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a9c1dddde10e4b8ca92de7ac324716c4 : SUCCESS in 1h 21m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/beb71543fb664dfb8f469a2ed22a1fa6 : SUCCESS in 1h 17m 31s","accounts_in_message":[],"_revision_number":2},{"id":"c5736aec0621541ea3ae22afc0e80704fe8b0692","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-25 15:11:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e66c7620e9eece441e8785cb72c97aa5b1c5c690","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-25 15:18:05.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aff61bfc1569259c21d04524041a2feaa3c54d2a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-25 16:49:50.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"814a67b8cf2dea30df0a9c344adba4364c601ef9","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-06-25 17:19:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nJust couple of minor things. I think the base logic is sound and good enough for the need.","accounts_in_message":[],"_revision_number":2},{"id":"d8c8f5051c63724e5e8e2b9e3aa45887ed93a348","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-25 17:49:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73d2aa0d438d89a0493c94e0542e41b9607fbe59","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-25 17:50:01.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nDan, Erno,\n\nThank you for review!","accounts_in_message":[],"_revision_number":2},{"id":"2e97808f05b3d06d291f280e07f314163447b1e1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-25 17:50:08.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"11973d32683f9ae907fe6f89a9afece320e0bf43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 23:41:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/abb0139a4fef4c0ab2523364b3c7ac1a : SUCCESS in 1h 05m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a5128520e38437d81560289dd6ea4ca : SUCCESS in 1h 04m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9561c50026fd410da5a88b343d589d64 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98431816988e468c90db5535a6eb5449 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2feb4f01c303461186176b71659b35df : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65375d953dd34c4dbb93539c5bab3b63 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5209cb23e9eb489cbe2e65cdc44b8487 : SUCCESS in 9m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f4492b7cc57549839cffb9a7ecc22ea6 : SUCCESS in 9m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7514a2c6b82547888097647227055d06 : SUCCESS in 9m 01s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9d917e9f42fa49af9285d08eeed9ff2e : SUCCESS in 5m 57s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d593acb8ddd54bc895a73969b17c2280 : SUCCESS in 1h 09m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17c856d7cf1a4fb2818890595a4834c6 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":3},{"id":"aeecea59edd84e6794550c9de21a5e1f63006e36","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-30 15:08:38.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"de2237eeb587d2fdf347a650c413c929c480347c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-03 15:15:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"92ca4ad1aa308db612dcd64a2ff27caaf98b118b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-03 18:35:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLooks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"1e5a4eec89d97b73f4468acdeebb2b4e74737fcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 18:35:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f1cdd940215000c41f50a5eb223b94ea46b7265c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 18:59:42.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4043cfe1a2a44e8c8849d025c6f9b0c7 : FAILURE in 16m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6cd90ea95e884cdf9795411922224c52 : FAILURE in 18m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/552e48be44534b07a8d47bedd6739605 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c24d51c5284a1985714de210bbef0f : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35b904afb47440feb4e3837865de8ac2 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c51c1d25610042c6b261dcb39b47f4b5 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c702c2a5fc241fab72bdbd3b02bcfae : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/33338658d4434c199a8d33d943845c51 : SUCCESS in 8m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/82ce9194953f4ca6b062975cbf471e91 : SUCCESS in 7m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4477541ff7724b8f8c7a45df1b18001f : FAILURE in 18m 42s","accounts_in_message":[],"_revision_number":3},{"id":"4dd3c001acd28857f579809b67a269387b07897e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-03 19:14:59.000000000","message":"Patch Set 3:\n\nHitting an issue from the new setuptools.","accounts_in_message":[],"_revision_number":3},{"id":"eb2c44c6c6d4821f440ee13cde8f4876516a20d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 19:39:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/58f3bfddb5bd44e9821e2536b32a2a2f : FAILURE in 19m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69ac52da7ecc4d3888d04f711e3ee499 : FAILURE in 17m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55a7e700d12648eaa7ec30359458564f : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869e09e8562546d5b6d6b4cacb7b20a2 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5c5260a99ce45a283e7fe75b6ee98fe : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63f8b5832f25446db96a2d141f8e51fc : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3698c2d68ab1402394acc8442f03da72 : SUCCESS in 8m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/988a0a72435047a08dcd18015169efbd : SUCCESS in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e77d47dee0e34da1b465db1a607b216a : SUCCESS in 9m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/736bf9c52e6c418b9e06ab06d2d0836d : SUCCESS in 4m 45s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bb6db68ee42944c9843d45a3f6abeee0 : FAILURE in 18m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6237af80fa7e4d6f9fba4570988333c8 : FAILURE in 14m 30s","accounts_in_message":[],"_revision_number":3},{"id":"5ce171815819e8608e0022c3da5d27b641ff92ee","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-06 03:14:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"778afa500ace94c3f9d7ac664e66d59f2f166a54","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-06 03:14:45.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"54609367c87f36744cf999b1c6b306043a8b41a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 04:30:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/00582d5b0bd043028eac52f435f6c2d8 : SUCCESS in 59m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/06182afe0b43450cb37f2b6eb91b52d4 : SUCCESS in 1h 06m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2eba1e1fd452497fba324c90143e3c63 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/487c4c31439f480fb1c4270ddcdcafdf : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb11bb57a8364a73a6297ae73b99f041 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06c70ef839da4e4c8eaebf3f2381e15e : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c29bc9374a4a319c917beec4a7d3c3 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1ca7cd36c834658b2ae886f263fc85d : SUCCESS in 10m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9c34aa35cf9144d8a930044b1c471904 : SUCCESS in 11m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4ef5948f6f9948f88c6fac50f62682d1 : SUCCESS in 4m 09s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/28cf98ce54bc44379ab57fde8f0264b2 : SUCCESS in 1h 11m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c1188c9354c4e60a7d18fd4c67d9e5d : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":3},{"id":"16609b567e9bc3bdb65dd7c340dc9eb922fb1ca8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 04:31:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c402be2e24bb219702de2ea41ea3b10da2424c87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 05:39:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bd0cf6da98a57b801f18eaccc037b8497a6b8979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 05:39:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/31bab569424f4971a1dc2dc57b5dc43d : SUCCESS in 56m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5459adc11c0a43d498cb8fac94dce9e8 : SUCCESS in 1h 05m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09b81c3e67bf4016b128db66c3f78cff : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/954397a8f6f94efb83fe6953c43e08dd : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13d813ee740b42aa85f8a8a792a7daac : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11365d431f824d94b8dd72ea9415504c : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0bbb69bb02446787c429358edccf6d : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8aa9fea6026e48ae9c8ef9d1d6010cfe : SUCCESS in 7m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a80054ae44a4489d88079220348b90fa : SUCCESS in 7m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6502d72ab3ce4c2c9a2c4ec1231b65f3 : SUCCESS in 52m 11s","accounts_in_message":[],"_revision_number":3},{"id":"b3c4e889da3e6d57ea2b913f0ce2cc42c151f4d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 05:40:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d14e1387daba48ae8d3b3770c54c351b : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":3},{"id":"df4439ab9a68ba51d83539d2fd41a03424af476d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-06 05:43:06.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 72510f9ce90970e75a95492c7647fd63ac2636dc","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"22d8f1fcbf4325d79ac18adfe87198c785f1df52","revisions":{"0cad494a398210b026ba91d1efcb166af0dcc6b0":{"kind":"REWORK","_number":1,"created":"2020-06-24 20:06:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/67/737867/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/737867/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/737867/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/737867/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/737867/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/737867/1"}}},"commit":{"parents":[{"commit":"e6db0b10a703037f754007bef6f56451086850cd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6db0b10a703037f754007bef6f56451086850cd"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-06-24 19:44:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-06-24 20:06:12.000000000","tz":0},"subject":"Fix: Interrupted copy-imgae leaking data on subsequent operation","message":"Fix: Interrupted copy-imgae leaking data on subsequent operation\n\nIf copying existing image in other stores fails while staging the data to\nstaging directory due to power, network or any other reason. Then\nsubsequent try may lead to data leaks in stores.\n\nTo fix this, added check to check the actual image size with the size\nof image file present it staging area. If it mismatch then deleting\nthe image file from staging area so that entire image will be staged\nagain.\n\nChange-Id: I44bfefb6eee421e18e5e95a0dafaef0ea4e170da\nCloses-Bug: #1885003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0cad494a398210b026ba91d1efcb166af0dcc6b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0cad494a398210b026ba91d1efcb166af0dcc6b0"}]},"branch":"refs/heads/master"},"46ab0ec607944df6df77e4e5103b6824c18a94ef":{"kind":"REWORK","_number":2,"created":"2020-06-25 07:16:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/67/737867/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/737867/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/737867/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/737867/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/737867/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/737867/2"}}},"commit":{"parents":[{"commit":"e6db0b10a703037f754007bef6f56451086850cd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6db0b10a703037f754007bef6f56451086850cd"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-06-24 19:44:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-06-25 07:13:24.000000000","tz":0},"subject":"Fix: Interrupted copy-image leaking data on subsequent operation","message":"Fix: Interrupted copy-image leaking data on subsequent operation\n\nIf copying existing image in other stores fails while staging the data to\nstaging directory due to power, network or any other reason. Then\nsubsequent try may lead to data leaks in stores.\n\nTo fix this, added check of the actual image size with the size\nof image file present in the staging area. If it does not match\nthen delete the image file from staging area so that the entire\nimage will be staged again.\n\nChange-Id: I44bfefb6eee421e18e5e95a0dafaef0ea4e170da\nCloses-Bug: #1885003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/46ab0ec607944df6df77e4e5103b6824c18a94ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/46ab0ec607944df6df77e4e5103b6824c18a94ef"}]},"branch":"refs/heads/master"},"22d8f1fcbf4325d79ac18adfe87198c785f1df52":{"kind":"REWORK","_number":3,"created":"2020-06-25 17:49:38.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/67/737867/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/737867/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/737867/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/737867/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/737867/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/737867/3"}}},"commit":{"parents":[{"commit":"e6db0b10a703037f754007bef6f56451086850cd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6db0b10a703037f754007bef6f56451086850cd"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-06-24 19:44:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-06-25 17:49:23.000000000","tz":0},"subject":"Fix: Interrupted copy-image leaking data on subsequent operation","message":"Fix: Interrupted copy-image leaking data on subsequent operation\n\nIf copying existing image in other stores fails while staging the data to\nstaging directory due to power, network or any other reason. Then\nsubsequent try may lead to data leaks in stores.\n\nTo fix this, added check of the actual image size with the size\nof image file present in the staging area. If it does not match\nthen delete the image file from staging area so that the entire\nimage will be staged again.\n\nChange-Id: I44bfefb6eee421e18e5e95a0dafaef0ea4e170da\nCloses-Bug: #1885003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/22d8f1fcbf4325d79ac18adfe87198c785f1df52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/22d8f1fcbf4325d79ac18adfe87198c785f1df52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
