)]}'
{"id":"openstack%2Fglance~737868","triplet_id":"openstack%2Fglance~master~Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8","project":"openstack/glance","branch":"master","topic":"copy-image_race","hashtags":[],"change_id":"Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8","subject":"Add image_set_property_atomic() helper","status":"MERGED","created":"2020-06-24 20:08:50.000000000","updated":"2020-07-22 20:18:51.000000000","submitted":"2020-07-22 20:16:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737868-1595449019206-798a8ba2","meta_rev_id":"b3476db343cbadf189cb8f54181a4af5b12d213c","_number":737868,"virtual_id_number":737868,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-17 14:22:41.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"date":"2020-07-20 19:27:34.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 20:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-22 18:47:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-07-17 14:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2020-07-20 20:36:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-22 18:47:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-07-17 14:22:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 18:28:02.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2020-06-26 10:24:42.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-06-29 10:02:50.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-17 14:22:41.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-20 20:36:29.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2020-07-22 18:47:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-22 20:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6cb9f4760ba15c91492c158c84cf63b3d96d1a36","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-24 20:08:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"017fc2b85287b3ef37cc8a0a6a4e65466a50e57c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-24 20:09:10.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"9c0d193e285e87576e0dbaaaf43d4699c474041a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-24 20:13:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9e856616177f33161b0e176a235ac0ff712afb53","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-24 20:13:38.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"f98c45c6baf8d40efc6db8f36a9bc6c46e2e2786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 23:33:53.000000000","message":"Patch Set 2: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b530f41b30b543468e2034ee9d02ae87 : SUCCESS in 1h 13m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eab2d58d06fc4925b942cc1a35da8ad2 : SUCCESS in 1h 38m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c9db2e273e246d8b3a5b7d78432436c : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ece31702ba64405b8b25c252bddf4a9 : FAILURE in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e860f56858374651b0ed7f224a3e0426 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c25c72a624b14d8d9a4f3fd16ba7e386 : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cc303b0f868443aa1227c19ab0a643f : SUCCESS in 8m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98fe81ecafdf4c0883cb3b272850748d : SUCCESS in 14m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e88e0d3bebab44c0b775167fcf72a636 : SUCCESS in 9m 31s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2ad6b021c45d429db45581761a15093d : SUCCESS in 7m 20s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a9f887c82d324846ba2b76687daeb709 : SUCCESS in 1h 05m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7116b45e4cf547aa9049abf9434a6570 : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":2},{"id":"a93e00456cb6836895efe4416a811e0cd4b9976d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-25 13:34:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"05da4f97497d5c0b27ee5487f4855c6e782a8f37","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-25 13:38:54.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"b1df0209c7da67ad18c7d1cf8f180033fc01ffbe","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-06-25 13:52:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)\n\nI would lose the \"autoload\" aspect pretty quickly here, the rest is nitpicking.","accounts_in_message":[],"_revision_number":3},{"id":"41f56766cca3e131d878bd5312ba6d2a86beb357","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-25 14:45:37.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThanks Mike!","accounts_in_message":[],"_revision_number":3},{"id":"9867d82190513f5047ae39d168e67cf9735af5f3","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-06-25 15:53:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9eb556e4c89a58a22ffc6fe0e54a71896239124c","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-06-25 15:55:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d7fb3493e61fdc0bd5af170d49f7dc645dd05d3d","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-06-25 16:05:08.000000000","message":"Patch Set 3:\n\nOK we\u0027ve gone over this in IRC I missed that the \"deleted\" column is the one we are trying to grab here so it shoudl be fine","accounts_in_message":[],"_revision_number":3},{"id":"0c630e5672de368746923a25e5ae0e8beead3dde","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-25 16:14:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"94ad8930b0ef14aa0b7205aa7b1ca5946f2e46c3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-25 16:14:26.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"0ce46269b63a382a66467a4a61d80c8d5bccf850","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-06-25 18:28:02.000000000","message":"Patch Set 4: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":4},{"id":"3196f413af105467254d86302c95f7f6df139bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 23:23:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6441622a833f4f3196c2104413d1afc8 : SUCCESS in 1h 05m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eeb282b9988a4fefb710da9c88db2407 : SUCCESS in 1h 14m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b48ba7753d064565a504c9742753ae44 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffa762228a2f49b9980100c30455505d : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b192d5e8ce14e1e88a3a3326324e7ff : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e27444a55a9452cbcb277b7a22eb263 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c65b10e4a5df4b6fbbec4549ba07b935 : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/049f972febd745eaba39e93a1583b41e : SUCCESS in 8m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/99c767cba8ca4d1b9026547ee1eed08d : SUCCESS in 8m 47s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/29d3dca8f8c44b0d9e5aafaec20aaf14 : SUCCESS in 5m 16s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/30fe7a2f24414b2096224ec014e9fe77 : SUCCESS in 1h 06m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8904ed2fbe14454983d30eb30f5025b9 : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":4},{"id":"3b6735927e2e18080ccc2315eb8da2f4a4287199","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-26 10:11:13.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e904c977143ba85fb9a5530de159df316514dc81","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-26 10:33:32.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"2d703060d5bf38fef4410a8b1277b162a45121e1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-26 10:33:52.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"1d29eaa17066686bee26014a886a160916d0e6d4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-26 13:32:34.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f9f8599d62688ff41673fb91e4c1bcaa6d4ddeea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 13:47:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/22e41c662dae4882a1946eb4dc89032b : SUCCESS in 57m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cb760ec68a7a4a0699147dcedb1ecedf : SUCCESS in 1h 06m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/746455a35430486b99d846ee46aa09b5 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f7bd9c1bc7442498e51913c6c85b9da : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9e6183e959e4ea3bd61cb6cf3fb5aa0 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/309a7f9cdf0d4b80af6222de45b84781 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e15ca51b9245d087c88168b7993704 : SUCCESS in 8m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1118e723846646e585805e77b73511c4 : SUCCESS in 10m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ae7629b4f97b4ab9b98775c14721e62f : SUCCESS in 8m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5cf473f9172b428ab02a008d12f3cd90 : SUCCESS in 5m 03s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c7b49af172a04a50868501a56f0745dd : SUCCESS in 1h 11m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec41cdb7fa63487d9c7c0be829876d71 : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":5},{"id":"3846a4dbfbb8d9f7002b4490c7a86116c5d7f00e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-26 14:11:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6563e74ef781389eb6d8e597a2b05a6c6a7ca9d8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-26 14:13:06.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"0b3d27970942b642fc0220ad352e89e6c5ab61ea","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-26 16:40:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6472dae51085b31c9357025a33af0ab7aae66c56","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-26 17:00:58.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"f2db4c0ff9779206374ee04fa1f16740e867c857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 20:53:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/eece6c27b1954d9bbe928b272343f125 : SUCCESS in 1h 06m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6e5aa223ab44207b4d74854ad0b25e5 : SUCCESS in 1h 17m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dd7cc38ee0141b38cca74e4172fee81 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d79c592b1c8e4b51aee7f610612169dd : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/942c62ce14af4551a4eefd9e82d14a6a : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39324ab876c94b29975dd321e95e0ff9 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0aea56c6d3741239ba6c75d1bb58169 : SUCCESS in 7m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d593148dc984b2580758e76d9053303 : SUCCESS in 10m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c242400bd80f4870b50ec4281a816cb0 : SUCCESS in 13m 23s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/705137e315d74a18b20b5f9e0f7495a1 : SUCCESS in 10m 09s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/14236ae5edc0459f9ecc434266bfb2e1 : SUCCESS in 52m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a512320b12624627a5d61efe9291659b : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":7},{"id":"872af2f8b1b007f41650ca854e602a95400e0b98","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-29 10:02:50.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+1\n\nThank you Dan!!","accounts_in_message":[],"_revision_number":7},{"id":"7d63ef5975c2eed125b32f5fac402770f5f7f412","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-02 05:05:35.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"2c10fd8f1f49df1333c54883f9a55fd4052a2e04","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-02 05:06:10.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":8},{"id":"0d19cc212b91d1c7c4349a13dbc58f5e393280fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 06:23:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/dc2e51d2cce84e2aa356ae59ef61bf8c : SUCCESS in 57m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7e4eac5b6087492f9af880d2ca3059bb : SUCCESS in 1h 17m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a746e4f25f274fa1b6677e8db3b5cf03 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d9c43ce43824ce892ba19d215204654 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbf895abf3674e48ba93104d9c006970 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb21ce39eaaf421fa4e4111cfe67da26 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61471aaa69f34e5c8e57ff77c570e7ca : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/10592c17748a46a9b2ce71fc94494390 : SUCCESS in 7m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d321c433f75b4155907767f4bed67577 : SUCCESS in 9m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2f443ca87c1c48df851d269a0a960525 : SUCCESS in 4m 08s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/819fbb4b34544df5a7004124292e477d : SUCCESS in 1h 13m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f528350b5544772b2719c4027eb8a68 : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":8},{"id":"3741811bafcd7bae30b898cc8bbad789d5be500f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-03 18:02:47.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"1b0888e5b7847ee7bf614c5c14f51d87ed5f1fd9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-03 18:03:09.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":9},{"id":"e3c18b54efbc070d078cef3e9454da10daae6b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 18:25:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/88d758dbe36c433bb11d1d6ee174ad06 : FAILURE in 21m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7e5ef0064cee4044a1e32dcd7ab77179 : FAILURE in 15m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7269e87bba284c56a326a19cd4efeebf : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cb33f99224548fd951cf3ec87231842 : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/520f4f459f884458bbb9e05556de65e2 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7b73393a2f14e4a9feba6f8f644580a : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aafb3a2a6be54b3d8bed7a9a15d75586 : SUCCESS in 9m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5e9b9e4f78048b0a68cbc4fdbdd8025 : SUCCESS in 12m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fd943f1bbdc0409abbbe47a217d1c10d : SUCCESS in 11m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d755e1d840814394a4a03f1f7483f00b : SUCCESS in 6m 46s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/942859fff8284bc78c0c88ff3b01620d : FAILURE in 13m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69cb775c597a46fda24e86e4f1db766f : FAILURE in 18m 21s","accounts_in_message":[],"_revision_number":9},{"id":"9de76e2adcf997477009b3531a9316ca867b7569","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-15 05:20:19.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"1c230d25a97183f8b4abf4cb0796167991cb8de7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-15 05:20:41.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":9},{"id":"52643272a12cdb5cd47ff8143eac4bcc99c82338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 06:32:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/00afcb112fa74a9a802ae3f8fdd0f5e3 : SUCCESS in 1h 05m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17ac5811df164a22b389148034ef2194 : SUCCESS in 1h 06m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1183999aa2c40a6974a83359d19bd5e : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a5c8972e229491c80d951abfb5dd866 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb3cb87d1ddc49ca84749f76e9f92231 : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/176612714708406b8aa779a2b5ece078 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad7b2aeb75842feb638e9a50c7427fd : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac4668dd2bd54c60acddeb9187c1bb29 : SUCCESS in 8m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3ceb7b62a2ec47d4841030b40144f162 : SUCCESS in 7m 08s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/248d285856a648298a4854bf56ecd344 : SUCCESS in 4m 53s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bb740d2bc6ff4875b84dec139125df2a : SUCCESS in 58m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ed7f55bb0754beb9d0973222f473356 : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":9},{"id":"db1e152499098d7324ad98c12e07fc02667eca9c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-17 14:22:17.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"edeb82fc4dcd1060db51316a05b11d3f750cc827","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-17 14:22:41.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":10},{"id":"71de4f64f9c53dcb1f82a81292817b5460d2461f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 20:07:47.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/87b13b846f2e4c669ab0b7be17a3d5c2 : SUCCESS in 1h 08m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ae6af2983c4481eb8bc84ce157827a4 : SUCCESS in 1h 19m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99bc2e2652d54dc79e7de71ac70f9334 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/643af7962e5c46a7b6a4f4562a73ba04 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bdb26af7ee64af2bb9e0953398a1f8a : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0b1ff6eaf374a5e984ae1c42d8d5ca8 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a112cd3d6ed642dd87e59bca9e2a3cac : SUCCESS in 10m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0e5a072d850e43b9854f4717804a9019 : SUCCESS in 9m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/013f0232565f4f9fb8bb311f7a340926 : SUCCESS in 7m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2377a9a17c18451d8e2a6e1a9014cfe9 : SUCCESS in 5m 26s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/91ac3cbc3aed4c439f13922cb57b1f6b : SUCCESS in 1h 15m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf50eb7414d14d728da150f7668385c0 : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":10},{"id":"29ad5da8a7e9e6811a548ff0314ddc479b2b912c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2020-07-20 18:09:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"fae9369f8bd819860717bb9d7ce80ce3757fdfc0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-20 19:27:34.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c63837bddd59d83afc42100024a7a8847f036d12","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2020-07-20 20:36:29.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"23f1e5bf3b4cb8e2d243ed8e5f15410946f5071f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-22 18:47:49.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"9e988f35201c84a87461edb5576b83e4ab53823a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 18:48:04.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"a3495f4917058e40012964fbc282233cac73d7dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:16:58.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5b6cc4fd9d4c496daccf2a6c8f8d934c : SUCCESS in 1h 09m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a7a3e5ff95134ac197fe043cd8ce64d1 : SUCCESS in 1h 11m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b5bee704d964ea1a9bb73763dd76bc7 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f09c7912a6348c48b4f5ec86a505c81 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d09999b53c9a4b9ab5d1c17905afc1bb : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f9b7fc668624f7fb4d642e18dbc5ac5 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89c17b54354457e8fbde0d725242cc9 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be16313777cc42a9910247477324e9c6 : SUCCESS in 10m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ce4a5b3790ac46c8989b30508000ca76 : SUCCESS in 9m 52s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/e8fd5e9dafab4ff5b119101529e83025 : SUCCESS in 1h 23m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8185c337a8464915a797e808a48a4724 : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":10},{"id":"776b12c83f3c70411a7dc7de8cd6f7259e30c05c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:16:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"58a666123d739b97076863093e97571ff7149504","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:18:51.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371e9af36a6549cba67cf981547423ee : SUCCESS in 59s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"2a51843138e27071bf84269f6b2a601b3ba9978f","revisions":{"bf7745b5370493f374098294ae61aadef35842ed":{"kind":"REWORK","_number":1,"created":"2020-06-24 20:08:50.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/737868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/737868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/737868/1"}}},"commit":{"parents":[{"commit":"c9c98727690702914784898f999a449f89100aba","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c9c98727690702914784898f999a449f89100aba"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"subject":"Add image_set_property_atomic() helper","message":"Add image_set_property_atomic() helper\n\nThis adds a new DB API method to atomically create a property on an image\nin a way that we can be sure it is created once and only once for the\npurposes of exclusion of multiple threads.\n\nChange-Id: Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8\nRelated-Bug: #1884596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf7745b5370493f374098294ae61aadef35842ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf7745b5370493f374098294ae61aadef35842ed"}]},"branch":"refs/heads/master"},"75d9b3e88de1b1f23c7bca4e0e77bdec27da1247":{"kind":"REWORK","_number":2,"created":"2020-06-24 20:13:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/737868/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/737868/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/737868/2"}}},"commit":{"parents":[{"commit":"c9c98727690702914784898f999a449f89100aba","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c9c98727690702914784898f999a449f89100aba"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:13:08.000000000","tz":-420},"subject":"Add image_set_property_atomic() helper","message":"Add image_set_property_atomic() helper\n\nThis adds a new DB API method to atomically create a property on an image\nin a way that we can be sure it is created once and only once for the\npurposes of exclusion of multiple threads.\n\nChange-Id: Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8\nRelated-Bug: #1884596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/75d9b3e88de1b1f23c7bca4e0e77bdec27da1247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/75d9b3e88de1b1f23c7bca4e0e77bdec27da1247"}]},"branch":"refs/heads/master"},"7f25c10937ba68773e38d9e02657fb556b29213c":{"kind":"REWORK","_number":3,"created":"2020-06-25 13:34:59.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/737868/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/737868/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/737868/3"}}},"commit":{"parents":[{"commit":"c9c98727690702914784898f999a449f89100aba","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c9c98727690702914784898f999a449f89100aba"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 13:35:03.000000000","tz":-420},"subject":"Add image_set_property_atomic() helper","message":"Add image_set_property_atomic() helper\n\nThis adds a new DB API method to atomically create a property on an image\nin a way that we can be sure it is created once and only once for the\npurposes of exclusion of multiple threads.\n\nChange-Id: Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8\nRelated-Bug: #1884596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f25c10937ba68773e38d9e02657fb556b29213c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f25c10937ba68773e38d9e02657fb556b29213c"}]},"branch":"refs/heads/master"},"af69b3a12a4360418c5b00f643d7a3f27d98b9d6":{"kind":"REWORK","_number":4,"created":"2020-06-25 16:14:05.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/737868/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/737868/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/737868/4"}}},"commit":{"parents":[{"commit":"c9c98727690702914784898f999a449f89100aba","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c9c98727690702914784898f999a449f89100aba"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 16:14:12.000000000","tz":-420},"subject":"Add image_set_property_atomic() helper","message":"Add image_set_property_atomic() helper\n\nThis adds a new DB API method to atomically create a property on an image\nin a way that we can be sure it is created once and only once for the\npurposes of exclusion of multiple threads.\n\nChange-Id: Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8\nRelated-Bug: #1884596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/af69b3a12a4360418c5b00f643d7a3f27d98b9d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/af69b3a12a4360418c5b00f643d7a3f27d98b9d6"}]},"branch":"refs/heads/master"},"e2efdea631d3ba7da1c0e70f3c24dc5a50cc002d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-06-26 10:33:32.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/68/737868/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/737868/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/737868/5"}}},"commit":{"parents":[{"commit":"e6db0b10a703037f754007bef6f56451086850cd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6db0b10a703037f754007bef6f56451086850cd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-06-26 08:49:49.000000000","tz":0},"subject":"Add image_set_property_atomic() helper","message":"Add image_set_property_atomic() helper\n\nThis adds a new DB API method to atomically create a property on an image\nin a way that we can be sure it is created once and only once for the\npurposes of exclusion of multiple threads.\n\nChange-Id: Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8\nRelated-Bug: #1884596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2efdea631d3ba7da1c0e70f3c24dc5a50cc002d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2efdea631d3ba7da1c0e70f3c24dc5a50cc002d"}]},"branch":"refs/heads/master"},"5f99919b844d10bb80d2f84481f5d4c4c938c589":{"kind":"REWORK","_number":6,"created":"2020-06-26 14:11:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/737868/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/737868/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/737868/6"}}},"commit":{"parents":[{"commit":"e6db0b10a703037f754007bef6f56451086850cd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6db0b10a703037f754007bef6f56451086850cd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-26 14:10:56.000000000","tz":-420},"subject":"Add image_set_property_atomic() helper","message":"Add image_set_property_atomic() helper\n\nThis adds a new DB API method to atomically create a property on an image\nin a way that we can be sure it is created once and only once for the\npurposes of exclusion of multiple threads.\n\nChange-Id: Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8\nRelated-Bug: #1884596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f99919b844d10bb80d2f84481f5d4c4c938c589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f99919b844d10bb80d2f84481f5d4c4c938c589"}]},"branch":"refs/heads/master"},"bdebef8aeb347cc008c9830f2ec4ed784ea5a697":{"kind":"REWORK","_number":7,"created":"2020-06-26 16:40:13.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/737868/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/737868/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/737868/7"}}},"commit":{"parents":[{"commit":"e6db0b10a703037f754007bef6f56451086850cd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6db0b10a703037f754007bef6f56451086850cd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-26 16:08:03.000000000","tz":-420},"subject":"Add image_set_property_atomic() helper","message":"Add image_set_property_atomic() helper\n\nThis adds a new DB API method to atomically create a property on an image\nin a way that we can be sure it is created once and only once for the\npurposes of exclusion of multiple threads.\n\nChange-Id: Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8\nRelated-Bug: #1884596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdebef8aeb347cc008c9830f2ec4ed784ea5a697"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdebef8aeb347cc008c9830f2ec4ed784ea5a697"}]},"branch":"refs/heads/master"},"1a0615a550c3db6ba314b9511c5e02a5815b7f91":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-07-02 05:05:35.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/68/737868/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/737868/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/737868/8"}}},"commit":{"parents":[{"commit":"fadf546a75e45d5382f7f68fd2888d75f780e3d2","subject":"Merge \"Check authorization before import for image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fadf546a75e45d5382f7f68fd2888d75f780e3d2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-02 04:33:52.000000000","tz":0},"subject":"Add image_set_property_atomic() helper","message":"Add image_set_property_atomic() helper\n\nThis adds a new DB API method to atomically create a property on an image\nin a way that we can be sure it is created once and only once for the\npurposes of exclusion of multiple threads.\n\nChange-Id: Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8\nRelated-Bug: #1884596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a0615a550c3db6ba314b9511c5e02a5815b7f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a0615a550c3db6ba314b9511c5e02a5815b7f91"}]},"branch":"refs/heads/master"},"c243786c7e596a054cc411e70bdef94360350a2d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-07-03 18:02:47.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/68/737868/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/737868/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/737868/9"}}},"commit":{"parents":[{"commit":"9840954bfad722636a0144ef46fe22cd6bbdfc8c","subject":"Merge \"Add a test to replicate the owner-required behavior of copy-image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9840954bfad722636a0144ef46fe22cd6bbdfc8c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-03 05:43:45.000000000","tz":0},"subject":"Add image_set_property_atomic() helper","message":"Add image_set_property_atomic() helper\n\nThis adds a new DB API method to atomically create a property on an image\nin a way that we can be sure it is created once and only once for the\npurposes of exclusion of multiple threads.\n\nChange-Id: Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8\nRelated-Bug: #1884596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c243786c7e596a054cc411e70bdef94360350a2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c243786c7e596a054cc411e70bdef94360350a2d"}]},"branch":"refs/heads/master"},"2a51843138e27071bf84269f6b2a601b3ba9978f":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-07-17 14:22:17.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/68/737868/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/737868/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/737868/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/737868/10"}}},"commit":{"parents":[{"commit":"7ed2cc9e14276055e98b6b925eddba7c495f87ac","subject":"Merge \"Stop to use the __future__ module.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ed2cc9e14276055e98b6b925eddba7c495f87ac"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-24 20:06:38.000000000","tz":-420},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-16 12:15:55.000000000","tz":60},"subject":"Add image_set_property_atomic() helper","message":"Add image_set_property_atomic() helper\n\nThis adds a new DB API method to atomically create a property on an image\nin a way that we can be sure it is created once and only once for the\npurposes of exclusion of multiple threads.\n\nChange-Id: Ifdb711cb241ef13eccaa5ae29a234f2fe4a52eb8\nRelated-Bug: #1884596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a51843138e27071bf84269f6b2a601b3ba9978f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a51843138e27071bf84269f6b2a601b3ba9978f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
