)]}'
{"id":"openstack%2Fglance~738101","triplet_id":"openstack%2Fglance~master~I284271a458117b753a88b847598ca10ce4fb94fa","project":"openstack/glance","branch":"master","topic":"bug/1884587","hashtags":[],"change_id":"I284271a458117b753a88b847598ca10ce4fb94fa","subject":"Add a test to replicate the owner-required behavior of copy-image","status":"MERGED","created":"2020-06-25 20:33:18.000000000","updated":"2020-07-02 17:50:02.000000000","submitted":"2020-07-02 17:47:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738101-1593712074743-72369fb8","meta_rev_id":"19841c7ce0b0e9fa6701c28d6e093dd53f36029a","_number":738101,"virtual_id_number":738101,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-01 15:42:23.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 17:47:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-01 15:41:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-07-01 15:41:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-07-02 15:14:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2020-07-01 15:41:55.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-07-01 15:41:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 20:35:12.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-06-30 14:52:28.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-01 15:42:23.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-02 15:14:34.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-02 17:47:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad84e9ef9434d4b85202beef9eeb39ae7fd317c4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-25 20:33:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d742ad976f76533ba22102fbd5ad34bb2c5cf405","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-25 20:33:44.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"4b00bda873e6d52df48e0751708df5aafaf6ede0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 23:56:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2469247987c14145b6474012c1971531 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0853087892e4341b23e807daffd946e : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7bd3128ef13443eb39a1e91800d3bee : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/816854c23acc417ba00164ec61ba75f2 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5362bee0e24497085b06b0fd33f32c2 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c177ca17385d4ac4b61ec67ca21d8d2e : SUCCESS in 8m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cf6a0dd9b30b4a8e84ca71c89f7305a3 : SUCCESS in 6m 58s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7efc99cda63c4c468a1f24d7a0a51ca3 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"753ff2cbb5acb297b20b4c6077d4b5c87b4f3be8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-26 07:55:13.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"1800385a71726c2b658f9a83511489b0ef394f4b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 14:52:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM.","accounts_in_message":[],"_revision_number":1},{"id":"137e08379ece4e70cded4a523f999caf1000849e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 14:57:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdfc7b032875ccc15bb99edd2a5406ce9126448b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 19:41:23.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"024087ec630e2166b7f59647b32b6f32b18d754c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-30 19:56:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d1934d137006cd51bb482f2283602842d8c5899b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 19:57:11.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"82cd54ae29224d3276ea3ed35865d0ec835a4d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 20:11:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04bfc51e7beb49b29bfd731862008c72 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e89a6a08d5ae4d5e999723b946deb013 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93945fca6be4405288e934971172d06f : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99e2dab128cb4be697efb1a84510b7e9 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5cb6dda90d0404b96dfffa58c496df2 : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/441ab91955964eff99b4af400800decd : SUCCESS in 7m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e1fb037f40d4bb68f814a0fe65e2124 : SUCCESS in 9m 10s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c08207a28e3e4233b9887de735f6b4a3 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e730d682b67ce15e4b1a447109652444c1ced532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 21:12:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3bdde62a50f0ab3ee2bc2008a8a138d4669adaf5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-30 21:44:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4bc45a5485b5ae0aff32f4a33bdb098ebd42fb83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 21:44:59.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"3304ccafe99a2a3879b72188f1e4d1bf1af63d44","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 21:45:59.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"160c492521bb181eb6da8eefee655ac2d1c49546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 22:18:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86cb1a5457074d0db21eb2c2a758fcc5 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66e989ec6b0245f2bd3e89b9d56890f7 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58f01d19734b4ba98ca8fabf3532d913 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed31771483ec4c38b9636a69306a288f : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff09e1dacf264c638c54ec1384447ca7 : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/77ca7cfdddff4a9ebbc7371b07134df6 : TIMED_OUT in 31m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b3e02a24c9234c7abd7173040f24a64d : SUCCESS in 7m 36s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/488591c9a7854b0087acea4192d1c337 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"6387aa4084240aeaa15658b7f06dfd38811b977b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-30 22:36:32.000000000","message":"Patch Set 2:\n\nrecheck something infra only unit changed but functional failed","accounts_in_message":[],"_revision_number":2},{"id":"b585544c286e1dd3985167cc513b7495d5a9473d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 22:36:57.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"e6413a07cd2041e129801c6c01e82ec7bd4cfa70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 22:49:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d7817acab2d4a2d94f3346d657b1517 : SUCCESS in 8m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a433265c47d74e678fc1d8b479d333e2 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6681fefa1a14f42bc0097c622dc1b92 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/024d468b798b480b911d8c6825b35b98 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a0651b5a38471697394a7613918c99 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/71a95445e6fa4752bf896c050c74ed05 : SUCCESS in 10m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/795d0f64dc934d4886bdc5d89a3ff607 : SUCCESS in 9m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ef721ba52eb44550b0630aefbe503e64 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":2},{"id":"5546148e77f08986c81b8e417f334e79e7a9679c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-01 15:41:55.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7a98b6f6669d092f90f632577e802b2aef14f3bd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-01 15:42:23.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"56a6ea0e3033218ffbbb1de9e3081b0059b9be1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 15:57:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc235d81f1cf475cab6d8b944853c15e : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2edc03e4b6064af8ad4d7856b12c85f1 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bf6090add1848c8ad0899fff4779a86 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adeba6a2ff1d4a01bc1ac326ef0ff456 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0f438289aea41aaa81bf90d96257d37 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e4d9e13a3624aedb7fa54cd0b98b4a6 : SUCCESS in 8m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6e0cf7e29fa5436d9f4717bd36f4a3d6 : SUCCESS in 14m 28s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1e6eb2f0d57242d594b2cd76103161a3 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":3},{"id":"9f296c88187bb6f0b5f4104010174b08835ec3d1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-02 15:14:34.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"73d8ef4300f0a761550095b5fc596bc5087a9520","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 15:14:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2f1999e0e41f1a3b750626c4dd4486ed3fd5f1e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 17:47:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/562a6e8926f34f4aa8c49a367777caa5 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ea166e8cffe4ca0b2af5150a873e28b : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd14f4b3f9bf484b934a6fca2898e467 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc91e76069314956a1fd435a9c86ee4e : SUCCESS in 11m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3099c6a60142458a9f621b23cc5435 : SUCCESS in 10m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8bae658efded4d3098fd76feebf1c77b : SUCCESS in 12m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c93e99eeea4146429ec1324685dfd2d3 : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":3},{"id":"0ed830e741db70364db0d0d0c1635897bd4b6b41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 17:47:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4f03bc8e23bc6beddf5faac7ee11b1265d2c8fbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 17:50:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da8c9e864aeb4b9ea36184feabb84b63 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f450b9cf0314afeb9aa64ed687218b92b6a49321","revisions":{"34ae329c45a2cfa209b33bc00ab04dd9cdf196c9":{"kind":"REWORK","_number":1,"created":"2020-06-25 20:33:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/738101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/738101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/738101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/738101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/738101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/738101/1"}}},"commit":{"parents":[{"commit":"165fb38d18c7bca5d9ef9100949a694f51171203","subject":"Check authorization before import for image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/165fb38d18c7bca5d9ef9100949a694f51171203"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 19:17:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 19:21:27.000000000","tz":-420},"subject":"Add a test to replicate the owner-required behavior of copy-image","message":"Add a test to replicate the owner-required behavior of copy-image\n\nRight now, image_import(method\u003d\u0027copy-image\u0027) requires you to be the owner\nof the image. This adds a test that validates that behavior so we can measure\nthe difference of the next set of changes to modify that.\n\nChange-Id: I284271a458117b753a88b847598ca10ce4fb94fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/34ae329c45a2cfa209b33bc00ab04dd9cdf196c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/34ae329c45a2cfa209b33bc00ab04dd9cdf196c9"}]},"branch":"refs/heads/master"},"6d1a457f96da2f520856cbce1a2d476c7e3a2419":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-30 21:44:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/738101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/738101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/738101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/738101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/738101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/738101/2"}}},"commit":{"parents":[{"commit":"c930638fcf58b43d53903cfc30e06fd6919bdad6","subject":"Check authorization before import for image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c930638fcf58b43d53903cfc30e06fd6919bdad6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 19:17:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-30 17:24:45.000000000","tz":-420},"subject":"Add a test to replicate the owner-required behavior of copy-image","message":"Add a test to replicate the owner-required behavior of copy-image\n\nRight now, image_import(method\u003d\u0027copy-image\u0027) requires you to be the owner\nof the image. This adds a test that validates that behavior so we can measure\nthe difference of the next set of changes to modify that.\n\nChange-Id: I284271a458117b753a88b847598ca10ce4fb94fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d1a457f96da2f520856cbce1a2d476c7e3a2419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d1a457f96da2f520856cbce1a2d476c7e3a2419"}]},"branch":"refs/heads/master"},"f450b9cf0314afeb9aa64ed687218b92b6a49321":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-01 15:41:55.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/01/738101/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/01/738101/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/738101/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/738101/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/01/738101/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/01/738101/3"}}},"commit":{"parents":[{"commit":"fadf546a75e45d5382f7f68fd2888d75f780e3d2","subject":"Merge \"Check authorization before import for image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fadf546a75e45d5382f7f68fd2888d75f780e3d2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 19:17:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-01 15:41:26.000000000","tz":-420},"subject":"Add a test to replicate the owner-required behavior of copy-image","message":"Add a test to replicate the owner-required behavior of copy-image\n\nRight now, image_import(method\u003d\u0027copy-image\u0027) requires you to be the owner\nof the image. This adds a test that validates that behavior so we can measure\nthe difference of the next set of changes to modify that.\n\nChange-Id: I284271a458117b753a88b847598ca10ce4fb94fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f450b9cf0314afeb9aa64ed687218b92b6a49321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f450b9cf0314afeb9aa64ed687218b92b6a49321"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
