)]}'
{"id":"openstack%2Fglance~738102","triplet_id":"openstack%2Fglance~master~Ief3440759fda4bd90979caa79641770cb022b7da","project":"openstack/glance","branch":"master","topic":"bp/copy-unowned-image","hashtags":[],"change_id":"Ief3440759fda4bd90979caa79641770cb022b7da","subject":"Refactor TaskFactory and Executor to take an admin ImageRepo","status":"MERGED","created":"2020-06-25 20:33:18.000000000","updated":"2020-07-14 07:08:26.000000000","submitted":"2020-07-14 07:04:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738102-1594710265072-72fcc681","meta_rev_id":"bfb056825d5fe6605a1260113f5d741f50ad1d40","_number":738102,"virtual_id_number":738102,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-10 16:00:40.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-14 07:04:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-10 15:59:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-07-13 05:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-07-14 05:35:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-07-14 07:08:26.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-10 13:35:28.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-10 16:00:40.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-14 05:35:09.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-14 07:04:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"455a5cea5ba36744203d559ce9b2d9c39f7033a5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-25 20:33:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"847c859efd71bf25802beab69c7eefce8dd6a450","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-25 20:33:59.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"1e98edcb5e962529f96e8f41d985f3514c32cf38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 00:50:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f1543fd355b6481d8c5eb08390f9248a : SUCCESS in 1h 01m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/176478950981489ca6af8f3ef7dda552 : SUCCESS in 1h 04m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57e3630e9bc346aa9a5365718ef1118f : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c7776fb8254b85a4efd778338e04bc : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/424b7d2cb668431fa78fb4d63a005d8f : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa4f4d3c1603450b9bcc157e3ff5f30f : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd0d6817096d4560930ad084277cb2e7 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/67e0d6a40c634b118578438337dca8ac : SUCCESS in 10m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2614eb4b15aa490f97f2bc2f4c066c11 : SUCCESS in 9m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/15b0fca6f751477eaedfb94938caeace : SUCCESS in 5m 01s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/59ff8ac752874e40b7c7226144ce911e : SUCCESS in 58m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/432640bb96684588a84f31620e3eef36 : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":1},{"id":"62d2840895a43f4a666bad34f7f561f606f4891c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-26 02:43:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74a9c0e182e010aadcec7e7f13b9d862961e08d0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-30 21:44:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"fe75354936faf194e2c5d8455d830949c03f00ab","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 21:46:28.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"87a38f0a215bd3f1375213ccfc62c3f6167ce306","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-30 21:47:30.000000000","message":"Patch Set 2: Workflow-1\n\nStill need to add tests for this","accounts_in_message":[],"_revision_number":2},{"id":"a5b603a5a8e2c4b6d211145567a81a417c01dec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 23:04:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0b4be588389c473c8d35e588a718cedc : SUCCESS in 1h 13m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/efba4e16de904cc08747e5a57985df38 : FAILURE in 1h 18m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d89d30684544a60b059b1b229bdad6b : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df0ed5c058f24dc38df7a35e465ee7d7 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3dcba4e84914c63ae8a367e3610d4a4 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee9528ec8d02410b8f8fb0c27bcc1e78 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af86d02af50c44819ffbf7cfdbc7b178 : SUCCESS in 9m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/368862c0ed764432816624613cb5ae39 : TIMED_OUT in 31m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4400742128734fbe88ad61e1002260a9 : SUCCESS in 17m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/23c1b5f489884352b2c9f8873a0f6bd6 : SUCCESS in 4m 27s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/605a90231c1b4dc88acb4d19e0435bb3 : FAILURE in 23m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25ec710583164f6d948cc84ea091d963 : SUCCESS in 1h 06m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d24b5828c5f4c52c87decf44e29763392cc93266","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-01 15:41:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"632d09d71d1a6dfbeb88613d7ea3c7bf3a983942","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-01 15:43:08.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"847fa16cd5609f6e30eab8ea8371c54047df3808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 17:05:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/99a55c2fb9ab47968b9b0449e929c327 : SUCCESS in 1h 03m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c9ef40623cdd40b1b0fe96ce46ea0ed8 : FAILURE in 1h 18m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35e898ef66f94d88a70bd7ad698ca397 : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b0e9356e8644f50b104dd452a883a48 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/760293ded296458e99d2840e6ed646e0 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e59d1f38abda4369aeed970c3d0feeab : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc3b76bfda3b4b869f10b5446b65cf96 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6459de89c6ed4d96a6f3b19d41769abf : SUCCESS in 9m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/14118e1a62594d9db5e4993fa8856acd : SUCCESS in 11m 10s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/38234afac1a6420facc5c3ef51485521 : SUCCESS in 7m 13s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8c9df80d3abe46e9b0adc606aae83a9c : SUCCESS in 1h 17m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ba8c5ace65b471b9cbf02b9ae0e4b01 : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":3},{"id":"b7d5f26dd758bd75d27aa3a865fb6340e31eb896","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-02 04:47:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f1c3fb675866d186320a19a79a176cfe3954f7d2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-02 04:48:12.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"1898e2612ce10d9b64c9b84d5784cb4ff50db5a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 06:05:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c5b7028c2d42499f8dc6e4d30461a839 : SUCCESS in 1h 07m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6feaf3058e134ccca682e2c52baf3225 : SUCCESS in 1h 17m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3e278aa76ad45568cf2b9b313fae641 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9079d02c13534e20a01155e08e9ef222 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3be36b08e97476c8d2b273a27d6b119 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1a368373087402aa90e601b08a03f71 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dfeb395f6ca4ecea9fbb8a6cc493a90 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d914fd36aa7945499425b1814d649155 : SUCCESS in 7m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/572d3b3cf089400c9590fd819ab6e67c : SUCCESS in 8m 05s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7f0879eb329142409f497a1367bebade : SUCCESS in 4m 49s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/04fcdd3091524881962f000564a6fd73 : SUCCESS in 1h 09m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ba0234c29fa4a3d8270400d4bfd668a : SUCCESS in 1h 01m 56s","accounts_in_message":[],"_revision_number":3},{"id":"eb7d9733320cf14f4c58c06ebaae298305ea304d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-08 13:53:36.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\nThank you for specific tests!","accounts_in_message":[],"_revision_number":3},{"id":"781184d94d4a7e020bd4208c4b37677ec8f96711","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-09 14:48:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6acd2a63595888331fb803f754e0ba154a9cab9e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-09 14:50:59.000000000","message":"Topic changed from bug/1884587 to bp/copy-unowned-image","accounts_in_message":[],"_revision_number":4},{"id":"a0a7cbcf9e5b34136101b520ebb41e3485e71665","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-09 18:57:27.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"06c4ed8f6e26639016d595a7a72374f7fbaa4290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 20:43:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c427b5d40b274a999d10ea37e12520fc : SUCCESS in 1h 10m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bb86bf79904f43eca4556d8228d6f24a : SUCCESS in 1h 08m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca1b2385f0d6445ebdbfe4dfe0602546 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e9bf59038b44b878d5f5282fef94936 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e88837f5d30d4c1c850f24d5e126f368 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0473d756c6e040b593425ebedea306e1 : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b971b126aff14d2da44a9608f92ac29d : SUCCESS in 8m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/01334bbfe5f64b61a95eb97e1154da4f : SUCCESS in 9m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/585d6e6fd6fa4737afac942faadb0423 : SUCCESS in 10m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/00663ae592a946caaead928df51517e9 : SUCCESS in 9m 11s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c60fea43962b49ddbc8650613a7504e6 : SUCCESS in 1h 04m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c318e9b5f004946b03e5ae4fc605def : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":4},{"id":"40aad9da3a20626b7e29bcfadbbd2edbe28f551a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-10 13:35:28.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"eb7265424f5b576c3016613c2baad4fb35dd0072","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-10 15:59:47.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"67e148d6a7570f45bbf76a025d4bf07992f8a00e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-10 16:00:40.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"b9c54b330d7191bfa3835eb86c2cff3d87dc1915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 18:07:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5279f322f5394e01841615af72e6587e : SUCCESS in 57m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b51bdf17977c4954850032f6336cd192 : SUCCESS in 1h 23m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55fa748d4bbe4d9992770891d390f83d : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03846c3b6d68434283990752c05a937f : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ca79abe429f48fd9dcd802480700674 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0a7081164d64a169cbefb207518a89b : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f42940f435ff488aa587a171f70b42be : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f68a31e6543a4f979678ec4a3ba63b88 : SUCCESS in 8m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f4cb5c759fb14beebd2b735507f8deab : SUCCESS in 11m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7ad54a9d7b5a4ed1b06d768ef42b842c : SUCCESS in 7m 47s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/89bc6f65db2c48b1bddbae31d072ca42 : SUCCESS in 48m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab3fe72d4e4b419c932f86899f8402d3 : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":5},{"id":"f4960c095fad044cf27a46eac64991df3fa11f71","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-13 05:25:13.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5573a6f636600d87a5d33ecd85abc8599ad939b2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-14 05:35:09.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ed4aa7ffef6b8aecf4865c73c6ef4604b32a4d87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 05:35:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7fdac81c1c43c30eefddc70771d01cb2aa91e759","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 07:04:24.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9191008865d4457ebf5e7a78f6ddbab1 : SUCCESS in 1h 03m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/518f76104f1244aa85e75fd4e7492f64 : SUCCESS in 1h 28m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78ea45cc93824037bbfb95e3bffc51a1 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52ecc9e8a51b481ea7191bc65011f8c7 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfb6435bf662476e8cd7ae1987ac225b : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80cc2acd5d5e4f2cb1170891a76dff6e : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2491ecefe6bb4c65b150f0a1c526f4c3 : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ba05b2effb054f3aba83fc95a9967fca : SUCCESS in 6m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7c72c9c4b9b94b3184f2028d04265adf : SUCCESS in 7m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b34b4d0cf377421b84df93147b336dbb : SUCCESS in 58m 13s","accounts_in_message":[],"_revision_number":5},{"id":"50cf1e72de0ac736a96f5da8f74127f31e069c05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 07:04:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"33dc85f4df2286d5b0ca66213856dd18ea346760","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 07:08:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18689036900043bd9d6bd4c3e34f5867 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"937a70ab831e045bf4b234030f34ac3bb8539467","revisions":{"8530a30f2c030e5cae2b96fbe909845ebdc6fc7a":{"kind":"REWORK","_number":1,"created":"2020-06-25 20:33:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/738102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/738102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/738102/1"}}},"commit":{"parents":[{"commit":"34ae329c45a2cfa209b33bc00ab04dd9cdf196c9","subject":"Add a test to replicate the owner-required behavior of copy-image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/34ae329c45a2cfa209b33bc00ab04dd9cdf196c9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 19:54:19.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 20:33:17.000000000","tz":-420},"subject":"WIP: Refactor TaskFactory and Executor to take an admin ImageRepo","message":"WIP: Refactor TaskFactory and Executor to take an admin ImageRepo\n\nThis refactors the task creation path to allow optional passing\nof an admin-enabled ImageRepo to tasks. This will be used for\nallowing import_image(method\u003d\u0027copy-image\u0027) to update non-owned\nimage metadata during a copy.\n\nI will add tests for this if the approach is acceptable\n\nChange-Id: Ief3440759fda4bd90979caa79641770cb022b7da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8530a30f2c030e5cae2b96fbe909845ebdc6fc7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8530a30f2c030e5cae2b96fbe909845ebdc6fc7a"}]},"branch":"refs/heads/master"},"552c882ecec43ee17e16864f9426c641205178cb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-30 21:44:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/738102/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/738102/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/738102/2"}}},"commit":{"parents":[{"commit":"fa164a88470b976b2ec1b68f36d808e55c0a131b","subject":"Add a functional test for non-owned image copying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa164a88470b976b2ec1b68f36d808e55c0a131b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 19:54:19.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-30 20:42:04.000000000","tz":-420},"subject":"WIP: Refactor TaskFactory and Executor to take an admin ImageRepo","message":"WIP: Refactor TaskFactory and Executor to take an admin ImageRepo\n\nThis refactors the task creation path to allow optional passing\nof an admin-enabled ImageRepo to tasks. This will be used for\nallowing import_image(method\u003d\u0027copy-image\u0027) to update non-owned\nimage metadata during a copy.\n\nI will add tests for this if the approach is acceptable\n\nChange-Id: Ief3440759fda4bd90979caa79641770cb022b7da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/552c882ecec43ee17e16864f9426c641205178cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/552c882ecec43ee17e16864f9426c641205178cb"}]},"branch":"refs/heads/master"},"bc7dd453a080d123aa37be1600e24fbbfb7fccd7":{"kind":"REWORK","_number":3,"created":"2020-07-01 15:41:55.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/738102/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/738102/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/738102/3"}}},"commit":{"parents":[{"commit":"276962a5deff56d5929516524c67f464aa372b6e","subject":"Add a functional test for non-owned image copying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/276962a5deff56d5929516524c67f464aa372b6e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 19:54:19.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-01 15:41:26.000000000","tz":-420},"subject":"Refactor TaskFactory and Executor to take an admin ImageRepo","message":"Refactor TaskFactory and Executor to take an admin ImageRepo\n\nThis refactors the task creation path to allow optional passing\nof an admin-enabled ImageRepo to tasks. This will be used for\nallowing import_image(method\u003d\u0027copy-image\u0027) to update non-owned\nimage metadata during a copy.\n\nChange-Id: Ief3440759fda4bd90979caa79641770cb022b7da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc7dd453a080d123aa37be1600e24fbbfb7fccd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc7dd453a080d123aa37be1600e24fbbfb7fccd7"}]},"branch":"refs/heads/master"},"6ddd7098807274709cceb8bdf6400d8882be2d15":{"kind":"REWORK","_number":4,"created":"2020-07-09 14:48:55.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/738102/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/738102/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/738102/4"}}},"commit":{"parents":[{"commit":"dba05a9e36a0c1ef49e2fe57a82cd86694c54c05","subject":"Add a functional test for non-owned image copying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dba05a9e36a0c1ef49e2fe57a82cd86694c54c05"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 19:54:19.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-09 14:48:53.000000000","tz":-420},"subject":"Refactor TaskFactory and Executor to take an admin ImageRepo","message":"Refactor TaskFactory and Executor to take an admin ImageRepo\n\nThis refactors the task creation path to allow optional passing\nof an admin-enabled ImageRepo to tasks. This will be used for\nallowing import_image(method\u003d\u0027copy-image\u0027) to update non-owned\nimage metadata during a copy.\n\nChange-Id: Ief3440759fda4bd90979caa79641770cb022b7da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ddd7098807274709cceb8bdf6400d8882be2d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ddd7098807274709cceb8bdf6400d8882be2d15"}]},"branch":"refs/heads/master"},"937a70ab831e045bf4b234030f34ac3bb8539467":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-07-10 15:59:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/738102/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/738102/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/738102/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/738102/5"}}},"commit":{"parents":[{"commit":"019afb789408bb0e179c849a72fd37a63202e2d1","subject":"Add a functional test for non-owned image copying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/019afb789408bb0e179c849a72fd37a63202e2d1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 19:54:19.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-10 15:59:53.000000000","tz":-420},"subject":"Refactor TaskFactory and Executor to take an admin ImageRepo","message":"Refactor TaskFactory and Executor to take an admin ImageRepo\n\nThis refactors the task creation path to allow optional passing\nof an admin-enabled ImageRepo to tasks. This will be used for\nallowing import_image(method\u003d\u0027copy-image\u0027) to update non-owned\nimage metadata during a copy.\n\nChange-Id: Ief3440759fda4bd90979caa79641770cb022b7da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/937a70ab831e045bf4b234030f34ac3bb8539467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/937a70ab831e045bf4b234030f34ac3bb8539467"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
