)]}'
{"id":"openstack%2Fglance~738103","triplet_id":"openstack%2Fglance~master~Iac75956e314ec6f41db18430486bd8be9754e780","project":"openstack/glance","branch":"master","topic":"bp/copy-unowned-image","hashtags":[],"change_id":"Iac75956e314ec6f41db18430486bd8be9754e780","subject":"Make import task capable of running as admin on behalf of user","status":"MERGED","created":"2020-06-25 20:33:18.000000000","updated":"2020-07-14 14:45:19.000000000","submitted":"2020-07-14 14:40:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738103-1594737637967-b2062de1","meta_rev_id":"2e640790983dc7c3f95ba3c2e45e5209dfe6c19f","_number":738103,"virtual_id_number":738103,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-10 16:01:35.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-14 14:40:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-14 13:08:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-07-14 05:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-14 13:08:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-07-14 05:38:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 14:59:22.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-30 17:27:24.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-10 16:01:35.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-14 05:38:53.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-14 13:08:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-14 14:40:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d74a7191087b497c76951aa78238809848572300","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-25 20:33:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2aa144bab68feca28e65954dae3d399ddf2294a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-25 20:34:22.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"cef6c58ac98c3c93372cfd55495e8f9c3239b35e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-25 20:38:14.000000000","message":"Patch Set 1:\n\nPretend we have a fine-grained method of granting the user to get past my new authorization check. This would pass an admin-enabled ImageRepo to copy-image tasks, and the ImportActionWrapper class would use it to do things like setting the image properties only. Right now, only get_flow() uses this, which is pointless, but it shows that we do that first operation as admin, which is what I was choking on, and what the test I added in HEAD~3 was hitting.","accounts_in_message":[],"_revision_number":1},{"id":"306cad0a7af7a958c068b34bf1ddfd16c034968c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 01:01:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/16ada8fff7ae4efbbe5b87ccfcc8322d : SUCCESS in 51m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/15df192f2d244c97abad8692d0e0a1c3 : SUCCESS in 1h 12m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a47254edd8746659127a76a8b2e0a42 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfab60ac627740c088f3cf3dca472683 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7090a6dcef2e48f099dac81bd39760da : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1eae470f003f4e13a66e81416f807612 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a8a9c6ae6f44c69ac8c374e3e3c366d : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/691f0ca7a73f4273863ea4be84bc74b9 : FAILURE in 8m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/727d89e49f6b4b598054a19f0e452d93 : FAILURE in 7m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a00b825eea0f41e4a38d6d374ead9c7a : SUCCESS in 5m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e18146e0b91243a4be8d7dd030d04feb : SUCCESS in 1h 03m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3ca7e0a52ce4c1ab6d6f9453aac23b3 : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":1},{"id":"707968c99091a1e7650dea79a024cb93fd44c259","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-26 02:44:18.000000000","message":"Patch Set 1:\n\nHaven\u0027t fixed up the functional tests yet, obviously.\n\nIf this seems like a reasonable approach, I\u0027ll fill this out, and work on enabling it to actually work by adding the policy knobs.","accounts_in_message":[],"_revision_number":1},{"id":"975c745edbebb54f93abb7317a789bbf78d38124","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-26 07:42:31.000000000","message":"Patch Set 1:\n\nI know this is PoC, Wrapper concept looks fine to me.\n\nJust to note that when I tried this to create image using import in all-stores I saw below behavior.\n\n1. When operation starts os_glance_importing_to_stores doesn\u0027t show first store and starts with \u0027,store2,store3,..\u0027\n2. When import workflow starts image state never changes to importing\n3. After completion of importing image to one store it gets deducted from os_glance_importing_to_stores and appended to os_glance_failed_import (I know this is ToDo, but just wants to highlight)\n4. After successful completion of operation, Image remains in uploading state only (I guess that is because we are not passing state to save in wrapper)","accounts_in_message":[],"_revision_number":1},{"id":"c55d7ecaff132d104439c1a362208e4ff82ef140","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-26 13:29:56.000000000","message":"Patch Set 1:\n\n\u003e I know this is PoC, Wrapper concept looks fine to me.\n \u003e \n \u003e Just to note that when I tried this to create image using import in\n \u003e all-stores I saw below behavior.\n \u003e \n \u003e 1. When operation starts os_glance_importing_to_stores doesn\u0027t show\n \u003e first store and starts with \u0027,store2,store3,..\u0027\n \u003e 2. When import workflow starts image state never changes to\n \u003e importing\n \u003e 3. After completion of importing image to one store it gets\n \u003e deducted from os_glance_importing_to_stores and appended to\n \u003e os_glance_failed_import (I know this is ToDo, but just wants to\n \u003e highlight)\n \u003e 4. After successful completion of operation, Image remains in\n \u003e uploading state only (I guess that is because we are not passing\n \u003e state to save in wrapper)\n\nI literally haven\u0027t even done that, so I\u0027m not surprised it\u0027s unhelpful. All I did was write enough of this to get past the forbidden check in the test I added before this. Not surprised that it has issues like the format of the string :)\n\nI\u0027d like to get a read from jokke too, and anyone else you think is important to sign-off on the idea, after which time I\u0027ll really work on making this work-work.","accounts_in_message":[],"_revision_number":1},{"id":"3d2f37b95ac0728e1a74833b5028f4d64592ee3e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-29 17:57:04.000000000","message":"Patch Set 1:\n\nThis looks like a good approach.","accounts_in_message":[],"_revision_number":1},{"id":"a198e3f167ce71eff741198a161a16fa7c314f88","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 14:59:22.000000000","message":"Patch Set 1:\n\nI think this makes sense as a way to handle this issue.","accounts_in_message":[],"_revision_number":1},{"id":"2971e2a0c86ccd6246675ca8121af514fe1e445f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-30 17:10:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7a28ab2a076a9582f3a9fd944c33a69e10a25029","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 17:11:19.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"0007df7ea1ee99be76e7d61a9b7c2df4f3907afd","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-06-30 17:27:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"430cda3f066f40eac480b2d2530068cfdf5644e2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-30 19:08:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"114bffe7c576ef50e87f801de59851b43cd1af6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 19:41:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/666f7adc22594cfa9d0a468bc9acdeae : SUCCESS in 1h 05m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0de177548510402abbcb0c3068fc164d : SUCCESS in 1h 21m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a34616347b9c41c9a06d36d13a556705 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d0cc77d9084d66bbc43681ecba16c1 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c35f716d9fb5452b9bbffaf8bae65226 : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8db1de57eaf0483a92db5591123653c7 : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aebb8e4c80414bae9d99fc22458d7a90 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a6adc884f3f4f4798eaadc321fce43c : FAILURE in 10m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/42bd9e36a5f24b31af609e8e8f5d3850 : FAILURE in 9m 47s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3610459e938342ce8b7d338ab35d0881 : SUCCESS in 6m 49s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3101d60a361f471ebad0958e149a7fd9 : FAILURE in 26m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6382cbe9a1949d8887e251e4a0bcebd : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":2},{"id":"488cceacdda6311a33e0514974ffdeaf17a14eb4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-30 21:44:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d024f5c1b98d08d9a51857eae4fc8bd67a5493b6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 21:47:01.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"5fa8db50be7fd2bb3a9fdac7b4304510031e1f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 22:59:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bf609624ac54431ea60592a936d18b57 : SUCCESS in 1h 09m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9894022810ef4924b2a0eea2768e5be4 : SUCCESS in 1h 06m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/771eb158566348c6938c76558c6ec70d : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e566658dbfe14cedbe140b1df884bc6b : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd5dbf21c6fd46548bcb01e42dcf91ec : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1149e6e7d146444f87510d7cc88a28c2 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c3b0c4721b547e08e9c67b95399675c : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6946399858cf4fc5a8a820c8e1245cbe : FAILURE in 17m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7a4ebbb3f49f4330b5890afc69f93e16 : SUCCESS in 11m 48s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e9975f3d3a0d4975927020dd2714fb2f : SUCCESS in 5m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/930a9854fbf840c8afb6a810c1f929f4 : FAILURE in 28m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6a3c2ffd5b746328a02590a74f43af0 : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":3},{"id":"8f46dcc02d5d9df52c3a535d9aac91859e0d5e83","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-01 05:34:51.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ec0c78540b4f39c97f155ea3c8dba50707fc2dbc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-01 09:37:37.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4c357d8efb4691c336ac53e072b93adf0e76905b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-01 09:40:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a01f5fceddd28a66e6605fadcd9f11a00221b527","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-01 09:40:48.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"2a548035199532d93577c58a9234de25881362dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 11:41:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e29f1289aac84bbea10ff8ba3445dee6 : SUCCESS in 58m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5befaed02b5740808a04a6cbc5d6e046 : SUCCESS in 1h 17m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ce9d0fa955e43ada2072ee6b1083696 : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9310e38128454aa5bb0f875912c7cae4 : SUCCESS in 9m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ee103491b364ceea271352da16de224 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0ec5c3127b64fb1b452c08f79b9174b : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/573e2bcf0fab49a08c1168bf056c2618 : SUCCESS in 9m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3fc89a24e46645fbab8fd9440d8e06c9 : SUCCESS in 9m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e0a1ef117de54483bf8b3420c46c444b : SUCCESS in 10m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0a3d1b15349845a7b33bd8f23875f587 : SUCCESS in 5m 51s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bf13ba36c01c41c8ba365b1386fe5621 : SUCCESS in 1h 06m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91d724e6012c4ce8bc27a408b89cd0d4 : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":3},{"id":"59844ce5726b5de548a433e100135ccc0c08f290","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-01 13:41:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d8cc94ec2e29ad1e12150ef711c481b861a8281","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-01 13:43:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bfe17ad3a52bb6a562d933bfda576ce92e0a40a4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-01 15:41:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9446337b1ed472d899a1b10d5dd9848e6157a208","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-01 15:43:48.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"73081243855cd5f05cffbaa65083462b06502704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 16:56:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/38293557546040778c9b082b87668ff0 : SUCCESS in 1h 02m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59778475f6174ea8a404804b9ac0abe1 : SUCCESS in 1h 05m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/908fa59d1a9c4dbca040fc0d056448bf : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58be54772eef4d90994ab5661daf2ef4 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a014ca98dd494561ae5d3545aed85e07 : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8331c7a86354620a5a50037b5c5e6c0 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863dfe04b05d49699f7247ad5d734238 : SUCCESS in 9m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/09996ea57f20457e971c057382362800 : SUCCESS in 9m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a05c65a12d874572a13344384a452768 : FAILURE in 7m 43s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b421cd81ccce4465b3b3d7c53c2b700b : SUCCESS in 6m 57s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/127d2bf7848840c29974bf968ecd7ecd : SUCCESS in 1h 04m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddabe148d8ed4bf6b878bab7f1b43421 : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":4},{"id":"1e6f41db7eda4c35dace4cae7942683a0e074d65","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-02 04:46:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e0b5beb130b4317160f467dd79b2d17538c973c1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-02 04:47:33.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"28e8960bd3d2972a128aa5f3712b31ffd1adcf9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 05:59:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/925ec6b358934c4f95c741eebaa590a3 : SUCCESS in 1h 04m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8e5ed9211e941248f68e268f80a074e : SUCCESS in 1h 11m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7082520b68249c68ad13fa26db88bbc : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb7a7aaf4b4346ae97099acbf8cd9ca8 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caa0358fa15d41c8b2346d2037b346dc : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2868f4b77ee44df8a78b526c6cb04b52 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4024545424f41988b82c1c91345a091 : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2b25c38d05b44a00968415fd4fa5b405 : SUCCESS in 9m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ba77b716aab4cf29182353a29d493ae : SUCCESS in 12m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ad1f26bf4fa74ddca02d9b0b6f7e2544 : SUCCESS in 6m 12s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ae936b522811439d85b72dbf94821d9e : SUCCESS in 1h 02m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/396e14442d624995ac10beb2c4264ae5 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":4},{"id":"4d070be859974ffb6eba649bd9afb65345f3d3ce","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-09 14:48:55.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"1b63b4f2ea78bba4fdebc6d56ef7513dbfa91480","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-09 14:51:22.000000000","message":"Topic changed from bug/1884587 to bp/copy-unowned-image","accounts_in_message":[],"_revision_number":5},{"id":"3935a1b2b95de4aa49019ea543f5f9179bf03201","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-09 18:58:29.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"2b845bab61e8d92cb504320173df06f4dfc2a7d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 20:43:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/147489adaa8641d9af5a4fc15713e819 : SUCCESS in 1h 00m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4fba597094d43658b5a1667f0d898b5 : SUCCESS in 1h 06m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9564c6578444520b5d8cba8a4114f48 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9455178c13f846ab80b140c7d1d9e769 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/784277472e6743e2b4065ff7ef5b0bc3 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e6f56a9b69846fe9839ecea9f4f3432 : SUCCESS in 13m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/122f797f098b42c481a1c16c186c384f : SUCCESS in 7m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/24a5171cd799448fb3aa199cb933c9f4 : SUCCESS in 17m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6b659e7c569c4b7fb87714669eeb1977 : SUCCESS in 10m 23s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/83318ff75efe4e50bc95a2b61d91f341 : SUCCESS in 7m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8cdbdfb7b07543cabf61538719df6522 : SUCCESS in 1h 08m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08f76e3d355a4b66b8763ac2f4704443 : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":5},{"id":"442fb942f5ced13521d6ff9a1b47de163688a615","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-10 15:59:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"215350630577b21e7595275db27d4ae37e26a94d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-10 16:01:35.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"e64c259f45a733c3a32b2cb017f0632dc25700c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 18:04:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b4b2251b8b9e47f8ade4d50338e6b521 : SUCCESS in 1h 05m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5072a0bab9e54d429d36bf43792a8475 : SUCCESS in 52m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b9250d817b640caa8d1ac3db565b892 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c40a895095f049e2ae4f509730887d96 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cda5228886842c497b6383a9be8f48a : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4db41f12d6b94920921949058fc31711 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c39240060abb4a07871e651d72841805 : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/caa34c0a0dec47a197a3298d03aad4f7 : SUCCESS in 9m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f335b6e584b46cf867ee59a75a4c6a8 : SUCCESS in 17m 00s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/309e51f6c7654009a8b33ad5b9f1f761 : SUCCESS in 7m 11s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6f022d439fd64805adb87472be701000 : SUCCESS in 1h 13m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e4373c60968473fb82fa3b1b83dae5b : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":6},{"id":"6c7b29652504a2cb0f94802b705b45ccc3442002","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-14 05:38:53.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1\n\nTested locally, working as per expectation, so far no regression found.\n\nThank you Dan!!","accounts_in_message":[],"_revision_number":6},{"id":"2c3c041a6b9bfbb7d49b4dc0a1b481815e131497","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-14 13:08:08.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nLooks good to me. Let\u0027s get it merged and see how it goes.","accounts_in_message":[],"_revision_number":6},{"id":"f998c0a5137e138eda073dbc0d0e9ba16d87203f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 13:08:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"899f610a0009c9ff7d5bc45b63657919e1e918e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 14:40:37.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f72f7f860c6d4b2c913b581d7af47f07 : SUCCESS in 57m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f0817edcc48a46a9ad4ca02d335b2093 : SUCCESS in 1h 15m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34038a3d8b7941e4a7519dcb9091d808 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dca76a4b5cf4916a0e58f25bfe34ed6 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/893ccb1c196a4474aabf9a7c3323eb8d : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ab69db93302462b929d0abd736b209a : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ad7bccc845427b963f2ae9c4de2e2d : SUCCESS in 10m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/020ed81d50104827ac826fe03ce19529 : SUCCESS in 7m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/07d226f4886c45ef9da568b60ab56bc9 : SUCCESS in 14m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e14f386f8e9f410da1f9890f39a139f4 : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":6},{"id":"2ca0d84fd886c6df034808b71f5d1f72271a717e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 14:40:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"2e640790983dc7c3f95ba3c2e45e5209dfe6c19f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 14:45:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/403acfbe8aed4f74a79030cae42ac81d : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2fd0c2573311db6a50228ed22adf8bc2b5638cfe","revisions":{"06b8875b8a8f93eead8ddf44edbc7e870f29b666":{"kind":"REWORK","_number":1,"created":"2020-06-25 20:33:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/03/738103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/738103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/738103/1"}}},"commit":{"parents":[{"commit":"8530a30f2c030e5cae2b96fbe909845ebdc6fc7a","subject":"WIP: Refactor TaskFactory and Executor to take an admin ImageRepo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8530a30f2c030e5cae2b96fbe909845ebdc6fc7a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 18:51:37.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 20:33:19.000000000","tz":-420},"subject":"WIP: Admin import action wrapper","message":"WIP: Admin import action wrapper\n\nThis is a PoC of my proposed way to encapsulate admin metadata\nmanipulation for copy-image imports.\n\nChange-Id: Iac75956e314ec6f41db18430486bd8be9754e780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06b8875b8a8f93eead8ddf44edbc7e870f29b666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06b8875b8a8f93eead8ddf44edbc7e870f29b666"}]},"branch":"refs/heads/master"},"09c9c141f629dfc024c15055ffdd44135d8ee49c":{"kind":"REWORK","_number":2,"created":"2020-06-30 17:10:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/03/738103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/738103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/738103/2"}}},"commit":{"parents":[{"commit":"8530a30f2c030e5cae2b96fbe909845ebdc6fc7a","subject":"WIP: Refactor TaskFactory and Executor to take an admin ImageRepo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8530a30f2c030e5cae2b96fbe909845ebdc6fc7a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 18:51:37.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-30 17:10:16.000000000","tz":-420},"subject":"WIP: Admin import action wrapper","message":"WIP: Admin import action wrapper\n\nThis is a PoC of my proposed way to encapsulate admin metadata\nmanipulation for copy-image imports.\n\nChange-Id: Iac75956e314ec6f41db18430486bd8be9754e780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/09c9c141f629dfc024c15055ffdd44135d8ee49c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/09c9c141f629dfc024c15055ffdd44135d8ee49c"}]},"branch":"refs/heads/master"},"316d45f02910855b5078ef1e4a34134aa7018d2c":{"kind":"REWORK","_number":3,"created":"2020-06-30 21:44:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/03/738103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/738103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/738103/3"}}},"commit":{"parents":[{"commit":"552c882ecec43ee17e16864f9426c641205178cb","subject":"WIP: Refactor TaskFactory and Executor to take an admin ImageRepo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/552c882ecec43ee17e16864f9426c641205178cb"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 18:51:37.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-30 21:37:41.000000000","tz":-420},"subject":"Make import task capable of running as admin on behalf of user","message":"Make import task capable of running as admin on behalf of user\n\nThis makes the api_image_import task capable of running as an admin on\nbehalf of a user if so authorized by the API. It includes a new object\ncalled ImportActionWrapper which provides a bundle of utility methods\nwhich can be run either against the user-authorized or admin-authorized\nImageRepo passed in from the API. It encapsulates all the actions\nwe are able and willing to run as an admin for the user.\n\nThis is currently not drivable by the API because the policy check is\nstill statically defined as \"admin or owner\" but this change is\noffered without any needed modification to the functional tests to\nprove that it does not regress existing functionality. The following\npatch will introduce a more robust knob for allowing users to do this,\nand it brings the functional test changes with it.\n\nChange-Id: Iac75956e314ec6f41db18430486bd8be9754e780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/316d45f02910855b5078ef1e4a34134aa7018d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/316d45f02910855b5078ef1e4a34134aa7018d2c"}]},"branch":"refs/heads/master"},"55e0eff6e0dbb1399e4e96bf0f268caced6f9eef":{"kind":"REWORK","_number":4,"created":"2020-07-01 15:41:55.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/03/738103/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/738103/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/738103/4"}}},"commit":{"parents":[{"commit":"4e84909c10578151d65edc96b9fe310fb913072e","subject":"Add context.elevated() helper for getting admin privileges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e84909c10578151d65edc96b9fe310fb913072e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 18:51:37.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-01 15:41:26.000000000","tz":-420},"subject":"Make import task capable of running as admin on behalf of user","message":"Make import task capable of running as admin on behalf of user\n\nThis makes the api_image_import task capable of running as an admin on\nbehalf of a user if so authorized by the API. It includes a new object\ncalled ImportActionWrapper which provides a bundle of utility methods\nwhich can be run either against the user-authorized or admin-authorized\nImageRepo passed in from the API. It encapsulates all the actions\nwe are able and willing to run as an admin for the user.\n\nThis is currently not drivable by the API because the policy check is\nstill statically defined as \"admin or owner\" but this change is\noffered without any needed modification to the functional tests to\nprove that it does not regress existing functionality. The following\npatch will introduce a more robust knob for allowing users to do this,\nand it brings the functional test changes with it.\n\nChange-Id: Iac75956e314ec6f41db18430486bd8be9754e780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55e0eff6e0dbb1399e4e96bf0f268caced6f9eef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55e0eff6e0dbb1399e4e96bf0f268caced6f9eef"}]},"branch":"refs/heads/master"},"7673ce0750fa439d47c782905a039f6ab7aa1658":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-07-09 14:48:55.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/03/738103/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/738103/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/738103/5"}}},"commit":{"parents":[{"commit":"bd9a20adbe7b665f3569e7616275a77521abcc76","subject":"Add context.elevated() helper for getting admin privileges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd9a20adbe7b665f3569e7616275a77521abcc76"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 18:51:37.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-09 14:48:53.000000000","tz":-420},"subject":"Make import task capable of running as admin on behalf of user","message":"Make import task capable of running as admin on behalf of user\n\nThis makes the api_image_import task capable of running as an admin on\nbehalf of a user if so authorized by the API. It includes a new object\ncalled ImportActionWrapper which provides a bundle of utility methods\nwhich can be run either against the user-authorized or admin-authorized\nImageRepo passed in from the API. It encapsulates all the actions\nwe are able and willing to run as an admin for the user.\n\nThis is currently not drivable by the API because the policy check is\nstill statically defined as \"admin or owner\" but this change is\noffered without any needed modification to the functional tests to\nprove that it does not regress existing functionality. The following\npatch will introduce a more robust knob for allowing users to do this,\nand it brings the functional test changes with it.\n\nChange-Id: Iac75956e314ec6f41db18430486bd8be9754e780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7673ce0750fa439d47c782905a039f6ab7aa1658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7673ce0750fa439d47c782905a039f6ab7aa1658"}]},"branch":"refs/heads/master"},"2fd0c2573311db6a50228ed22adf8bc2b5638cfe":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-07-10 15:59:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/03/738103/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/738103/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/738103/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/738103/6"}}},"commit":{"parents":[{"commit":"c59ed1bce82c960b205663ba9ac6d038633588dc","subject":"Add context.elevated() helper for getting admin privileges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c59ed1bce82c960b205663ba9ac6d038633588dc"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-25 18:51:37.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-10 15:59:54.000000000","tz":-420},"subject":"Make import task capable of running as admin on behalf of user","message":"Make import task capable of running as admin on behalf of user\n\nThis makes the api_image_import task capable of running as an admin on\nbehalf of a user if so authorized by the API. It includes a new object\ncalled ImportActionWrapper which provides a bundle of utility methods\nwhich can be run either against the user-authorized or admin-authorized\nImageRepo passed in from the API. It encapsulates all the actions\nwe are able and willing to run as an admin for the user.\n\nThis is currently not drivable by the API because the policy check is\nstill statically defined as \"admin or owner\" but this change is\noffered without any needed modification to the functional tests to\nprove that it does not regress existing functionality. The following\npatch will introduce a more robust knob for allowing users to do this,\nand it brings the functional test changes with it.\n\nChange-Id: Iac75956e314ec6f41db18430486bd8be9754e780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2fd0c2573311db6a50228ed22adf8bc2b5638cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2fd0c2573311db6a50228ed22adf8bc2b5638cfe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
