)]}'
{"id":"openstack%2Fglance~738241","triplet_id":"openstack%2Fglance~master~Ifc4b3bc5c2a09b9a54f3e3ccd7d4432c9d9b88fc","project":"openstack/glance","branch":"master","topic":"bug/1884596","hashtags":[],"change_id":"Ifc4b3bc5c2a09b9a54f3e3ccd7d4432c9d9b88fc","subject":"WIP: FakeImage test fixture is not reflective of reality","status":"ABANDONED","created":"2020-06-26 16:40:13.000000000","updated":"2022-07-07 15:36:36.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4bf098882da862df88f31271281f19fdfb5dd59d","_number":738241,"virtual_id_number":738241,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-26 17:01:43.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"date":"2020-06-29 14:49:20.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-26 19:50:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-26 16:41:28.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2020-06-29 14:57:22.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 16:41:28.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-06-26 17:01:43.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-26 19:50:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-29 14:57:22.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"da300a0663a10e8dff790e5c1049b48626f99732","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-26 16:40:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7e6c8eb4e57b18d00eaa307c37063f9c7376a88","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-26 17:01:43.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"78076823e1798b89a2fe7ccfbec5a252a4fa276e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 19:50:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1052aa4084db44f889f4738f94c4fbcc : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32befdbe38154a0290c28a384c6f12ce : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36ef9f1305f3488a841416c416b21680 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e2677d3c19147c68ac74d5e35161e61 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ea8881c05cc4395962a6dbb8bda94b1 : SUCCESS in 7m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15da0cbb077847bbb4cfb143dfd9336c : SUCCESS in 12m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/342169d2bfbc4a5387e77d0af301b669 : SUCCESS in 12m 07s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/07b7a3458fd74efcbdd91f138a3c53ff : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5387108c009bb73abfbe0955c592a624deb6b3fc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-29 05:40:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e08766aabb2c2199615d7a8b1e8ab1156dfe9242","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-06-29 14:49:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3324e20eb3f27ca9149c2813c898cdffcf4bf24","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-06-29 14:57:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4bf098882da862df88f31271281f19fdfb5dd59d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-07 15:36:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d9269c4a7e2597a2db569a7cd86a36ebe08b521","revisions":{"2d9269c4a7e2597a2db569a7cd86a36ebe08b521":{"kind":"REWORK","_number":1,"created":"2020-06-26 16:40:13.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/41/738241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/41/738241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/738241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/738241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/738241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/41/738241/1"}}},"commit":{"parents":[{"commit":"bdebef8aeb347cc008c9830f2ec4ed784ea5a697","subject":"Add image_set_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdebef8aeb347cc008c9830f2ec4ed784ea5a697"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-26 16:34:32.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-26 16:38:20.000000000","tz":-420},"subject":"WIP: FakeImage test fixture is not reflective of reality","message":"WIP: FakeImage test fixture is not reflective of reality\n\nSurely I must be wrong about this, but it looks to me like FakeImage\nin the test_images_resource test is not a proper representation of\nproxy.Image(), specifically around the image_id field. Presumably it\nis reflective of the *external* REST representation of an image, but\nis \"close enough\" that it got used extensively for mocking out the\nresult of DB calls as well. I\u0027m putting this in a separate patch\nbecause if this is true, the tests should probably be fixed in a\nseparate patch from any other changes. This hack is just here to show\nthat it makes the following patch work.\n\nIs it possible that these tests have been wrong for a long time?\n\nChange-Id: Ifc4b3bc5c2a09b9a54f3e3ccd7d4432c9d9b88fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d9269c4a7e2597a2db569a7cd86a36ebe08b521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d9269c4a7e2597a2db569a7cd86a36ebe08b521"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
