)]}'
{"id":"openstack%2Fglance~738671","triplet_id":"openstack%2Fglance~master~I75014bd50bf382efebe56bd89c20ffefbdde25f5","project":"openstack/glance","branch":"master","topic":"cleanout_registry_simpledb","hashtags":[],"change_id":"I75014bd50bf382efebe56bd89c20ffefbdde25f5","subject":"Deprecation cleanout Registry and related","status":"MERGED","created":"2020-06-30 16:49:54.000000000","updated":"2020-07-02 08:04:49.000000000","submitted":"2020-07-02 08:02:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738671-1593676961046-fca765a5","meta_rev_id":"bbd2fe29d69f77436ffa04b1511bcd6fe899986c","_number":738671,"virtual_id_number":738671,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-30 19:43:54.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 08:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-01 11:04:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-01 05:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-02 05:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 19:43:54.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-01 11:04:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-02 05:55:18.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-02 08:02:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4fb9aef5258e1c4c78636fc7eb4895b3079b39e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-06-30 16:49:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b9e9d316f9c4a60dccf613d7f088d15c68fbfd4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 16:50:16.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"3067505a9a72ed76799f39988365283bf940f264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 19:14:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/081577c6e0514aaca753a423a1807dbe : SUCCESS in 4m 22s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fa4e6865f6ef4895b2e3b616e380f504 : SUCCESS in 7m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/af007cf163504f87afe213d08afae008 : FAILURE in 6m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/3bc55439c0244ebb89b18a622d8022d4 : SUCCESS in 1h 05m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc493a26174d47fa8b8aa77be4341892 : SUCCESS in 1h 15m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e32fb6350e6e441e9f4c64c6501fe5d2 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64db1c69a55d4e468444b80bac60a970 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1668c8ba0eee41d68dba510a725a0e6d : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9538bef6564408193e02334e60dcc1c : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1127b00402c4451388a6805e100103ed : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de78ff078e2344e9a1e2448be0c81090 : SUCCESS in 7m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b8ff521938c441eb1981b37e4b5894e : SUCCESS in 10m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/52e9fe71be824194864636c702133908 : SUCCESS in 9m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/75a88c9111dd44a5ad789d31e630ab5f : SUCCESS in 9m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/074f1d84ee8444cea57d68b9d93bffc4 : FAILURE in 29m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/898a29763d3d48d588b385c6aeddda0e : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3aa6b0d7f4895090f3bedc4e33a1ce58030adce3","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-06-30 19:43:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5cfc818fae51f6edc3a338025dd66f216e30a010","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 19:43:54.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"e3ffcf4c8e76341c0fd9e0f4c52e426e8411f3ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 21:06:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/589c486dcbb74583929598d1851815c6 : SUCCESS in 5m 39s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6626022f8bfe4a3fa9430222431438dd : SUCCESS in 5m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c8648dacb0234d9ab8b4ebb6b0b36494 : SUCCESS in 5m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/96a71aa836f64b7791b14c2013a5d32b : SUCCESS in 1h 04m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/427289dff1a5455cb28ec34ae4ecfd70 : SUCCESS in 1h 17m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48acbca455454978a27ab3e36d802dd5 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f0196851c6b4e1d8865b953cc2fc3e5 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03314eb78698448d8ad9c995c2131df2 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6229683eafe34a77aa7a71fc3d4581f9 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e344ae8062545f9afad25a4e47562f8 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/923b4739d5594ba491616991e8bc3f16 : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5b80bab7274c488fb90c2ccfe286c529 : SUCCESS in 6m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e65842de20cc407ba543b0793215d61c : SUCCESS in 15m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a1a1fb2bcef44a42a63b15d335150c25 : SUCCESS in 8m 23s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/26b387c99e3e41f18d3bc78166525eb0 : FAILURE in 26m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd53bfb742054f5794df2d6d1c1b4280 : SUCCESS in 52m 31s","accounts_in_message":[],"_revision_number":2},{"id":"3d78d738b217569cb230b1f8eeee6daa9cd16fbd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-01 05:43:08.000000000","message":"Patch Set 2: Code-Review+2\n\nThank you Erno!!","accounts_in_message":[],"_revision_number":2},{"id":"d6d0d955cb920f4509cafee8cb280dc61bfb5f59","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-01 11:04:21.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good to me, but since this touches a lot of code, will leave approval to someone else.","accounts_in_message":[],"_revision_number":2},{"id":"23d786edbafd9b96231d2193dc47877b8ee1ebbf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-02 05:55:18.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0dd3ec533ef7bddfbc6b7108c6c5af7224976d11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 05:55:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"764bf996ca394625d05afa06344fc5b1e671d8b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 08:02:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b80f8d0969fb490a8c2543288f330abc : SUCCESS in 5m 49s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b1f03ef2052947c198c095b410644757 : SUCCESS in 11m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/763dc5b8acf346389e830e5dce3bb7ed : SUCCESS in 5m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/0d07538a95a04da9b33edf74ed4f455e : SUCCESS in 1h 04m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e766d97edb14b3392f0fea00f1dc6be : SUCCESS in 1h 14m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6af3936fe5634d7682944b7b034d3f91 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbad870a41c3456592281a543ba6eb96 : SUCCESS in 10m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41cb745754c44487b178b4785ff54bea : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/704c1a00f952493d987b826f9a9cb0e5 : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb466566e404595a4f6e3f2c62e6f0a : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6b426a293e1444fb1fde67770c0b6b0 : SUCCESS in 8m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a49214e2d6bd49eaaab0b5dfad10a19a : SUCCESS in 11m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/625762d8f8d948df86699bbcab9fc6ad : SUCCESS in 14m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07b9bb4b3dc443e48577c193909056d0 : SUCCESS in 49m 09s","accounts_in_message":[],"_revision_number":2},{"id":"a13746feaf7eafeee5fc15e096ec61d1ab77f947","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 08:02:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8a9f3a3b012a0396659e69231c2d12f9148d78eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 08:04:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/1c8e792eadb04f929d3bdcbffc15a899 : SUCCESS in 1m 21s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ccb4141bedf481f8263fe6287fc27eb : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e26900b989b74862982f9c69c7b73015 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"673666cbf4f5ee97b28f157f3b1e7115b72c5fee","revisions":{"34e0489167eb4955d1c74c4e048ac910259b469a":{"kind":"REWORK","_number":1,"created":"2020-06-30 16:49:54.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/71/738671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/71/738671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/738671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/738671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/738671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/71/738671/1"}}},"commit":{"parents":[{"commit":"e6db0b10a703037f754007bef6f56451086850cd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6db0b10a703037f754007bef6f56451086850cd"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-06-30 14:13:54.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-06-30 15:30:56.000000000","tz":60},"subject":"Deprecation cleanout Registry and related","message":"Deprecation cleanout Registry and related\n\nThis patch removes majority of the registry and it\u0027s related\nendpoints and config options that has been deprecated for\nremoval in various releases.\n\nChange-Id: I75014bd50bf382efebe56bd89c20ffefbdde25f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/34e0489167eb4955d1c74c4e048ac910259b469a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/34e0489167eb4955d1c74c4e048ac910259b469a"}]},"branch":"refs/heads/master"},"673666cbf4f5ee97b28f157f3b1e7115b72c5fee":{"kind":"REWORK","_number":2,"created":"2020-06-30 19:43:26.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/71/738671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/71/738671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/738671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/738671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/738671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/71/738671/2"}}},"commit":{"parents":[{"commit":"e6db0b10a703037f754007bef6f56451086850cd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6db0b10a703037f754007bef6f56451086850cd"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-06-30 14:13:54.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-06-30 19:41:30.000000000","tz":60},"subject":"Deprecation cleanout Registry and related","message":"Deprecation cleanout Registry and related\n\nThis patch removes majority of the registry and it\u0027s related\nendpoints and config options that has been deprecated for\nremoval in various releases.\n\nChange-Id: I75014bd50bf382efebe56bd89c20ffefbdde25f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/673666cbf4f5ee97b28f157f3b1e7115b72c5fee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/673666cbf4f5ee97b28f157f3b1e7115b72c5fee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
