)]}'
{"id":"openstack%2Fglance~738672","triplet_id":"openstack%2Fglance~master~Ic83e7e8077b3fc939990c1f8e9c99b13b7fcd7ad","project":"openstack/glance","branch":"master","topic":"cleanout_registry_simpledb","hashtags":[],"change_id":"Ic83e7e8077b3fc939990c1f8e9c99b13b7fcd7ad","subject":"Removal of \u0027enable_v2_api\u0027","status":"MERGED","created":"2020-06-30 16:49:54.000000000","updated":"2020-07-13 20:58:06.000000000","submitted":"2020-07-13 20:56:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738672-1594673770921-bfbdc47d","meta_rev_id":"c956f634e4ffa264389a91fae9b2158b84e94a4d","_number":738672,"virtual_id_number":738672,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-10 14:39:56.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2020-07-13 19:42:02.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-13 20:56:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-13 19:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-07-13 05:21:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-13 19:36:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 14:54:46.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-07-10 14:39:56.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-13 05:21:51.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-13 19:36:01.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-13 20:56:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"180f9806d828a7a8c625cfc79455c0e9a2275315","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-06-30 16:49:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e22b43b83648556fb86177b041ea666e50fb88d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 16:50:27.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"b5075ed6eb5a877ff918018eba97e0f125e49bdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 19:20:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/366cbff703174daeaea5ca45ba2409b5 : SUCCESS in 1h 15m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/159c8ec0017a4263bb6a8e52c985b3e0 : SUCCESS in 1h 15m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39e277dccea342b08b78a22b5d47f3fc : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/202d446df2374c4fb98a92200a99c4a7 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3d34f239fe24fd19369fd106bf4845f : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdff730b63d94016b7a4cc20fde9957d : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1927262e9b884206b0d81e5a82e94413 : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4e601b02dd64c8f8571f742d96c2f9d : SUCCESS in 6m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f94b225318464bf2b4b24d118bef3f99 : SUCCESS in 10m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/711b89fdabe14e0da1f4c7c94113d7e5 : SUCCESS in 11m 31s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e5db372b67214aa2b81412e5a56509e1 : SUCCESS in 11m 06s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/efc89f205587431b9fac426032127b9a : FAILURE in 26m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1b7155460de4e6bb2d278c575d62d50 : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3df7d4e52e664487f5895fa80c67e89a6d64adbc","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-06-30 19:43:26.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"89bb4e0127e5195ad53d6baf3a943a045419baad","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-30 19:44:07.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"b50242f575f291cf2646014e0af5c20f5e566c4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 21:21:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/52503ed504e746debef32ea80baeb444 : SUCCESS in 1h 05m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fa17f3185f1d4d88acbbea243db863e6 : SUCCESS in 1h 37m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23aba5be42e24d0dbc6c7f8489c5ea16 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6f45270ad4435c9b090c7a010b8bbc : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83eff5e8a1a640b79b7de78cab1a041a : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8756023fc6224381aaed929b69d203e4 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4d4a38746f4a1eb0d951270e6f330d : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae3b99bd75174a51939130c051873d2e : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1fa1b77b0cee4b4584dae68e65ee6fac : SUCCESS in 19m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1bca188f3c854aad90b3b17bc14159ed : SUCCESS in 18m 43s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0907f61c181847b6acd3964132bbbb04 : SUCCESS in 4m 34s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5513ed2dce694fe5becb708aa2067545 : FAILURE in 24m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d02efcde39de4fb58d80fa9dcd8024c7 : SUCCESS in 1h 22m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f2b6702e82a3a2765bcfb859cf1c7838fad9baa4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-01 05:46:33.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nThank you Erno!","accounts_in_message":[],"_revision_number":2},{"id":"1a808e35e02c04080196ccee97153e5a1b24a53d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-02 14:54:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d19c56efaf8fc90e8eaedc5c0d534ad498382be7","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-03 19:01:43.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6bf6d4ba9b19a9c55e4924e5653afb525008df9f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-03 19:02:04.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"a5019787a0bbedf107e856e6c2a2052569a3e813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 19:25:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e6738ee6fd864fec9ff225808cebca15 : FAILURE in 19m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/090667577eff43e9b704497cf8bc88a4 : FAILURE in 17m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af3d70ce278e49f4964a291507c1bbd8 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92fb93a3bdcc4ae891f7451473dddd04 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30cb1c354fc24ba2ba9d0bc8a55f0967 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad6b40bc661343c2a1c9c18dbcce53a1 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68a1abac5c4f4b57bf067cf5ced240b2 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f964cfdf1924f33af6212456db002a0 : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/009f0c7d74b44a94992b9ea032968d88 : SUCCESS in 7m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/72642ee947184f1c8a02aa5d1d1ce30e : SUCCESS in 7m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0dc46e114f454a5e8113296243d2788f : SUCCESS in 5m 00s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e95cf0437ac242d2b653f45c9f91d7e9 : FAILURE in 18m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e9c6fcf87f144a399725630f0c5d87b : FAILURE in 17m 23s","accounts_in_message":[],"_revision_number":3},{"id":"dbb30960b3b1077f3e85554488e2f7f7a0f07a0b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-06 11:27:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"633e449fc4f5807ce120428e7a99d7950b757c5b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-06 11:27:55.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"17df226b1c2ac99cb7bb95f5c6cb40efa3628b22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 12:57:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c0315a3fc5b947e8a43b5a478475bb03 : SUCCESS in 1h 02m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fd0d2f77e1f4eef933cdf62b09ce4ea : SUCCESS in 1h 04m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d152e3c3904450396d290cfaaf8024c : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8078ee9268a7490fa7f18c982062af1f : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40c3f2ae39e54701b4d3a61da6d770cf : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f7f8170522d453f9012507aa108f538 : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5cdca30b744d0fb8fd733688d183a9 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3335b000cf874865b517371d0cfafb64 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2a7cf13695f9416eb1871dfb178b683d : TIMED_OUT in 31m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f00bc2509d8a45778ba1043f85ee53cd : SUCCESS in 10m 05s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/37b29c694a874b5ebaa413348b78512d : SUCCESS in 8m 17s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/cc2804afc7dc46ba9c40e06b537996e6 : SUCCESS in 1h 09m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93ef68dc7eca4e0e8eb068671fdef3ea : SUCCESS in 51m 36s","accounts_in_message":[],"_revision_number":3},{"id":"099357fde32dd2888d0c20cae754ff3fe229f850","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-06 13:28:12.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4df31ddbe696e721672ea0939e13a1080d52ad6f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-06 13:28:33.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"5f6db220cac32635569b19b9b215a63f6b41a764","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:50:48.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d44cc607854c4c428a18b19b8c097a6a : SUCCESS in 1h 13m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/66945a37eb154c318032892e8112de81 : SUCCESS in 1h 26m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fcb18be197040e78145cd235e508be8 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3014039c62c240309965e4b670044ccc : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d67c8bdd55d3474bb4b4e70c5a289bc3 : SUCCESS in 12m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8471885ef4064c4e88cea6b3aa4b4a25 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ebb3f5d609460f9807b35b23af5dc6 : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84d535c1f5124813a2d07ea8d57eaa00 : POST_FAILURE in 17m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2f8b2a469bf141a8987a5a1f4c7a26e6 : SUCCESS in 25m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d33d8b1cb6204bb49693ac135cc7298d : SUCCESS in 18m 10s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ade6c2b8bc8a49f88d7ba188f7fd0824 : SUCCESS in 5m 58s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8606848ade3d48989e495b70529fc650 : SUCCESS in 1h 14m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e468f8e76e648f0bbfbaaf285edf3e9 : SUCCESS in 50m 22s","accounts_in_message":[],"_revision_number":3},{"id":"218f58864a5886e98bd3d8b452325eeab599adae","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-06 16:41:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"95406100ad742e71a2abca54880d6944a33e3aca","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-06 16:41:46.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"fd020ab071017cfdc32f82154c9c86fe64097ae4","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-06 17:31:58.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2530e056e67f62a4bd4f03ae5606b15e5342d755","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-06 17:32:20.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"867c39ac8eef9d5b0d314001576c3a4a8dfd7404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 18:50:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3706dd0172074525a6f8826b51a22396 : SUCCESS in 1h 08m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0741f60b3aeb4503a10d95e6c0d56431 : SUCCESS in 1h 17m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/229945e7be3543a09053b3ea0e8b0e9a : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07194227b8e6416b8776610273b21fc7 : SUCCESS in 10m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3c1f9e7739b49b88583ab734ac30047 : SUCCESS in 11m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/248dadabb31e4c75acfea6fe1fa8819b : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/930a991fd18e49edb2e1b562e293abec : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b535d9b490d4e27a0df822e374096d6 : SUCCESS in 9m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c868515174f42b7ac6f150a1c3d5f9d : TIMED_OUT in 32m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0d1fb260a0124355b6f006a88a667392 : SUCCESS in 13m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8c7350aac88542738a54abd0e6368ae8 : SUCCESS in 12m 33s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c55ec2c4a5714b3eaa65d88d8917fccf : SUCCESS in 1h 08m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c7df1fe99174301b15a998776fe5ff3 : SUCCESS in 1h 23m 45s","accounts_in_message":[],"_revision_number":3},{"id":"2ba93fae7b63faec07972433231b7b1eadeb6d69","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-07 14:16:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bb5929cb230714ad39517357d1fcb775cfa77262","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-07 14:16:56.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"a494c9349983320dae9f9554d22ba6181df2e76f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 18:33:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2305e6f7d3de45e3b2061a65845cd1ed : SUCCESS in 1h 15m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a260f37b9f7413db712308a29bce708 : FAILURE in 1h 02m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1b76cf1bb914bd797c5363ddbde7ced : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6bfdb1ec3604894935887bddc110363 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5fadd265d07443fa9c6aeebeef22d06 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02e6815536b64fa5815df104165c8b19 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1734b647f9c346fcbdb5bf557522e10b : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/596f34281bcc4c91a35118cdbd1119b2 : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d50d59314cdf44eb9ce84aad0a908dfa : SUCCESS in 8m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eaa6dc9b458e4c1daa8dfc4250e14788 : SUCCESS in 13m 36s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4fb4a8240d1647b997f28473202d5330 : SUCCESS in 6m 27s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5edf4eb52fc54942a5db0b39be21952c : SUCCESS in 1h 21m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d06b15532cab418e9a9976b11b1154eb : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":3},{"id":"b2b17eca7e56e61c7915cc820421d2303b3ccb21","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-08 13:18:17.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d98571f30558f92534b7b7aa11a4f0e8092b01a2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-08 16:00:22.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"78c4c4b861f60251aa47ce63d1b8e749f8aac791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 21:02:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7e17cc7bfdba4ac1b3384a04ddef7357 : SUCCESS in 59m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3980903674734d6c89cd59441568e2cb : SUCCESS in 1h 06m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a24be0421c224970a5c096c2ca8a35e9 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c60af85aa18c4c6985866d71333c9736 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c08a00301d14f6a993bf03fc53a0efb : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8059709e9b54756ae2d72927584dcc8 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e81dcfc36364cc9bc3fbd3230eac3e0 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/906dad9df6ae481f950e8cb4edee1d2a : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b1bb3fc5c9941a09bd5f4af70b9d047 : TIMED_OUT in 32m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/633a78454d0a4288a94d703b84379329 : SUCCESS in 9m 15s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9bf9c969abee40cbb5b00e1c177aeb11 : SUCCESS in 5m 32s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/558350bf06e64977a23d6ad5782f0280 : SUCCESS in 1h 26m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce779d74ad3c4180a02b4a181a66069f : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":3},{"id":"61ff38d42475447211f21f0b4fe151a199ade23e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-10 13:49:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3e44f7fb66d073487b9994ef77bab62be54fc5af","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-10 14:39:56.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"5822b0a8d617f67ff8e97296d91ec5f21e275e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:33:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5abc061f049c47eda4e85c5d325a312f : SUCCESS in 1h 16m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/14e5452b57cf4a3ca34b49bd42ef63d2 : SUCCESS in 53m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54182c7472ec4fe1980d7f045c726eea : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0bfe13ef14049c9a360cfa3c0d8bc55 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/444ced3b829d4aa5af75a39687ecc4fc : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64be4e198e144e5cbf083bb110faa0e1 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/745616b8b98a42a1b631313cc3a181d7 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a0f9173929b4473a12505c283dae6eb : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ceeaa7d77074413684616be8555b4208 : SUCCESS in 8m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cdad8f20968a48058cefd2379893566d : SUCCESS in 8m 47s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/49f61666f9434c558d6bc1af3f4ae72c : SUCCESS in 5m 18s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d0e92a97938d4a0f8283eb55085080cb : SUCCESS in 1h 04m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e58b518e00f4feab9db352adfc2e387 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":3},{"id":"af652ac74cfbd395f41891864a67dd4f2e6f68fe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-13 05:21:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"17d9cbd02ef98bb58cff4754e93c353c4783ef3f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-13 19:36:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)\n\nSome nits noted inline, but they can be followups if you decide to do them.","accounts_in_message":[],"_revision_number":3},{"id":"901e5cdbea02535c39063d6a1a3f812baf73ff03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 19:36:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"21426563f8015be586fbc58db7c60829f8f0e412","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-13 19:42:02.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nwill discuss about the reno placement, cheers.","accounts_in_message":[],"_revision_number":3},{"id":"8ca00ed224e9c86db62c8ad0d4440de961795641","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 20:56:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1abf7e4a74d34e3293256f66388693ba : SUCCESS in 1h 09m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77b4b82dea5847e1b863ff1e237ff174 : SUCCESS in 1h 13m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d754d05b15c4c879495d4368291ccb2 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91975685c3c4444e90a7788a45b73958 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bd444eab6cd443c91367953f9a4ab5e : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b54104784ea54eb5a6edfb9bc4db18af : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9b9a51d4c1347728f18cfccbb748ae1 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dcf3d0cf37c4c8b8dceb525eb58dbcd : SUCCESS in 8m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2be6be678cb24e3b8f5848151032173e : SUCCESS in 10m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9c545f3c0c614314a207b64a65543366 : SUCCESS in 13m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dee13bd9c8d640f19fe53ecc3878e95b : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":3},{"id":"557fcc75208d428097acc2e15e67e231be71c0ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 20:56:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"45b341a32a521b52559a6d6c8bd069486639e050","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 20:58:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f96a7acda6db4f0c8fcd19334cd22185 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83da6883df1743b1bacf0435c9b3bec9 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bbb3ede89596fb8d9dd10754f91dcdac76bd52f1","revisions":{"f8d42a3aa233ff5caf3a1f8a7d6eabcbce2741a2":{"kind":"REWORK","_number":1,"created":"2020-06-30 16:49:54.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/72/738672/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/738672/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/738672/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/738672/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/738672/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/738672/1"}}},"commit":{"parents":[{"commit":"34e0489167eb4955d1c74c4e048ac910259b469a","subject":"Deprecation cleanout Registry and related","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/34e0489167eb4955d1c74c4e048ac910259b469a"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-06-30 15:44:10.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-06-30 15:44:10.000000000","tz":60},"subject":"Removal of \u0027enable_v2_api\u0027","message":"Removal of \u0027enable_v2_api\u0027\n\nRemoval of config optio \u0027enable_v2_api\u0027 and it\u0027s related\noperations and tests.\n\nChange-Id: Ic83e7e8077b3fc939990c1f8e9c99b13b7fcd7ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f8d42a3aa233ff5caf3a1f8a7d6eabcbce2741a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f8d42a3aa233ff5caf3a1f8a7d6eabcbce2741a2"}]},"branch":"refs/heads/master"},"1a92766a6f7c20d96cfdfbcf9c03ec3002f5733e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-30 19:43:26.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/72/738672/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/738672/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/738672/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/738672/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/738672/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/738672/2"}}},"commit":{"parents":[{"commit":"673666cbf4f5ee97b28f157f3b1e7115b72c5fee","subject":"Deprecation cleanout Registry and related","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/673666cbf4f5ee97b28f157f3b1e7115b72c5fee"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-06-30 15:44:10.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-06-30 19:42:55.000000000","tz":60},"subject":"Removal of \u0027enable_v2_api\u0027","message":"Removal of \u0027enable_v2_api\u0027\n\nRemoval of config optio \u0027enable_v2_api\u0027 and it\u0027s related\noperations and tests.\n\nChange-Id: Ic83e7e8077b3fc939990c1f8e9c99b13b7fcd7ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a92766a6f7c20d96cfdfbcf9c03ec3002f5733e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a92766a6f7c20d96cfdfbcf9c03ec3002f5733e"}]},"branch":"refs/heads/master"},"bbb3ede89596fb8d9dd10754f91dcdac76bd52f1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-07-03 19:01:43.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/72/738672/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/738672/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/738672/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/738672/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/738672/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/738672/3"}}},"commit":{"parents":[{"commit":"673666cbf4f5ee97b28f157f3b1e7115b72c5fee","subject":"Deprecation cleanout Registry and related","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/673666cbf4f5ee97b28f157f3b1e7115b72c5fee"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-06-30 15:44:10.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-03 19:01:39.000000000","tz":0},"subject":"Removal of \u0027enable_v2_api\u0027","message":"Removal of \u0027enable_v2_api\u0027\n\nRemoval of config option \u0027enable_v2_api\u0027 and its related\noperations and tests.\n\nChange-Id: Ic83e7e8077b3fc939990c1f8e9c99b13b7fcd7ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbb3ede89596fb8d9dd10754f91dcdac76bd52f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbb3ede89596fb8d9dd10754f91dcdac76bd52f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
