)]}'
{"id":"openstack%2Fglance~738737","triplet_id":"openstack%2Fglance~stable%2Fussuri~I385f222c5e3b46978b40bdefdc28fcb20d9c67d3","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1884587-stable/ussuri","hashtags":[],"change_id":"I385f222c5e3b46978b40bdefdc28fcb20d9c67d3","subject":"Check authorization before import for image","status":"MERGED","created":"2020-07-01 06:46:09.000000000","updated":"2020-07-02 08:04:49.000000000","submitted":"2020-07-02 08:02:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738737-1593676959714-326f21e0","meta_rev_id":"9799f05d5ffaddadc11ac9812339da777fc2a8b9","_number":738737,"virtual_id_number":738737,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2020-07-02 04:25:26.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 08:02:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2020-07-01 10:44:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2020-07-01 10:44:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-07-01 06:46:09.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-01 06:46:09.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-07-01 10:44:20.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-02 08:02:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"240f2c98f604de3e8533a591e03732913778644e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-01 06:46:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2650f102603e3f119ef1deff9d51586aa60932cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 08:37:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/fb00c39057b0483589b03639080b8193 : SUCCESS in 1h 11m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aa2eab21e9184a0cb9cde6d782e6fc3d : SUCCESS in 1h 10m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffa75958691841aaa8bbc047dceeb912 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a74905ec96b24714a6f6a19ec33e0535 : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb59afbb58db4fbc906c707fa8d98d9e : SUCCESS in 9m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e8c1bb9f2754d7db8f542ace3d3c92c : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/303ab70dd6074b1ca3df403a3989ba97 : SUCCESS in 10m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ffb7e9988b544a6a1d6b2106e49df96 : SUCCESS in 15m 21s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4d2e963606e04dd3980722545623d954 : SUCCESS in 16m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ee6168c002f149848dc0e15e5c627a3b : SUCCESS in 8m 43s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c750c4ef5ca5437bbee55e3f00eca482 : SUCCESS in 1h 30m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00cbb76f0cd644299abcf0f39c44b919 : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f912e2ad09a1d3b2e1a29ce382a53fc51b0533f0","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-01 10:44:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b491c4eb1af4288c5a7f4420beb69bdca7938bcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 10:45:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2fb3177753bb76bd73e2d3f695a4dea369e0228","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 13:27:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/97e3d9bc1a5f44dfbb5466c245be53b2 : SUCCESS in 1h 22m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf6d574389d74b34a82569be95939682 : SUCCESS in 1h 12m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee67ab12c8084c4ca5f5a5b3f5bd5d28 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee35ce2d72b4207b2f341de305e232f : POST_FAILURE in 34m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13b1b7e95b434117a02c0e0f2cc7e9ff : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee1423145b3d40b19bb7a4ce39f97194 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd29d3643474d9b907fab7bc4a53558 : SUCCESS in 11m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/548fc197608f4b3093a956e1d8c84b66 : SUCCESS in 7m 46s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1f860796f8174bc0a5f4ea4be23c9821 : SUCCESS in 27m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b34ac816d308483a84973706b01eed6a : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":1},{"id":"964818830ed60a6cbf5c133884ac138ab40de748","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-02 04:25:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b793bf1cdb1bfefe7ff1f6a43c4d8965bd03bebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 05:34:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f97a628b3b3e49859d3a7198e2abed63 : SUCCESS in 47m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e467415b3554a46b5b0da4678efb470 : SUCCESS in 47m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb3b0d424cb04d659cd087c1420417c2 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bd840a20269423584a53fee745cc02f : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcd62f1fc943484dbd82f9be13d56d99 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc815c32465c4dc79bbe5e551dbac36d : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e368b924a4dd490284716120682151cc : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/52f07577ce99483db428f44434199eef : SUCCESS in 6m 45s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ff1f0fb837b64cec85d950c36a409ee1 : SUCCESS in 6m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/acc4cba555654b38bee0d3b298198708 : SUCCESS in 4m 12s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8d025b73f49e485e9dff44004f862062 : SUCCESS in 1h 08m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fe2eec0dc374e79a4efe88ae31d06e4 : SUCCESS in 48m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c373b9af0a759a3d5aa5a7bd993939efc56930a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 05:34:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d01f390aed1ecd93ec74c40747399660a901a7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 08:02:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7042139139cb4a5d9c3a4a0bf5a84b04 : SUCCESS in 1h 23m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f29776513c042118e763b7c95c0612b : SUCCESS in 1h 23m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2f2d24688f44af8a046ca7d1766cbad : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1482e9b92b24638a0648e3228c6c688 : SUCCESS in 10m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f79aca5f085248ddafd8818cf6383cc9 : SUCCESS in 9m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5aec8c8e8aa445291d6545978f825a3 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1989ea6779914b79b77e71120ca81c79 : SUCCESS in 10m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b8df0fa70bb4ccb8a8a1e1b921cf83c : SUCCESS in 11m 04s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b72c3809a26e4a999d84b30deadf9e25 : SUCCESS in 10m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ac0dd668ffe4c66aaf8ab2ef8510721 : SUCCESS in 43m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7be299afda31c84a2c6a6ec2164685041ffc317d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 08:02:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4a1ae443856f83401eaf8a145e3ea771b801bb9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 08:04:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48881aed45bd47b1a94fdea4878ee2fd : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcba03989feed85a3301e53b252f766eabeeba3c","revisions":{"fcba03989feed85a3301e53b252f766eabeeba3c":{"kind":"REWORK","_number":1,"created":"2020-07-01 06:46:09.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/37/738737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/37/738737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/738737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/738737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/37/738737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/37/738737/1"}}},"commit":{"parents":[{"commit":"c3fd60f70b07ddf7d2b79c5fc2704bf1619f4c96","subject":"Exclude http store if --all-stores specified for import/copy operation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c3fd60f70b07ddf7d2b79c5fc2704bf1619f4c96"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-06-23 14:12:12.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-01 06:46:09.000000000","tz":0},"subject":"Check authorization before import for image","message":"Check authorization before import for image\n\nRight now we only check to see if the user can see the image before\nwe kick off an import operation. However, that will never work unless\nthe user is the *owner* of the image (or an admin) which means we\nreturn a 202 to the API caller and then the task fails immediately.\n\nThis change makes us check that authorization up front and return an\nappropriate error to the user so they know it failed, and avoid\nstarting a task destined for failure.\n\nNote that there was already a check for a Forbidden result when calling\nthe import API. However, that used a context.owner\u003dNone which could never\nhappen in reality. A more suitable check would have been to use a context\nwith a different real owner, but it turns out that the task creation\nwould have succeeded in that case as well. This test is changed to use\nan alternate owner and ensure that we get the forbidden result from the\nnew check immediately.\n\nChange-Id: I385f222c5e3b46978b40bdefdc28fcb20d9c67d3\nCloses-Bug: #1884587\n(cherry picked from commit c930638fcf58b43d53903cfc30e06fd6919bdad6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcba03989feed85a3301e53b252f766eabeeba3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fcba03989feed85a3301e53b252f766eabeeba3c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
