)]}'
{"id":"openstack%2Fglance~739056","triplet_id":"openstack%2Fglance~stable%2Ftrain~I1aed6ba9054f8940d55e9fdb76bcd0615d00889c","project":"openstack/glance","branch":"stable/train","topic":"native-zuulv3-migration)","hashtags":[],"change_id":"I1aed6ba9054f8940d55e9fdb76bcd0615d00889c","subject":"zuul: switch to the \"plain\" grenade job here too","status":"MERGED","created":"2020-07-02 14:37:02.000000000","updated":"2020-07-03 13:31:18.000000000","submitted":"2020-07-03 13:26:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739056-1593782768178-ea166436","meta_rev_id":"9e07219820679f99726f98ada43c3200815a34e1","_number":739056,"virtual_id_number":739056,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-03 13:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-02 14:47:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-03 11:52:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-03 11:52:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 14:47:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-03 11:52:13.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-03 13:26:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25bcb307a5d8e2a62394a15f2816936ff036acde","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-02 14:37:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9f8edba53c1913e671bd469a0b7e05b7d1c5f4b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-02 14:38:17.000000000","message":"Topic set to native-zuulv3-migration)","accounts_in_message":[],"_revision_number":1},{"id":"8d4bf06ae525f320d6bccb2d9c68ba7de376ec0b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-02 14:47:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de32f8d1f625dfbbda5ca64798eedbf097efe9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 20:14:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/61fa04424ed54a48bde61f3dd9b2e238 : SUCCESS in 1h 05m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/077e40c954b842c9850d8e969381e332 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/381a13de024840babc0fd80e1e27ab1d : SUCCESS in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54301bdb103748d18b6c46731cfe52c3 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22a7026f769343098b3f86e51ee0edaf : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/341e4409487f4a289b6f66b5bf62aace : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41bf5e7b065f42b28bb81145f3bedf80 : SUCCESS in 9m 26s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9c0f793688764cec97093aa168fa3f75 : SUCCESS in 19m 06s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/3ff2de0790f74da1a08919fed3b66164 : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4392ccd244b8464d5caefdeb15bdfd46d5448633","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-03 11:52:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks Luigi!","accounts_in_message":[],"_revision_number":1},{"id":"4e23a21fd84b5345d9cef5d0b385307974a9d129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 11:52:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c18a28daf05670c90a5d979cd564e68050aefb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 13:26:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1609c460c76248a8b7fd0ceda5456ba7 : SUCCESS in 54m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b6627e8ed5f410f86f5e28c0113ea6b : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd7da3e5f9464f61aa2d104cd08d4d27 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58e7c36b0c56478697f72cb3e8fb77b7 : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb16d7d4f8a545a08cce5d17391a5c3d : SUCCESS in 9m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a45eae176284b6592670b8405b59a87 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4441e9773591470e9164f81c01f19d1a : SUCCESS in 11m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5c12e4ddde9943d1b22799d09201e266 : SUCCESS in 16m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/060592d4e648494190e7405d453cb431 : SUCCESS in 16m 23s","accounts_in_message":[],"_revision_number":1},{"id":"79aec7a176017188e30d9e537462742fb814ebcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 13:26:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9e07219820679f99726f98ada43c3200815a34e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 13:31:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/830e5ab6d3614b668104b34bfd84649c : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8abe20d520b977fa16f1e60e108530b2880bad92","revisions":{"8abe20d520b977fa16f1e60e108530b2880bad92":{"kind":"REWORK","_number":1,"created":"2020-07-02 14:37:02.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/56/739056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/739056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/739056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/739056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/739056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/739056/1"}}},"commit":{"parents":[{"commit":"3e16bc7d40922f4e6176486b5e7af42989a50774","subject":"Merge \"fix properties\u0027 missing underline for VirtCPUTopology\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e16bc7d40922f4e6176486b5e7af42989a50774"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-02 14:35:37.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-02 14:35:37.000000000","tz":120},"subject":"zuul: switch to the \"plain\" grenade job here too","message":"zuul: switch to the \"plain\" grenade job here too\n\nThe native Zuulv3 grenade jobs have been recently backported\nto stable/train. Use them instead of the old neutron-grenade.\n\nChange-Id: I1aed6ba9054f8940d55e9fdb76bcd0615d00889c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8abe20d520b977fa16f1e60e108530b2880bad92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8abe20d520b977fa16f1e60e108530b2880bad92"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
