)]}'
{"id":"openstack%2Fglance~739376","triplet_id":"openstack%2Fglance~stable%2Ftrain~I845bce79e487656a284e52e4c605d648a5411822","project":"openstack/glance","branch":"stable/train","topic":"native-zuulv3-migration","hashtags":[],"change_id":"I845bce79e487656a284e52e4c605d648a5411822","subject":"Use grenade-multinode instead of the custom legacy job","status":"MERGED","created":"2020-07-05 23:06:42.000000000","updated":"2020-07-08 03:33:55.000000000","submitted":"2020-07-08 03:30:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739376-1594179057803-de16a12a","meta_rev_id":"08486e5bd991eed003597ffd18e10025f2fa32ae","_number":739376,"virtual_id_number":739376,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-07-07 20:55:01.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-08 03:30:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2020-07-07 14:46:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-07-06 13:49:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2020-07-07 14:46:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 13:49:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-07 14:46:54.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-08 03:30:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1e4e90e617aa19774e95ad0c0ccd2400972a835","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-05 23:06:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0354b20dcfbef85c2fa8ac88735996f11993352","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 00:40:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/eb39dea2f73647cab190f59d659cb450 : SUCCESS in 1h 32m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3b4df34a6b0e423eab8c552f66db762d : SUCCESS in 1h 01m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/2cf8eaa578b34ebc961379e0ec148d56 : SUCCESS in 58m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/445205fd84eb42e1b3f971456d5d0afb : SUCCESS in 1h 11m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93d3ca3e87204f519ce0daab52f2d878 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645e7011b341403aac21d7a940b5df04 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85aa7b430ce446ab93b831e4d1374038 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbf0c8c9e32c4344ba4de029694459cf : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b36325d0b1b2405681e9c7f39e2173a8 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96feb0410e144bd484bca89d0f20be5f : SUCCESS in 6m 42s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4fedd758586042778eeac6cc8a66aaf2 : SUCCESS in 8m 24s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/79cd8528355949d4a8ed1690775fe838 : SUCCESS in 9m 10s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/69758f6563574bf7b0cfd0f9893cee95 : FAILURE in 5m 01s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2f5a3eab4c6642dba3017f58bef82a85 : SUCCESS in 4m 26s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3b0bc6c306574a238782ddd6f92514eb : SUCCESS in 1h 11m 09s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c3e105f6e6e74bf396cee64e3b7d1622 : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a9167224113161274035a61750eb373ca818cebf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-06 13:49:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a27c6815ff4f3fe810a374b50487c44d0408daa","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-07 14:46:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks Luigi!","accounts_in_message":[],"_revision_number":1},{"id":"d5dc76e4210f60bba0c25f8ab44c8a2aba0d5f69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 14:49:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f191bfbbce583bdb649b59eb689767b891de25cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 17:01:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/af2aeb9210894d0f95561348b8dc18c2 : FAILURE in 1h 32m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c565356961ce4ef8ad289f319b8c3b71 : SUCCESS in 42m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/a269b7dc824c4ea780c0c254f82063c0 : SUCCESS in 1h 10m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6863d603933c43dcb5e4d919a4900048 : SUCCESS in 1h 51m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65cfea9c7ce646ac87d09def63cb442a : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d776ef68ff44cd787d67818e2d706bd : SUCCESS in 8m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74ca38fac56b4a008618eab55ea4a46d : SUCCESS in 9m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74c37364d2334a3b8a2709a364fd5bb2 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4dff4022b8e14e10a0809e076403a7b9 : SUCCESS in 12m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae230a3a6989404ea535224f1c0402e9 : SUCCESS in 10m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/767cc139971046a5bfd2feee018b6bc7 : SUCCESS in 13m 55s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2e8b8843319d4745a6348562e9f33ce4 : SUCCESS in 20m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8587aae847c94942b2274382b3f32747 : SUCCESS in 1h 01m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f081a28a3f3f604f47cacdb6ab02ca3c7cd61d0b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-07 20:55:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c61f03c82344f016a0ca681907b06438042312ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 01:09:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c27546c683174d4aa709652b07944601 : SUCCESS in 1h 45m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2c2cfd559991404480d718f1959ace0a : SUCCESS in 1h 08m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/97fbbc73c0054d63a411fdaeba181857 : SUCCESS in 1h 13m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1cd451e26b3040e197b534c91d734620 : SUCCESS in 1h 25m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9a83c674a7c4c8baaeda62a37108ce1 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3d6176ea4b54baaa7cafe8053f31582 : SUCCESS in 10m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d07a08e74114665848a8c9b73b267a3 : SUCCESS in 10m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/996146d0bc634750b9c437589d650212 : SUCCESS in 10m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b67b03cd0a64011aebc896099fc5567 : SUCCESS in 10m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b06fd2e835a4936a49a239cdab823ca : SUCCESS in 9m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/02345e20200b432e97c701406cc3ee2c : SUCCESS in 11m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/af1a0b73ea8a4c4a9d8a2f181a322ef4 : SUCCESS in 12m 37s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/871d7ab7a0b14bc2ba03905ae439273d : FAILURE in 8m 18s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/906894d1b00549e9a443b986b9a9426c : SUCCESS in 7m 13s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e9138266dffd4cafbc2010f660cdda9c : SUCCESS in 1h 17m 31s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4c9ba379fd5a46199360c6f213748584 : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2ddfc9945e22742139df7067fb2182ddf4826bca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 01:09:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d03aac9a5efdf6ca595b337e0619bc3c1aedc092","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 03:30:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/4dcc65cb50d34746b02a20afa3db8bcd : SUCCESS in 1h 23m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3f04a32c23be4a71a7e0f51dbc83e5d8 : SUCCESS in 56m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/711293e65bca40408cd84d0fe9cb6ead : SUCCESS in 51m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d72e8a75375640d1920c205ded2fa890 : SUCCESS in 1h 17m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9773a05eabf24e86aa10397363f039d2 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb48e802c364c46805de61c2f09c8f7 : SUCCESS in 7m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e006a5a43ae402fbf74412bd3df6c0b : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d901c4137c7a43239ee22608e39c9af1 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6aef3886aa4143daa233e13cc2c43ef4 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d6ac2f696240349fa2dea164a756d5 : SUCCESS in 8m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1636a1172bbd4291a5d27234afb3128b : SUCCESS in 13m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9ee787a02df84159aeab1697a6aab663 : SUCCESS in 13m 32s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5d43177211074de797839089d52ab68d : SUCCESS in 1h 00m 04s","accounts_in_message":[],"_revision_number":1},{"id":"424b1e9de8a7781e48f1d58b6722487505fc827a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 03:30:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4204e50920fe04745b0491f86e90cde2c2fbb5ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 03:33:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e394c9b9d40b4213bf33f1be76395c28 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdba1feacb7fa088650667e69290aa6e9bfad65c","revisions":{"bdba1feacb7fa088650667e69290aa6e9bfad65c":{"kind":"REWORK","_number":1,"created":"2020-07-05 23:06:42.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/76/739376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/739376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/739376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/739376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/739376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/739376/1"}}},"commit":{"parents":[{"commit":"8abe20d520b977fa16f1e60e108530b2880bad92","subject":"zuul: switch to the \"plain\" grenade job here too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8abe20d520b977fa16f1e60e108530b2880bad92"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-06-02 22:56:11.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-05 23:05:51.000000000","tz":120},"subject":"Use grenade-multinode instead of the custom legacy job","message":"Use grenade-multinode instead of the custom legacy job\n\nglance-dsvm-grenade-multinode is a simple multinode grenade job and\nit can thus be replaced by the existing grenade-multinode.\n\nConflicts:\n\t.zuul.yaml\n\tplaybooks/legacy/glance-dsvm-grenade-multinode/run.yaml\nConflicts due to the evolution of the 2 files over the releases.\n\nChange-Id: I845bce79e487656a284e52e4c605d648a5411822\n(cherry picked from commit a506e941152d1dacba57c800c34623d9ad02bebe)\n(cherry picked from commit 0855cc1eaa6b25b823db3253afee10ca13fc39c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdba1feacb7fa088650667e69290aa6e9bfad65c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdba1feacb7fa088650667e69290aa6e9bfad65c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
