)]}'
{"id":"openstack%2Fglance~739403","triplet_id":"openstack%2Fglance~stable%2Fussuri~I44bfefb6eee421e18e5e95a0dafaef0ea4e170da","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1885003-stable/ussuri","hashtags":[],"change_id":"I44bfefb6eee421e18e5e95a0dafaef0ea4e170da","subject":"Fix: Interrupted copy-image leaking data on subsequent operation","status":"MERGED","created":"2020-07-06 05:43:06.000000000","updated":"2020-07-06 16:41:35.000000000","submitted":"2020-07-06 16:37:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739403-1594053478027-de16a12a","meta_rev_id":"48284c0a5dc967f6dce28fb3704696c7a9adc0ef","_number":739403,"virtual_id_number":739403,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 16:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-06 12:33:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-06 15:06:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-06 15:06:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-07-06 05:43:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 12:33:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-06 15:06:51.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-06 16:37:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d6c71cb5d2e70e5bf84163c14159b010a3eed04","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-06 05:43:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4b9deb2e6b0e4c337ab752a89d6cfddbc987376f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-06 05:43:15.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"4c0f4c7c4d8b9b489fc0b5bb0d6a07de6f7a5f97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 06:58:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f4ae45b9c8d04d4e83de27d9f860d646 : SUCCESS in 1h 14m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e070b6beb3734a60a4230a297e8596ae : SUCCESS in 1h 11m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53a743e9731b4317bfeac9cec754c422 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26747b78d8bc4adc819378d292e2156d : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/312fa24587e743afbbe7f58271e207af : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7cd62f9b971945ee89dc2c5b40087ab2 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/955422da1e674e35b8849ef8994c3af0 : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/598c6cd2a0c841c78b4550ab1fa79418 : SUCCESS in 7m 16s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c988c91621b3499e90bae7170c8f9385 : SUCCESS in 8m 11s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/de0ef856d25441988803a4c0023782d4 : SUCCESS in 4m 24s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b41ab8c826fb469d88e83ac874ffbbd4 : SUCCESS in 1h 05m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d700234955214dbdbb7df94e2872a966 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":1},{"id":"34eb4152f414bb7b80fc4c88ae1a0de3de296137","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-06 12:33:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"765852dc1326886b6c4ca0b2e5f245b61e60e6c6","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-06 15:06:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73081510cc078a2cc5b64368f8144bc1f10382aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:07:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0e7b333e7dfe9291e07b3c451e444d38c292985","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:37:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/830c2cd5453b45ebaf21c737f12338b3 : SUCCESS in 54m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e0f38aa5b2f48139867ed318cc375b3 : SUCCESS in 1h 12m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/238299824ec741e691a6c7e5e2a6287d : SUCCESS in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01b1b323d54342f5968ebbfd4984c3c5 : SUCCESS in 11m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fb22f65f48b4293bdc4fd427d3aafc1 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ca7cd894521455ab597f486856d4413 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5750e1ede0f0421b92e3251ab3ee76d5 : SUCCESS in 8m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ea9faaecc07d42b9a101058e7ea9fe3f : SUCCESS in 10m 04s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c29929fb50b04475a715d922c2a5754c : SUCCESS in 10m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c7b95a003a7460faf0f7225a60be99e : SUCCESS in 1h 17m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8f8368ed8a7366f16856488914b9c479a6a54cea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:37:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"19d7dcdaebd3ca3a9e4cd7c23d328539efd01cae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:41:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5e47bf0a184549baba1ab7691e3fa7 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72510f9ce90970e75a95492c7647fd63ac2636dc","revisions":{"72510f9ce90970e75a95492c7647fd63ac2636dc":{"kind":"REWORK","_number":1,"created":"2020-07-06 05:43:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/03/739403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/03/739403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/739403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/739403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/03/739403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/03/739403/1"}}},"commit":{"parents":[{"commit":"75862b395c271a3b8e55d6db105ac42ed06fe141","subject":"Merge \"Use grenade-multinode instead of the custom legacy job\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/75862b395c271a3b8e55d6db105ac42ed06fe141"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-06-24 19:44:54.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-06 05:43:06.000000000","tz":0},"subject":"Fix: Interrupted copy-image leaking data on subsequent operation","message":"Fix: Interrupted copy-image leaking data on subsequent operation\n\nIf copying existing image in other stores fails while staging the data to\nstaging directory due to power, network or any other reason. Then\nsubsequent try may lead to data leaks in stores.\n\nTo fix this, added check of the actual image size with the size\nof image file present in the staging area. If it does not match\nthen delete the image file from staging area so that the entire\nimage will be staged again.\n\nChange-Id: I44bfefb6eee421e18e5e95a0dafaef0ea4e170da\nCloses-Bug: #1885003\n(cherry picked from commit 22d8f1fcbf4325d79ac18adfe87198c785f1df52)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/72510f9ce90970e75a95492c7647fd63ac2636dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/72510f9ce90970e75a95492c7647fd63ac2636dc"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
