)]}'
{"id":"openstack%2Fglance~740108","triplet_id":"openstack%2Fglance~master~Iaa0ce6e1fde5c4c546b2f4a53969287b05d295de","project":"openstack/glance","branch":"master","topic":"visibility-all-filter","hashtags":[],"change_id":"Iaa0ce6e1fde5c4c546b2f4a53969287b05d295de","subject":"Add \u0027all\u0027 visibility filter for listing images in docs","status":"MERGED","created":"2020-07-08 20:27:19.000000000","updated":"2020-07-10 15:49:30.000000000","submitted":"2020-07-10 15:44:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740108-1594395890909-9763ba5f","meta_rev_id":"8e21eccc0079b350a9788fc8a7acd0a90f9ca2aa","_number":740108,"virtual_id_number":740108,"owner":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-08 20:37:52.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},{"value":0,"_account_id":27515,"name":"Florian Engelmann","email":"engelmann@d-g-c.de","username":"engel75"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 15:44:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-08 20:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-10 13:54:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-07-09 12:32:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},{"value":1,"date":"2020-07-08 21:35:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27515,"name":"Florian Engelmann","email":"engelmann@d-g-c.de","username":"engel75"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-10 13:54:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},{"value":0,"_account_id":27515,"name":"Florian Engelmann","email":"engelmann@d-g-c.de","username":"engel75"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},{"value":0,"_account_id":27515,"name":"Florian Engelmann","email":"engelmann@d-g-c.de","username":"engel75"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27515,"name":"Florian Engelmann","email":"engelmann@d-g-c.de","username":"engel75"},{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 20:37:52.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-08 20:39:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-08 21:35:38.000000000","updated_by":{"_account_id":27515,"name":"Florian Engelmann","email":"engelmann@d-g-c.de","username":"engel75"},"reviewer":{"_account_id":27515,"name":"Florian Engelmann","email":"engelmann@d-g-c.de","username":"engel75"},"state":"REVIEWER"},{"updated":"2020-07-09 12:32:33.000000000","updated_by":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"reviewer":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"state":"REVIEWER"},{"updated":"2020-07-10 13:54:14.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-10 15:44:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35d6126d8822955eef4426b72d132f5fe883a465","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-07-08 20:27:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51ca373ec93bd9dd148712e5a464a08f5293b6ba","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-08 20:27:41.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"911148472283d50ca6b6bbad0361f735a0de44bd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-08 20:33:46.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks for updating the docs. If you can reformat the commit message, that would be great.","accounts_in_message":[],"_revision_number":1},{"id":"0f3c30eb86d2a93dd741504f8f4e6b0c9f169d8b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-08 20:33:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f94aec28ecc05bd134229dccff413e966a616a77","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-07-08 20:34:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5ea45cf5c8e8638f05fb349b9359c844807c52c8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-08 20:34:38.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"2a6cba5de7bbfde5f41a9fa3baac127465e6c2c2","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2020-07-08 20:37:24.000000000","message":"Patch Set 1: Code-Review+1\n\nNice catch. This change makes the documentation more accurate with the available options when listing images.","accounts_in_message":[],"_revision_number":1},{"id":"df6bc396d4486623d004e073ee8d6c9e36d484dc","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-07-08 20:37:31.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4a81284d4f65d3c741abc9b23593d806c48d119c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-08 20:37:52.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"908326e6fff807bee2d811f927915e4f6c755d24","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-08 20:39:05.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"c639e51b37a8a5795bbca9c94baf8cf351ef3962","author":{"_account_id":27515,"name":"Florian Engelmann","email":"engelmann@d-g-c.de","username":"engel75"},"date":"2020-07-08 21:35:38.000000000","message":"Patch Set 3: Code-Review+1\n\nSaved me a lot of time!","accounts_in_message":[],"_revision_number":3},{"id":"c46f042854ea30a947a59da70c0d68b621dbc25c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 22:51:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cd284b3474644ac4a37327413531ae47 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c96f94a785b94089941ee92b92ca9c12 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f58d160a13d8469caedb3c98986d4eec : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":3},{"id":"6c8d4bfd5b5e9326bb24ed9eef8ea9b19709cdde","author":{"_account_id":30695,"name":"Pedro Henrique Pereira Martins","email":"phpm13@gmail.com","username":"pedrohpmartins"},"date":"2020-07-09 12:32:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a735650a5f2d26728d46e4a77b6b549d7c426345","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-10 13:54:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks Rafael!","accounts_in_message":[],"_revision_number":3},{"id":"da05080d2344f1a38e46a896029343b36e9ef3ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 13:54:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4f5392d08adfed886eef832f0879da8cb469007e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:44:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cc72bd601f06cb839d2f2cf9a9f47fd4122ac0d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:44:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cd0d448019e24fd388f241da885ba5c0 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d2defbdada4e45b8f0de614f5d3c43 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0517c740bf74418a86eb64f226dae17 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":3},{"id":"97f6b0a69a90ed3cd577a9c4368fd6a8ca047551","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:49:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/82f08562d13842f198afaa8116a18610 : SUCCESS in 2m 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf6d70be29a4462db5559cb4b16c6b01 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e2e050c0b1fc9b33ea78a204f38530136e59e7c5","revisions":{"d32d5e3527c4b221a2d852e6c472944bb13ad9b3":{"kind":"REWORK","_number":1,"created":"2020-07-08 20:27:19.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/08/740108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/740108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/740108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/740108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/740108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/740108/1"}}},"commit":{"parents":[{"commit":"63d982588f4c4b290985ca5fe5b288c2cd3579df","subject":"Merge \"Update lower-constraints versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63d982588f4c4b290985ca5fe5b288c2cd3579df"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-07-08 20:26:51.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-07-08 20:27:16.000000000","tz":-180},"subject":"Add \u0027all\u0027 visibility filter for listing images in docs","message":"Add \u0027all\u0027 visibility filter for listing images in docs\n\nSince this commit https://github.com/openstack/glance/commit/cfa9b6e8e814a52181d4e27fd23a659e8489dee8,\nthe visibility all is support in the listing of images. However, the documentation does not present it. This can make it hard for operators to be aware of this option when listing images. Therefore, we\npropose to add it in the documentation\n\nChange-Id: Iaa0ce6e1fde5c4c546b2f4a53969287b05d295de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d32d5e3527c4b221a2d852e6c472944bb13ad9b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d32d5e3527c4b221a2d852e6c472944bb13ad9b3"}]},"branch":"refs/heads/master"},"03a73dbd7f6805c581213505f691009536766595":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-08 20:34:17.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/08/740108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/740108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/740108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/740108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/740108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/740108/2"}}},"commit":{"parents":[{"commit":"63d982588f4c4b290985ca5fe5b288c2cd3579df","subject":"Merge \"Update lower-constraints versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63d982588f4c4b290985ca5fe5b288c2cd3579df"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-07-08 20:26:51.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-07-08 20:33:34.000000000","tz":-180},"subject":"Add \u0027all\u0027 visibility filter for listing images in docs","message":"Add \u0027all\u0027 visibility filter for listing images in docs\n\nSince this commit https://github.com/openstack/glance/commit/cfa9b6e8e814a52181d4e27fd23a659e8489dee8,\nthe visibility all is support in the listing of images. However,\nthe documentation does not present it. This can make it hard for operators\nto be aware of this option when listing images. Therefore, we propose to add\nit in the documentation.\n\nChange-Id: Iaa0ce6e1fde5c4c546b2f4a53969287b05d295de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03a73dbd7f6805c581213505f691009536766595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03a73dbd7f6805c581213505f691009536766595"}]},"branch":"refs/heads/master"},"e2e050c0b1fc9b33ea78a204f38530136e59e7c5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-07-08 20:37:31.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/08/740108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/740108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/740108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/740108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/740108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/740108/3"}}},"commit":{"parents":[{"commit":"63d982588f4c4b290985ca5fe5b288c2cd3579df","subject":"Merge \"Update lower-constraints versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63d982588f4c4b290985ca5fe5b288c2cd3579df"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-07-08 20:26:51.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-07-08 20:36:58.000000000","tz":-180},"subject":"Add \u0027all\u0027 visibility filter for listing images in docs","message":"Add \u0027all\u0027 visibility filter for listing images in docs\n\nSince this commit https://github.com/openstack/glance/commit/cfa9b6e8e814a52181d4e27fd23a659e8489dee8,\nthe visibility all is supported in the listing of images. However,\nthe documentation does not present it. This can make it hard for operators\nto be aware of this option when listing images. Therefore, we propose to add\nit in the documentation.\n\nChange-Id: Iaa0ce6e1fde5c4c546b2f4a53969287b05d295de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2e050c0b1fc9b33ea78a204f38530136e59e7c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2e050c0b1fc9b33ea78a204f38530136e59e7c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
