)]}'
{"id":"openstack%2Fglance~740384","triplet_id":"openstack%2Fglance~master~Idf9be5afea5d034b9679ad451306a8f2d365aa51","project":"openstack/glance","branch":"master","topic":"bug/1887099","hashtags":[],"change_id":"Idf9be5afea5d034b9679ad451306a8f2d365aa51","subject":"Fix metadefs for compute-watchdog","status":"MERGED","created":"2020-07-10 00:56:00.000000000","updated":"2020-08-29 03:04:26.000000000","submitted":"2020-08-29 03:02:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740384-1598670176972-bc9ee2cc","meta_rev_id":"cfa9b91420a20e2bd272c72392269cacb9c8691f","_number":740384,"virtual_id_number":740384,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-08-29 00:35:30.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-07-10 00:57:29.000000000","_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-29 03:02:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-28 19:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2020-07-14 07:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-28 19:08:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-29 00:35:08.000000000","_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2020-07-14 07:31:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-14 07:31:30.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-28 19:08:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-29 00:35:08.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-08-29 00:35:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-29 03:02:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ff7001f389d66f9e92bbe66390467d13ee9a5d4","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2020-07-10 00:56:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ed335481567bababbcc6bab4204cd2419eb247e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-10 00:56:21.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"d5aba91769f2cc9652555f56d8c34a40f0546a2f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2020-07-10 00:57:29.000000000","message":"Patch Set 1:\n\nTo be honest, I have not really tested the change and do not really know this feature.\n\nThe patch is based upon the feedback of a user.","accounts_in_message":[],"_revision_number":1},{"id":"3e353f27393fcffe6fd122d3fdd5c7aaa9224816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 03:23:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1da1b914d14a402f9e8717b038b606c1 : FAILURE in 2h 19m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f58a1f03cdcb4ee097eade646ed1dc67 : TIMED_OUT in 2h 12m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c5b8409b3204d35b47c6945e98ebc9c : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb9bd65f418c497fb9347bedeaeb9273 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca3ba9241f944550b1c545b476b9caad : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5506063a46448ff96f0369ac48c50c6 : FAILURE in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce4bcd2eeba4d2cb725baa8fe3507dd : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/42146c5aee3b48b2baca70be8f00b0cd : SUCCESS in 7m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a060389f77247b99408159dbbc638c5 : SUCCESS in 8m 20s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/287e4ce50a9c48fd86d6b56576aa4328 : SUCCESS in 1h 04m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4e7bfbeb4214fcab67d07a5a4b14328 : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6627090720b75ed8148a2a9a35cfe8076e60ce0d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-13 16:39:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c53d80b5d6cc76df9889d58dcd8a7aa6de34ec99","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-13 16:39:34.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"56e23761a4c1e1b35d9296d8f3e26161d83963c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 18:49:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4222b9bff1d249a0a4ea689947f698b2 : SUCCESS in 1h 14m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8badc53106e24273983942ff3934d173 : SUCCESS in 1h 26m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f158c7db9e43470499650e0ad303551a : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa69de73d1684aedb963d1dfd16328d4 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d4d43787dc54cea99eab1665f071209 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40cd24d125a44364a267c845188e99cf : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b48ddde1774247c3be6e8a556712d310 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c83730cad3d5491da78508a7179ec73b : TIMED_OUT in 31m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/05b9c71632e94cea9fa7c43a84f7f17a : SUCCESS in 13m 19s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e33ad89840754c16b8c01cd0805e3ab4 : SUCCESS in 1h 29m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c5748dacabd46b693f426a95bd790f5 : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2da922a94bf435ecc675425cba040f9db693e042","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-14 05:32:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d6e9311d0608c6d77de78ec19b314d455ea7337","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-14 05:33:10.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"21d03d19844f3e658c930c817ca0e03655347acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 07:21:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/80c09626769c46e0a19da3b8de9f3744 : SUCCESS in 1h 39m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86da57f2c4c74b6bbeb74ec9952d6784 : SUCCESS in 1h 47m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc68cfb0351d47f590771784d13f9e2d : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9ab13a3c13a4ce78a8d91dcf39182c9 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97fdca61a61c46cfa4fb44e0d80406a7 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc5ab14f94354e98b27866d0800367a8 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3367a74ef7d45498541219bad3417ac : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/483b1294977b4ebb92fb6133f6b05794 : SUCCESS in 7m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d6b05d8490ec4331beec4043cc47a53f : SUCCESS in 8m 09s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/48e931bcafd24ed498f05ceae0ddbf90 : SUCCESS in 1h 01m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/935163193c4a47d2978b3d7e35a6c057 : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c62fb816218d67b860547841902282a2a3ffd76d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-14 07:31:30.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"9aa86e3f80008969b9a2aa47612fc3da38891ab4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-28 19:08:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"906077c2686c2393483ea9f6cab394ce895cb1f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 19:08:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"229095a2772e9ca9863967fd77e1a6c74ad74fd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 22:49:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/41e7aaae076c4af9992f750d6d384418 : SUCCESS in 46m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9ed8ff64984a405c8c50db63c3e63250 : RETRY_LIMIT in 0s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e5b834d69ce4f139e684a8851904281 : RETRY_LIMIT in 0s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/768d4afe60da4b7f9a978fa501a8aa69 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7d9637947e740408b472b64706b5e79 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6a7114881ab4cdebaa41d924e900b26 : RETRY_LIMIT in 0s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d913cac3f748babfb0ec02f969e826 : RETRY_LIMIT in 0s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0841b4bd61d34f469830eb379f631a4f : SUCCESS in 7m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7bc35680717c48a68a3158d3cc64aa43 : SUCCESS in 11m 03s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/b19fdb96622c4345a80dd66913c58308 : SUCCESS in 1h 32m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81986c4e994041159875206fef46a31d : RETRY_LIMIT in 0s","accounts_in_message":[],"_revision_number":1},{"id":"af2d3e9ce6de7c47323a1d5a9958b044863160d4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-29 00:35:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"081c984c59fc1a268c3fa638b7d42201bc7bd9eb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-29 00:35:30.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"20d2c92c60e20603a4d1bf7b2980f3e917e8e66f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 01:40:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/fc0a4d84a03a48ca90c5ba2680af6279 : SUCCESS in 49m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a030f6b7bca84ff8ac4f9fcb1b14eacb : SUCCESS in 1h 00m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f94de50d03444ffb7780bc39f6f6cf6 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5364826c35e4c0ea34cef9b2cd6b502 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a70698166fd14b9cab607a666c8a627c : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1753b120d6b540d2a6210628aa09fde7 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/215310dda7934f3ba2598d58792ebad2 : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa81148464614e26960b560f2de94ff2 : SUCCESS in 7m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/64ee14934d404cdfb3e45513ee3e9971 : SUCCESS in 12m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3332b488e50f45d089bd0f7b47070598 : SUCCESS in 1h 00m 45s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/42b280e58bcc45cda4499224a32e6925 : SUCCESS in 1h 03m 18s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/4513d07dff2f489a97f97ff3d30a5f5d : SUCCESS in 1h 01m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8ed6c4c79d2462a8881f35a9c90b0da : SUCCESS in 49m 38s","accounts_in_message":[],"_revision_number":1},{"id":"19c49095254e3ea85d05af367a5d6d51c657cf47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 01:40:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f23384fa4f1e144507e5c96b5d9690c053b94e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 03:02:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4d5d25e9ad7749cdaca40d296930cd0c : SUCCESS in 1h 08m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6568d20afec466d9855288a8efd4048 : SUCCESS in 1h 08m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3749a33426d4152ad59addf759adec0 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/905c7fdaa0294f95999aae635ee2c2cc : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73eacf4e1dd1498fb245b9d9ead302e7 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d623ceee15a4213a65e2d19856ce7fd : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/408a2e8c51dc4ae2b47ec34269c86c02 : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ee134aa837424d52b14e6a2088eb4db2 : SUCCESS in 7m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59bc1446524c4c45bb0e76910a190264 : SUCCESS in 8m 10s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/30e0a7fefff446d58e507de1b30d9ef7 : SUCCESS in 1h 21m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/451dcc4d657c4050be977e118a5422da : SUCCESS in 54m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d5540ed5b5f3e634ddfa8bf30b193d4c4992cd18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 03:02:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bc5707d10affa1ce1c7a4933143fddca6d01c8e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 03:04:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a363a5d52294af4b4718985836092bd : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24c281230e42a1592b563af0ae50518ecb8bb10e","revisions":{"24c281230e42a1592b563af0ae50518ecb8bb10e":{"kind":"REWORK","_number":1,"created":"2020-07-10 00:56:00.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/84/740384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/84/740384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/740384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/740384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/84/740384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/84/740384/1"}}},"commit":{"parents":[{"commit":"63d982588f4c4b290985ca5fe5b288c2cd3579df","subject":"Merge \"Update lower-constraints versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63d982588f4c4b290985ca5fe5b288c2cd3579df"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2020-07-10 00:55:11.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2020-07-10 00:55:11.000000000","tz":120},"subject":"Fix metadefs for compute-watchdog","message":"Fix metadefs for compute-watchdog\n\nChange-Id: Idf9be5afea5d034b9679ad451306a8f2d365aa51\nCloses-Bug: #1887099\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/24c281230e42a1592b563af0ae50518ecb8bb10e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/24c281230e42a1592b563af0ae50518ecb8bb10e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
