)]}'
{"id":"openstack%2Fglance~741086","triplet_id":"openstack%2Fglance~stable%2Ftrain~I789fa7adfb459e7861c90a51f418a635c0c22244","project":"openstack/glance","branch":"stable/train","topic":"bug/1886374-stable/ussuri-stable/train","hashtags":[],"change_id":"I789fa7adfb459e7861c90a51f418a635c0c22244","subject":"Improve lazy loading mechanism for multiple stores","status":"MERGED","created":"2020-07-15 05:19:13.000000000","updated":"2020-07-18 03:32:16.000000000","submitted":"2020-07-18 03:30:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741086-1595043034332-45274524","meta_rev_id":"7a8f8755480f5744997a4babf18aa2c0255c8d19","_number":741086,"virtual_id_number":741086,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-18 03:30:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-16 19:38:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-17 16:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-17 16:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 19:38:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-17 16:57:02.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-18 03:30:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9674437509399878360b3200b86ad2ec9ba05695","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-15 05:19:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"f3694a0b101a921a17be1189fcc6996bfa9ce48a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 07:18:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ff9120427c0445daa3ed9726e9cb74d3 : SUCCESS in 1h 57m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5f7aa97f5fdb42e8aa2dfa4e6ae8e203 : SUCCESS in 1h 04m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/ed43607b87a741809d07781ece1bcf8f : SUCCESS in 1h 09m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d18c674f4ae24668ade6296d4cfa67a7 : SUCCESS in 1h 41m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03dc99d8153b40eb8b6969dfec6ab705 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4416c1b606aa48d0a92f0c3da7e7fd4a : SUCCESS in 7m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ca7fba6fbf04d77b268407fb540a38f : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57fd7813b28642eebd96bce9daf40229 : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/571de5d52d284a32a20a4650a3cf87e7 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9cab760d6a846a080a8f36f11d61d52 : SUCCESS in 8m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/93ebe84507d84adaaa405f5d871f91cb : SUCCESS in 9m 11s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/f5c32fe4a68c40ecbc12fd88420522eb : SUCCESS in 15m 00s\n- glance-eventlet-ssl-handshake-broken-py3 https://zuul.opendev.org/t/openstack/build/f43292f687e64ad7baa880fb7508e735 : FAILURE in 8m 23s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1a35214a43ac41fabd82312bfffab266 : SUCCESS in 4m 48s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e900a119c6f544ef9c6062b5bb888d23 : SUCCESS in 1h 12m 49s (non-voting)\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/153ee79f282f41159981b3bef8a9f1ef : SUCCESS in 1h 06m 07s","accounts_in_message":[],"_revision_number":1},{"id":"42ea0c49750771991e64e5fe71fabf1c6771281f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-16 19:38:10.000000000","message":"Patch Set 1: Code-Review+2\n\nMerged in ussuri, good to backport.","accounts_in_message":[],"_revision_number":1},{"id":"00e21c65464fe78c6ae9f109c982adce31dbaa65","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-17 16:57:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a4dc8243cad1fe8558ff881bb0b36e1719d1e1eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 16:57:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6be606095e7b742cb2743cde1f1f1678ed5c4638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 22:58:55.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b6495b693abd9bc937bb2d9599dc4f00118431f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 03:30:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/fe703ffeca04411ebaefabbcf7b53661 : SUCCESS in 1h 20m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6ea374db673a4a2f90aa9057c8ac2a85 : SUCCESS in 59m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/aadd4ac708a9441ab2d7c1215f7ce66e : SUCCESS in 49m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1b4913a78684b62bec7367abc2b8869 : SUCCESS in 58m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b9f2d5d64a04594a7ef4f9cc01d8d2b : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ffd78ffb96c46109ffe7dda26f409d6 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73db3f2f81f44beb833e30c512f8a235 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de273b337e3d49008f117445120d7101 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/090691173b844f18ae7a626ff501f292 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c2af66f1144328a55324e9c9e47828 : SUCCESS in 6m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ff0bf4c0c5024ea085acdd594c02e532 : SUCCESS in 9m 57s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/efb971c0cdf44d918eebc1fc43e4fda2 : SUCCESS in 12m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4d735370657f418a91d621276a310f69 : SUCCESS in 1h 04m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1904cad7d3d29af226f36bd17697c37e5c60be3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 03:30:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ed16a1d734abe2a1ba6480cfdc5f14b5941952d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 03:32:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5102e125ee3f4d15853ebd59cbf7a52a : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"baf6b5ebf05ce7a2fb012f4a90cf34c61059d97a","revisions":{"baf6b5ebf05ce7a2fb012f4a90cf34c61059d97a":{"kind":"REWORK","_number":1,"created":"2020-07-15 05:19:13.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/86/741086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/86/741086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/741086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/741086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/741086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/86/741086/1"}}},"commit":{"parents":[{"commit":"bdba1feacb7fa088650667e69290aa6e9bfad65c","subject":"Use grenade-multinode instead of the custom legacy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdba1feacb7fa088650667e69290aa6e9bfad65c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-06 07:49:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-15 05:19:13.000000000","tz":0},"subject":"Improve lazy loading mechanism for multiple stores","message":"Improve lazy loading mechanism for multiple stores\n\nGlance has a facility lazy loading for legacy images which will be called\non get/list api calls to add store information in image\u0027s location metadata\nbased on location URL of image. Even if admin decides to change the store\nnames in glance-api.conf same will also be updated in location metadata\nfor all images related to that particular store. Current implementation of\nlegacy image performs this operation on each get/list call as location metadata\nis not getting updated in database or it doesn\u0027t handle to perform store name\ncheck in glance-api.conf.\n\nImprovements done:\n1. Save updated location metadata information in database permenantly\n2. Add logic to perform lazy loading only if store information is not present\nin location metadata or store present in location metadata is not defined in\nglance\u0027s enbaled_backends configuration option.\n\nChange-Id: I789fa7adfb459e7861c90a51f418a635c0c22244\nCloses-Bug: #1886374\n(cherry picked from commit ab0e5268a9c2614572659d763b3c0b6fc36dd0cf)\n(cherry picked from commit 9a104acbae328788451442f71aefbc5c20a98f3c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/baf6b5ebf05ce7a2fb012f4a90cf34c61059d97a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/baf6b5ebf05ce7a2fb012f4a90cf34c61059d97a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
