)]}'
{"id":"openstack%2Fglance~741276","triplet_id":"openstack%2Fglance~master~I7851e2b5ed21534e739327ce9562d458064eab1c","project":"openstack/glance","branch":"master","topic":"glance-import-job","hashtags":[],"change_id":"I7851e2b5ed21534e739327ce9562d458064eab1c","subject":"Make our tempest job use import, standalone, and inject_metadata","status":"MERGED","created":"2020-07-15 17:23:39.000000000","updated":"2020-07-22 04:15:16.000000000","submitted":"2020-07-22 04:12:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741276-1595391148870-bfbdc47d","meta_rev_id":"c2102f2ebcfbfa92162d8a377cb1524c3d37500d","_number":741276,"virtual_id_number":741276,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-21 22:19:17.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 04:12:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-21 22:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-21 22:19:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-21 22:23:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-07-22 04:15:16.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-21 15:04:03.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-21 17:12:27.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-21 17:12:34.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-21 22:19:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-21 22:19:40.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-07-21 22:23:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-22 04:12:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e78b834667403721a191bd6e42f76909424b8274","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-15 17:23:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"843f6dd1195240469f6631cbce03def1bea48dd2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-15 17:24:00.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"4c576d6af85816f02279dc6a86e9fbca2ae50ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 21:45:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c90db9f9cf544e6b2b2699d5702fa9c : SUCCESS in 51m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa66ae829e954062833f239336db80fe : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d663a3d6aab4423bb8365320f108520d : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3171ab69b54349de9274b6f74c664954 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1e047d52f70451e99d65a33b79c94b2 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e9591a7c5f04ee58bfceb2fcef97097 : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98c881fbc14d430dbf059db4b3a73cc9 : SUCCESS in 14m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f86d9725268c41638f47cabec4d70a8e : SUCCESS in 8m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ab0ff57223d14277bebe3dcacb2e3f93 : SUCCESS in 1h 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"72e6bdf820ef1cebb5553c9728e3b317d41e92a0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-16 08:50:25.000000000","message":"Patch Set 1:\n\nLooks like this job is still using OSC instead of glance image-create-via-import command","accounts_in_message":[],"_revision_number":1},{"id":"c76d9f7799b7329e99ada44f697f7339793b2e8b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-16 13:39:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7b10cc9a06e2f06b91533741dcbf0f1b0dc9e087","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-16 13:40:00.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"a6efa38a0a2350d650e76acc872e5d3e3f702d92","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-16 13:42:36.000000000","message":"Patch Set 1:\n\n\u003e Looks like this job is still using OSC instead of glance\n \u003e image-create-via-import command\n\nNo, it\u0027s using import:\n\n openstack --os-cloud\u003ddevstack-admin --os-region-name\u003dRegionOne image create cirros-0.5.1-x86_64-disk --public --container-format bare --disk-format qcow2 --import\n\nIn OSC, \u0027image create --import\u0027 does the import workflow. I think you can tell that this is the case because the image is in queued state, right?\n\n | status           | queued  \n\nI just pushed up more of this which adds a post-run task that makes sure the image got the properties I added. You can see in the glance log that tasks are running and it seemed to do the metadata injection:\n\n Jul 15 20:58:56.119927 ubuntu-bionic-rax-ord-0018263174 glance-api[12067]: DEBUG glance.async_.taskflow_executor [-] Task \u0027api_image_import-InjectMetadataProperties-2e29b682-dece-4f65-ae9a-e66d00545f6e\u0027 (74620958-1ae1-4841-88d7-a2736bf3afb9) transitioned into state \u0027RUNNING\u0027 from state \u0027PENDING\u0027 {{(pid\u003d12155) _task_receiver /usr/local/lib/python3.6/dist-packages/taskflow/listeners/logging.py:192}}\n Jul 15 20:58:56.121779 ubuntu-bionic-rax-ord-0018263174 glance-api[12067]: DEBUG glance.async_.taskflow_executor [-] Task \u0027api_image_import-InjectMetadataProperties-2e29b682-dece-4f65-ae9a-e66d00545f6e\u0027 (74620958-1ae1-4841-88d7-a2736bf3afb9) transitioned into state \u0027SUCCESS\u0027 from state \u0027RUNNING\u0027 with result \u0027None\u0027 {{(pid\u003d12155) _task_receiver /usr/local/lib/python3.6/dist-packages/taskflow/listeners/logging.py:181}}","accounts_in_message":[],"_revision_number":1},{"id":"b29d4364986964ece4af624da81ff320849800b0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-16 14:28:12.000000000","message":"Patch Set 2:\n\n\u003e \u003e Looks like this job is still using OSC instead of glance\n \u003e \u003e image-create-via-import command\n \u003e \n \u003e No, it\u0027s using import:\n \u003e \n \u003e openstack --os-cloud\u003ddevstack-admin --os-region-name\u003dRegionOne\n \u003e image create cirros-0.5.1-x86_64-disk --public --container-format\n \u003e bare --disk-format qcow2 --import\n \u003e \n \u003e In OSC, \u0027image create --import\u0027 does the import workflow. I think\n \u003e you can tell that this is the case because the image is in queued\n \u003e state, right?\n \u003e \n \u003e | status           | queued\n \u003e \n \u003e I just pushed up more of this which adds a post-run task that makes\n \u003e sure the image got the properties I added. You can see in the\n \u003e glance log that tasks are running and it seemed to do the metadata\n \u003e injection:\n \u003e \n \u003e Jul 15 20:58:56.119927 ubuntu-bionic-rax-ord-0018263174\n \u003e glance-api[12067]: DEBUG glance.async_.taskflow_executor [-] Task\n \u003e \u0027api_image_import-InjectMetadataProperties-2e29b682-dece-4f65-ae9a-e66d00545f6e\u0027\n \u003e (74620958-1ae1-4841-88d7-a2736bf3afb9) transitioned into state\n \u003e \u0027RUNNING\u0027 from state \u0027PENDING\u0027 {{(pid\u003d12155) _task_receiver\n \u003e /usr/local/lib/python3.6/dist-packages/taskflow/listeners/logging.py:192}}\n \u003e Jul 15 20:58:56.121779 ubuntu-bionic-rax-ord-0018263174\n \u003e glance-api[12067]: DEBUG glance.async_.taskflow_executor [-] Task\n \u003e \u0027api_image_import-InjectMetadataProperties-2e29b682-dece-4f65-ae9a-e66d00545f6e\u0027\n \u003e (74620958-1ae1-4841-88d7-a2736bf3afb9) transitioned into state\n \u003e \u0027SUCCESS\u0027 from state \u0027RUNNING\u0027 with result \u0027None\u0027 {{(pid\u003d12155)\n \u003e _task_receiver /usr/local/lib/python3.6/dist-packages/taskflow/listeners/logging.py:181}}\n\nAhh, I overlooked --import flag","accounts_in_message":[],"_revision_number":2},{"id":"ac2f2e7261fd96f3953e6e1c7cd0450ba5ced2ee","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-16 15:06:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e6756829d778b255ce68b7ce4f0e6519b8f7d2bd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-16 15:06:22.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"d8f1407b23033f2e2caffe824512b6ab954200f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 17:54:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/22aef57b9feb4ebdac0c346859f71d5f : SUCCESS in 1h 03m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8b857c3b107c4ec6aec9444aa8d6a133 : SUCCESS in 1h 08m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/171201b2f62848149a4cc1c962c0b072 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af801c4d14b425cb4de22f9f1586943 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5e08e03dec24fd3a91fab3f34328d53 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e5e747810a34740b96d1642e0d0cb72 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff21398a2c354b3999dd853654ed201e : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8db7f1b9b06241fc85bfac0f2c7ae8a9 : SUCCESS in 8m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dea5ad53a34c44a0a6ac8f3f6fe71f69 : SUCCESS in 9m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d8c247b3ec0a49e68ffe359b062d86a6 : SUCCESS in 6m 08s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0474409127b34c6b812cc18e02b24fd9 : SUCCESS in 1h 12m 50s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a3053572e3374b35bb7b8b5986402d10 : POST_FAILURE in 1h 14m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8647aeee44b64531ade3e5b443f568ef : SUCCESS in 55m 23s","accounts_in_message":[],"_revision_number":3},{"id":"ed1fa67dd13edd37cdf3b73c98b5a099cd538f65","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-17 01:12:30.000000000","message":"Patch Set 3: Workflow-1\n\nThis is failing because the injection isn\u0027t happening for some reason. So the post check is good and I must just be missing something from the config.","accounts_in_message":[],"_revision_number":3},{"id":"39019a9f30552cf7a93c153bc84eb1ed3f378d37","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-17 15:55:21.000000000","message":"Patch Set 3:\n\n\u003e This is failing because the injection isn\u0027t happening for some\n \u003e reason. So the post check is good and I must just be missing\n \u003e something from the config.\n\nI\u0027m thinking because of:\n\n [inject_metadata_properties]\n ignore_user_roles \u003d admin\n\nwhich is default.","accounts_in_message":[],"_revision_number":3},{"id":"f4eb76e7c29c37a2d4c122727c05c5ffc4d6a04a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-17 15:59:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c10546474d52b365199d67d06d30c90817a6254e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-17 15:59:22.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"58a55545e14fca57bd24e77f1b55b6a6fa3618fe","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-17 16:16:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b0d8da796787da15f96b087f615267841f2efd4d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-17 16:16:50.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"6ceaf04301c2a7e102a988fbe8c33b2b5b227dcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 00:31:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d221a0c032f54815830763ed378b7644 : SUCCESS in 1h 10m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c070f26da0e143e18f45818692bb0d78 : SUCCESS in 1h 25m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/958743690c90425ba64ff302e1a35489 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d4176211ae0471d8586175a9a3585ff : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a85c381300b4977a5afb05b18453fd2 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d96852026ec4c84bbdac9dbd902a922 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c394bbc04885452eaa29ca33fc8d3110 : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2df4e0fa9ee34d1d906629cce6982e34 : SUCCESS in 8m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c1843077924449d490bee91769e90e38 : SUCCESS in 9m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/329bd40b081148798b7b444188ee8a60 : SUCCESS in 7m 26s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/46ad4aa09a894cf9aea7d1811942eb50 : SUCCESS in 1h 07m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8396232e354b47d1a71ba5ddd818b797 : SUCCESS in 1h 13m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0110df0d5f34cd08ff3021dd16667a3 : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":5},{"id":"4a8ccfdf015df0aaf7b61befaa5d292077992457","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-20 20:36:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6224ef1c1b7a616680fee3b23db1c418b29894d2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-20 20:36:54.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"c7cd0e67c78b8fa07adec5c017823aeb7f9704e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 22:37:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a61fb5b438f4490bb235420f2612e550 : SUCCESS in 1h 00m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49642372fcee425b9b9cd3788f3b6b4f : SUCCESS in 1h 19m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ffd44ff47ed496d8ac01bb4ed4dc163 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4327d16cd359471eb8d2b79edd0ae324 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff7b408e1b7a4ee98c487e671f8a78f7 : SUCCESS in 8m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ce9c437e65d4973ae0208fae7c93081 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a404e098beea435bb3f71d13564e84aa : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2c169645c504890a52e7baec90ab628 : FAILURE in 28m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6525d279349044e0bf92e9c67d10182f : TIMED_OUT in 30m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5223fd3200ce41c0bdb959d67d633219 : SUCCESS in 4m 17s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d582e28b749b4c399766948f25ce350f : SUCCESS in 59m 28s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0430c0837540475f8c21183bcb3635d9 : SUCCESS in 1h 12m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b00cb5e63004754b7aabe1a89cbedf5 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":6},{"id":"bc6c4c0e90bd810fa8f3a78454374ad904329328","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-21 06:22:48.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"79b9b394f613ec60e45fc401319c59d4fe71101d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-21 06:23:11.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"e6466fc86ae36cd50099a47aae51b8a280ad493f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 08:53:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8dbf9f00972a483697408c418bd428b4 : SUCCESS in 1h 22m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/74e1903d44a04b10bf22ff851b7c3634 : SUCCESS in 1h 18m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8be9889bfde147ed97441832dd76ea3f : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3c461c1e156490181badcd37d8e7567 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ed32f1ef6db41f3b9eb7b98243c719f : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9349e2ad4ecf40d69c45f30c16d405c1 : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986ddc23bdb34edfb56b2e22eb563cbb : SUCCESS in 8m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/28869b9314ab44a997330a17ec61d473 : SUCCESS in 9m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/65e6627eb3094ede93f1f79d27105c58 : SUCCESS in 10m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9a69ea2a9c544d5cbfd72400366614e2 : SUCCESS in 6m 58s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c0ca323883cc4793a2945af52b0b4ee5 : SUCCESS in 57m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/351bf0744c2444f8a71ae7468e20ab07 : SUCCESS in 1h 27m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/253d8a19ff404563a690510c35552855 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":6},{"id":"0fde3ed9a0ebea40518479428f18224aebc17458","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-21 09:04:22.000000000","message":"Patch Set 6: Code-Review+2\n\nLooks good to me, thanks for working on it.\nCan we think on more meaningful name?","accounts_in_message":[],"_revision_number":6},{"id":"ee65a1de4f34e0529da0d3b06c362ba6e0ecf836","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-21 14:18:14.000000000","message":"Patch Set 6:\n\n\u003e Looks good to me, thanks for working on it.\n \u003e Can we think on more meaningful name?\n\nSure, what do you want to call it? I come up with boring names :)","accounts_in_message":[],"_revision_number":6},{"id":"fc27962f5d88cd77393fbfd597e7326015f2ac3b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-21 14:49:19.000000000","message":"Patch Set 6:\n\n\u003e \u003e Looks good to me, thanks for working on it.\n \u003e \u003e Can we think on more meaningful name?\n \u003e \n \u003e Sure, what do you want to call it? I come up with boring names :)\n\nI am also not either good in naming :D\nI think tempest-integrated-storage-import-workflow\nor\ntempest-integrated-image-import","accounts_in_message":[],"_revision_number":6},{"id":"655181423960bca1b5cc1cf0734f0cb82c7f9af0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-21 14:56:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4314e362c07f3fe9e90f694b3fc0b8e2e1866bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 14:57:10.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  Job tempest-integrated-storage-import not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate-storage\n        - openstack-lower-constraints-jobs\n        - openstack-python3-victoria-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional-py36\n          - openstack-tox-functional-py38\n          - glance-code-constants-check\n          - devstack-plugin-ceph-tempest-py3:\n              voting: false\n              irrelevant-files: \u0026tempest-irrelevant-files\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^glance/hacking/.*$\n                - ^glance/locale/.*$\n                - ^glance/tests/.*$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^tools/.*$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - tempest-integrated-storage-import:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-ipv6-only:\n              irrelevant-files: *tempest-irrelevant-files\n  \n      gate:\n        jobs:\n          - openstack-tox-functional-py36\n          - openstack-tox-functional-py38\n          - tempest-integrated-storage-import:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-ipv6-only:\n              irrelevant-files: *tempest-irrelevant-files\n      experimental:\n        jobs:\n          - glance-tox-py38-glance_store-tips\n          - glance-tox-py36-glance_store-tips\n          - glance-tox-functional-py38-glance_store-tips\n          - glance-tox-functional-py36-glance_store-tips\n          - barbican-simple-crypto-devstack-tempest\n          - grenade-multinode\n          - tempest-pg-full:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-full-py3-opensuse15:\n              irrelevant-files: *tempest-irrelevant-files\n      periodic:\n        jobs:\n          # NOTE(rosmaita): we only want the \"tips\" jobs to be run against\n          # master, hence the \u0027branches\u0027 qualifiers below.  Without them, when\n          # a stable branch is cut, the tests would be run against the stable\n          # branch as well, which is pointless because these libraries are\n          # frozen (more or less) in the stable branches.\n          #\n          # The \"tips\" jobs can be removed from the stable branch .zuul.yaml\n          # files if someone is so inclined, but that would require manual\n          # maintenance, so we do not do it by default.  Another option is\n          # to define these jobs in the openstack/project-config repo.\n          # That would make us less agile in adjusting these tests, so we\n          # aren\u0027t doing that either.\n          - glance-tox-functional-py38-oslo-tips:\n              branches: master\n          - glance-tox-functional-py36-oslo-tips:\n              branches: master\n          - glance-tox-py38-keystone-tips:\n              branches: master\n          - glance-tox-py36-keystone-tips:\n              branches: master\n          - glance-tox-functional-py38-keystone-tips:\n              branches: master\n          - glance-tox-functional-py36-keystone-tips:\n              branches: master\n          - glance-tox-py38-glance_store-tips:\n              branches: master\n          - glance-tox-py36-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py38-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py36-glance_store-tips:\n              branches: master\n          - glance-tox-py38-cursive-tips:\n              branches: master\n          - glance-tox-py36-cursive-tips:\n              branches: master\n          - glance-tox-functional-py38-cursive-tips:\n              branches: master\n          - glance-tox-functional-py36-cursive-tips:\n              branches: master\n\n  in \"openstack/glance/.zuul.yaml@master\", line 231, column 3","accounts_in_message":[],"_revision_number":7},{"id":"ea88dd2f819fec5788c02992d1b91ed6269b0cf6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-21 14:57:14.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"e1f68ae34ed2bcedebf7710c1f2959577f55cdea","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-21 14:59:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f9d405fde2c4623384d4423e9d2b388192ba2925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 15:00:11.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  Job tempest-integrated-storage-import not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate-storage\n        - openstack-lower-constraints-jobs\n        - openstack-python3-victoria-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional-py36\n          - openstack-tox-functional-py38\n          - glance-code-constants-check\n          - devstack-plugin-ceph-tempest-py3:\n              voting: false\n              irrelevant-files: \u0026tempest-irrelevant-files\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^glance/hacking/.*$\n                - ^glance/locale/.*$\n                - ^glance/tests/.*$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^tools/.*$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - tempest-integrated-storage-import-workflow:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-ipv6-only:\n              irrelevant-files: *tempest-irrelevant-files\n  \n      gate:\n        jobs:\n          - openstack-tox-functional-py36\n          - openstack-tox-functional-py38\n          - tempest-integrated-storage-import:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-ipv6-only:\n              irrelevant-files: *tempest-irrelevant-files\n      experimental:\n        jobs:\n          - glance-tox-py38-glance_store-tips\n          - glance-tox-py36-glance_store-tips\n          - glance-tox-functional-py38-glance_store-tips\n          - glance-tox-functional-py36-glance_store-tips\n          - barbican-simple-crypto-devstack-tempest\n          - grenade-multinode\n          - tempest-pg-full:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-full-py3-opensuse15:\n              irrelevant-files: *tempest-irrelevant-files\n      periodic:\n        jobs:\n          # NOTE(rosmaita): we only want the \"tips\" jobs to be run against\n          # master, hence the \u0027branches\u0027 qualifiers below.  Without them, when\n          # a stable branch is cut, the tests would be run against the stable\n          # branch as well, which is pointless because these libraries are\n          # frozen (more or less) in the stable branches.\n          #\n          # The \"tips\" jobs can be removed from the stable branch .zuul.yaml\n          # files if someone is so inclined, but that would require manual\n          # maintenance, so we do not do it by default.  Another option is\n          # to define these jobs in the openstack/project-config repo.\n          # That would make us less agile in adjusting these tests, so we\n          # aren\u0027t doing that either.\n          - glance-tox-functional-py38-oslo-tips:\n              branches: master\n          - glance-tox-functional-py36-oslo-tips:\n              branches: master\n          - glance-tox-py38-keystone-tips:\n              branches: master\n          - glance-tox-py36-keystone-tips:\n              branches: master\n          - glance-tox-functional-py38-keystone-tips:\n              branches: master\n          - glance-tox-functional-py36-keystone-tips:\n              branches: master\n          - glance-tox-py38-glance_store-tips:\n              branches: master\n          - glance-tox-py36-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py38-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py36-glance_store-tips:\n              branches: master\n          - glance-tox-py38-cursive-tips:\n              branches: master\n          - glance-tox-py36-cursive-tips:\n              branches: master\n          - glance-tox-functional-py38-cursive-tips:\n              branches: master\n          - glance-tox-functional-py36-cursive-tips:\n              branches: master\n\n  in \"openstack/glance/.zuul.yaml@master\", line 231, column 3","accounts_in_message":[],"_revision_number":8},{"id":"abd86d97597e639e1f4baf1d667ca38db28c4397","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-21 15:00:21.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":8},{"id":"eecd4a5d2219480250f2cac4e585266810f99261","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-21 15:01:25.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ff04c7c18286480d9d3367bc4cb5d0a7842f666e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-21 15:01:47.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":9},{"id":"252701b3f274944c98b8c449ffcb449d9fec4ac5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-21 15:04:03.000000000","message":"Patch Set 9: Code-Review+2 Review-Priority+1\n\nThank you for your efforts!\nFirst tempest job for import workflow \\o/","accounts_in_message":[],"_revision_number":9},{"id":"262ab0fec38320bc39c2aafd6791333bd9ee55b1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-21 17:13:01.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"febc63c390a0f75e90e3e2ffe36a151bcd594126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 21:40:14.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/21954798823b47f7aff866a5fb14f00d : SUCCESS in 1h 03m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/855850ac5e9a44f5b3ed2b76c868eb22 : SUCCESS in 1h 10m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33c9ae0cb89b4a76bec143ba513a7e88 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f31f3310b0346e4a2bdaa067615c4ad : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/448d71f56f4145b68ccd0f72426f5155 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2bd0e1454ec44aaa49ff70869f900be : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89011afd221b473daa0882f7c14cb7bb : SUCCESS in 9m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f44c789a80c345fa9d6d34f00bf13810 : SUCCESS in 10m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ab9a9506272a4f13a7298e0b7d9919f9 : SUCCESS in 11m 29s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2579c54999f7481dadca04784fad378d : SUCCESS in 6m 46s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d8c22833a03649569e950188b190a47c : SUCCESS in 1h 58m 15s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/2ae9ec4d938f40838a04e9afedebc2a3 : FAILURE in 1h 48m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e06d81d52d734e32aff77f71d15e8dfd : SUCCESS in 1h 32m 02s","accounts_in_message":[],"_revision_number":9},{"id":"4612bdf99317f2d8b47272b53c021383a77ef361","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-21 22:01:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4d1b32219b8758c793d04baa58311f83c4f37de4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-21 22:18:56.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2035c6764d159c623c1bb5998691cfcfc3db424e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-21 22:19:17.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":10},{"id":"9c0dae407845284606ced7f04e269004775925a7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-21 22:19:40.000000000","message":"Patch Set 10: Code-Review+1\n\nthanks","accounts_in_message":[],"_revision_number":10},{"id":"05e41407a403d0746761719fed623aa3f9f09d5d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-21 22:23:39.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nMinor tweak since last approved rev, so I am going to restore workflow on this to keep things moving.","accounts_in_message":[],"_revision_number":10},{"id":"7b8d082d01936ab8db20bb010dc5c0372a78a89f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 02:49:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d92d5cda66a743e6a8294d7ba919c2c8 : SUCCESS in 1h 07m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8eb23918e0614a8fb1e6e090fa4bd476 : SUCCESS in 1h 12m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ad3591e2fc143d7ba85a9d3550f4f5f : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14f00fd0b39492babf1df95654a9d45 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c059f6fc6c240eb8978dad589984678 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6345f364f8eb4f1bb902a3e036df9f2b : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6f6f04ba8948a09803d30aa6baa8ee : SUCCESS in 5m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3610718118c94ff48a61ac4d00d08487 : SUCCESS in 7m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0f24d3dba9a4e62a9812200036a136d : SUCCESS in 8m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/823bd1bde4c7485687c507284914b2fe : SUCCESS in 4m 17s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a0be9c41df1f4c3baf116993644e69b4 : SUCCESS in 1h 06m 23s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d0a886e7b92049eca5f2b962611fb56d : SUCCESS in 1h 31m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dc5d962afda4bb7b3c085f420acce10 : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":10},{"id":"0a1be16cd3f54573933774709d678c43795870b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 02:49:47.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"534afbb92e01b7519bbdf49dcd77323202fefd43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:12:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"6f36714a66f91501a64d7f959fe6dbb2e30fb42a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:12:28.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/721286f7ed50482e80b10a3cd59a564d : SUCCESS in 1h 02m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/196acf8b9624465988b8437cb30da790 : SUCCESS in 1h 09m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d5ec9cda6ab43978e7335a41d1e1e9d : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81c70624a3604511ad053d900bd35857 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94f38fce02954d7788298ca2b9901e8d : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92049e57e8514a1bb15901c8981796d0 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a25aedc041f4e3eb3dc802e5481a095 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3e5656cb3ce34b1f99c7c0fc8ddaa1e4 : SUCCESS in 7m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/60dbd1fd599441d1ba09146a539a9155 : SUCCESS in 8m 32s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/37ad31bdd96d41058efa562fceff0665 : SUCCESS in 1h 10m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b13adc2aad74efc8d06681ed6f16931 : SUCCESS in 1h 09m 54s","accounts_in_message":[],"_revision_number":10},{"id":"8d82dd109b27400df76a7e3635f8ccd5fb9a3196","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:15:16.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f32d90e121854dab957bf99c71e96f65 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"7d49b7ffc22d8835610c7670acbac0b910afcc63","revisions":{"8fd1c0391ac90dbe5102fca62c825c16049f1d9e":{"kind":"REWORK","_number":1,"created":"2020-07-15 17:23:39.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/76/741276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/741276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/741276/1"}}},"commit":{"parents":[{"commit":"9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6","subject":"Merge \"Improve lazy loading mechanism for multiple stores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 17:22:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 17:22:55.000000000","tz":-420},"subject":"WIP: Make our tempest job use import, standalone, and inject_metadata","message":"WIP: Make our tempest job use import, standalone, and inject_metadata\n\nThis should configure Glance to inject a random piece of metadata\ninto images that are created via import, which we can check later to\nmake sure it works.\n\nChange-Id: I7851e2b5ed21534e739327ce9562d458064eab1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fd1c0391ac90dbe5102fca62c825c16049f1d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fd1c0391ac90dbe5102fca62c825c16049f1d9e"}]},"branch":"refs/heads/master"},"50c0438178df9fb56ce6d6c29f7ca3f30b90fdee":{"kind":"REWORK","_number":2,"created":"2020-07-16 13:39:39.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/76/741276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/741276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/741276/2"}}},"commit":{"parents":[{"commit":"9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6","subject":"Merge \"Improve lazy loading mechanism for multiple stores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 17:22:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-16 13:39:49.000000000","tz":-420},"subject":"WIP: Make our tempest job use import, standalone, and inject_metadata","message":"WIP: Make our tempest job use import, standalone, and inject_metadata\n\nThis should configure Glance to inject a random piece of metadata\ninto images that are created via import, and execute a post-run job\nto make sure the cirros image that was uploaded via import got that\nmetadata as expected.\n\nChange-Id: I7851e2b5ed21534e739327ce9562d458064eab1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/50c0438178df9fb56ce6d6c29f7ca3f30b90fdee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/50c0438178df9fb56ce6d6c29f7ca3f30b90fdee"}]},"branch":"refs/heads/master"},"2554398776037488e15f310755ae3805cc70644f":{"kind":"REWORK","_number":3,"created":"2020-07-16 15:06:01.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/76/741276/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/741276/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/741276/3"}}},"commit":{"parents":[{"commit":"9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6","subject":"Merge \"Improve lazy loading mechanism for multiple stores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 17:22:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-16 15:06:11.000000000","tz":-420},"subject":"WIP: Make our tempest job use import, standalone, and inject_metadata","message":"WIP: Make our tempest job use import, standalone, and inject_metadata\n\nThis should configure Glance to inject a random piece of metadata\ninto images that are created via import, and execute a post-run job\nto make sure the cirros image that was uploaded via import got that\nmetadata as expected.\n\nChange-Id: I7851e2b5ed21534e739327ce9562d458064eab1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2554398776037488e15f310755ae3805cc70644f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2554398776037488e15f310755ae3805cc70644f"}]},"branch":"refs/heads/master"},"252bb5a36ab23231393cf84c5e57920ad2ec4622":{"kind":"REWORK","_number":4,"created":"2020-07-17 15:59:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/76/741276/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/741276/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/741276/4"}}},"commit":{"parents":[{"commit":"9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6","subject":"Merge \"Improve lazy loading mechanism for multiple stores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 17:22:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-17 15:59:12.000000000","tz":-420},"subject":"WIP: Make our tempest job use import, standalone, and inject_metadata","message":"WIP: Make our tempest job use import, standalone, and inject_metadata\n\nThis should configure Glance to inject a random piece of metadata\ninto images that are created via import, and execute a post-run job\nto make sure the cirros image that was uploaded via import got that\nmetadata as expected.\n\nChange-Id: I7851e2b5ed21534e739327ce9562d458064eab1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/252bb5a36ab23231393cf84c5e57920ad2ec4622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/252bb5a36ab23231393cf84c5e57920ad2ec4622"}]},"branch":"refs/heads/master"},"27467dc6017eca93369ba97c0b7762008c1cd4ef":{"kind":"REWORK","_number":5,"created":"2020-07-17 16:16:29.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/76/741276/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/741276/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/741276/5"}}},"commit":{"parents":[{"commit":"9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6","subject":"Merge \"Improve lazy loading mechanism for multiple stores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 17:22:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-17 16:16:41.000000000","tz":-420},"subject":"WIP: Make our tempest job use import, standalone, and inject_metadata","message":"WIP: Make our tempest job use import, standalone, and inject_metadata\n\nThis should configure Glance to inject a random piece of metadata\ninto images that are created via import, and execute a post-run job\nto make sure the cirros image that was uploaded via import got that\nmetadata as expected.\n\nChange-Id: I7851e2b5ed21534e739327ce9562d458064eab1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/27467dc6017eca93369ba97c0b7762008c1cd4ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/27467dc6017eca93369ba97c0b7762008c1cd4ef"}]},"branch":"refs/heads/master"},"3692571605cd0239e1565a8b8cf35d479bf3c74f":{"kind":"REWORK","_number":6,"created":"2020-07-20 20:36:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/76/741276/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/741276/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/741276/6"}}},"commit":{"parents":[{"commit":"9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6","subject":"Merge \"Improve lazy loading mechanism for multiple stores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 17:22:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-20 20:36:07.000000000","tz":-420},"subject":"Make our tempest job use import, standalone, and inject_metadata","message":"Make our tempest job use import, standalone, and inject_metadata\n\nThis configures Glance to inject a random piece of metadata\ninto images that are created via import, cause devstack to use import,\nand execute a post-run job to make sure the cirros image that was\nuploaded via import got that metadata as expected.\n\nChange-Id: I7851e2b5ed21534e739327ce9562d458064eab1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3692571605cd0239e1565a8b8cf35d479bf3c74f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3692571605cd0239e1565a8b8cf35d479bf3c74f"}]},"branch":"refs/heads/master"},"cc46eb8d9b05ee4f5a1652e0202466b66267d1e0":{"kind":"REWORK","_number":7,"created":"2020-07-21 14:56:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/76/741276/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/741276/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/741276/7"}}},"commit":{"parents":[{"commit":"9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6","subject":"Merge \"Improve lazy loading mechanism for multiple stores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 17:22:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 14:57:07.000000000","tz":-420},"subject":"Make our tempest job use import, standalone, and inject_metadata","message":"Make our tempest job use import, standalone, and inject_metadata\n\nThis configures Glance to inject a random piece of metadata\ninto images that are created via import, cause devstack to use import,\nand execute a post-run job to make sure the cirros image that was\nuploaded via import got that metadata as expected.\n\nChange-Id: I7851e2b5ed21534e739327ce9562d458064eab1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc46eb8d9b05ee4f5a1652e0202466b66267d1e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc46eb8d9b05ee4f5a1652e0202466b66267d1e0"}]},"branch":"refs/heads/master"},"da836bdd2dcb50ce70458b09b86656f46d13810d":{"kind":"REWORK","_number":8,"created":"2020-07-21 14:59:56.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/76/741276/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/741276/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/741276/8"}}},"commit":{"parents":[{"commit":"9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6","subject":"Merge \"Improve lazy loading mechanism for multiple stores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 17:22:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 15:00:09.000000000","tz":-420},"subject":"Make our tempest job use import, standalone, and inject_metadata","message":"Make our tempest job use import, standalone, and inject_metadata\n\nThis configures Glance to inject a random piece of metadata\ninto images that are created via import, cause devstack to use import,\nand execute a post-run job to make sure the cirros image that was\nuploaded via import got that metadata as expected.\n\nChange-Id: I7851e2b5ed21534e739327ce9562d458064eab1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da836bdd2dcb50ce70458b09b86656f46d13810d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da836bdd2dcb50ce70458b09b86656f46d13810d"}]},"branch":"refs/heads/master"},"b6427cbc1503780744837b1f87b968182befd939":{"kind":"REWORK","_number":9,"created":"2020-07-21 15:01:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/76/741276/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/741276/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/741276/9"}}},"commit":{"parents":[{"commit":"9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6","subject":"Merge \"Improve lazy loading mechanism for multiple stores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 17:22:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 15:01:38.000000000","tz":-420},"subject":"Make our tempest job use import, standalone, and inject_metadata","message":"Make our tempest job use import, standalone, and inject_metadata\n\nThis configures Glance to inject a random piece of metadata\ninto images that are created via import, cause devstack to use import,\nand execute a post-run job to make sure the cirros image that was\nuploaded via import got that metadata as expected.\n\nChange-Id: I7851e2b5ed21534e739327ce9562d458064eab1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6427cbc1503780744837b1f87b968182befd939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6427cbc1503780744837b1f87b968182befd939"}]},"branch":"refs/heads/master"},"7d49b7ffc22d8835610c7670acbac0b910afcc63":{"kind":"REWORK","_number":10,"created":"2020-07-21 22:18:56.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/76/741276/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/741276/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/741276/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/741276/10"}}},"commit":{"parents":[{"commit":"9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6","subject":"Merge \"Improve lazy loading mechanism for multiple stores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c66f46a148407c10f5bdfb0fcad9a9dc0fa63d6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 17:22:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 22:18:48.000000000","tz":-420},"subject":"Make our tempest job use import, standalone, and inject_metadata","message":"Make our tempest job use import, standalone, and inject_metadata\n\nThis configures Glance to inject a random piece of metadata\ninto images that are created via import, cause devstack to use import,\nand execute a post-run job to make sure the cirros image that was\nuploaded via import got that metadata as expected.\n\nChange-Id: I7851e2b5ed21534e739327ce9562d458064eab1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d49b7ffc22d8835610c7670acbac0b910afcc63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d49b7ffc22d8835610c7670acbac0b910afcc63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
