)]}'
{"id":"openstack%2Fglance~741312","triplet_id":"openstack%2Fglance~master~Icf3aded7975d99dbc01b9378905f5e9dece07f04","project":"openstack/glance","branch":"master","topic":"bp/copy-unowned-image","hashtags":[],"change_id":"Icf3aded7975d99dbc01b9378905f5e9dece07f04","subject":"Remove unused \"copy_from\" policy rule","status":"MERGED","created":"2020-07-15 20:02:53.000000000","updated":"2020-07-18 03:32:07.000000000","submitted":"2020-07-18 03:30:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741312-1595043047130-dfe663bb","meta_rev_id":"2141f9014cc75ce5ed5d45e7597cb6c6e15e9f4a","_number":741312,"virtual_id_number":741312,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-17 15:52:55.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2020-07-17 15:50:20.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-18 03:30:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-15 20:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-16 02:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-16 02:00:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 20:09:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-16 02:00:24.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-17 15:52:55.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-18 03:30:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb5d31ca25ca1029a39e0270f0e492d24aca8334","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-15 20:02:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4ec6c3a5d64e7d7b6df9a99d6311dae4c0afd8e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 20:09:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d3b00d7cd50c88a9b5a7306cf06c0383cee4e9ec","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-15 21:38:50.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"a24de6be664be04ae94a112c93a625abad7ff264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 22:30:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bd6a2950dccc4020a70a9f454cc71002 : SUCCESS in 1h 01m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42d6047de8744872ad617708e66555d8 : SUCCESS in 1h 03m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff6d3161ac6c483683692b5a3e90a7dd : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4eac42107c241298943a901917a4823 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cbd8091700b445bba78d0434577595a : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6d20be4a4904a75b0a64db6c482202d : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3c9b07b2504e7eba9e8abddfebdffd : SUCCESS in 8m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8964460d9066425689c55412e4fbf853 : SUCCESS in 6m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4e6ec9f7415f49258feaa9c1de69fd4d : SUCCESS in 14m 10s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/22721a82984242af965bf9ed9b152a5a : SUCCESS in 7m 17s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/cd748678a53e4aa894491ceeb5da4be4 : SUCCESS in 59m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9245e5b0fc344e6ba056a10bbf9aebfd : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":1},{"id":"69973e2b4cbde98bb4059daa4cc540a29fdecb10","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-16 02:00:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"098287527678b9453cfe70539fc1dec082c732bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 12:56:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fd2e9330b9f1cdfbd99e1037dff0763fd656c7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 16:12:22.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"4a575c2255cdb3b4bf6a6824a2f1e0bacf81392f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-17 15:50:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"11e473c2f101b30728cffc7a6f8ed79a9709d01c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-17 15:52:55.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"fb3ec261bbf543d5378f2260a64ee6950319eca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 21:34:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ffd965b185254328928d974beb954758 : SUCCESS in 1h 04m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc2808ce46c14c048e1f03b89821826b : SUCCESS in 1h 08m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b2545ff206b40b69e89b473548f1245 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/152706d85e7d45549a2729f9006fba74 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6736575b5a8d4bc3a97e82b566f32858 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9465170089f4485ab629578985ba7010 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f011bb7a44dd4747946ede56ff8c4664 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/57db2c2d409b4ccca71b4147c881dc02 : SUCCESS in 13m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4ffea3b1fbd048d7afe1ccd21fff397d : SUCCESS in 10m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/dddc9e1d91c342f3b13e4a683616ec7d : SUCCESS in 6m 22s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d7fb570c98404597b979cf331fde4743 : SUCCESS in 1h 24m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2d92c6529ee4eba817b22e204fb5db0 : SUCCESS in 49m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3163a3c04afddfc9a8ab0741a3ad824d37849169","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 21:35:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37288821e72b464943fa238080c9e4c0d9cffb1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 22:59:32.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86a05a35a44c43b2964baaf1cd6b2c633b3b3cbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 03:30:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ea3f38bbf8ab4b4688bd5555f36750fa : SUCCESS in 1h 07m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/348238f43c154ce9a3f9d763ea60eca1 : SUCCESS in 1h 12m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fd5e1ea0ba5458f979b628f617f0b33 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e66bf51b90044504b915801186549d97 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/025e29d6cf324c2f8d6c7d289a5fb303 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b959156bf8424783a8777800b226f61d : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86fd4ab25150439688487db2524359d2 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d81253c230ae4119ba3e8b3d9a85e205 : SUCCESS in 7m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4d064214f9b5449dbfab4ecd2b071b9e : SUCCESS in 8m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec0df39a3af2480fa4de645078cd117b : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5b7c7250a899061cb38d91781a75dabd07c81f7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 03:30:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4ddf5c5dfd0f4f39ab40796ee8c4fa0258fe2ea6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 03:32:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27df602d033e412b873a5106aa82ff7c : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"363a78ddfb15ac07ad7d712dddbc7e7cda734973","revisions":{"363a78ddfb15ac07ad7d712dddbc7e7cda734973":{"kind":"REWORK","_number":1,"created":"2020-07-15 20:02:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/741312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/741312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/741312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/741312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/741312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/741312/1"}}},"commit":{"parents":[{"commit":"ee8a69d50675097ec58dcb61c84966fe6255eb80","subject":"Add a policy knob for allowing non-owned image copying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee8a69d50675097ec58dcb61c84966fe6255eb80"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 20:01:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-15 20:01:58.000000000","tz":-420},"subject":"Remove unused \"copy_from\" policy rule","message":"Remove unused \"copy_from\" policy rule\n\nThis rule was only used in v1, which is now gone. Remove this to avoid\nconfusion over the difference between it and \"copy_image\" (and of course,\nconfusion over it doing anything at all).\n\nLoosely related to blueprint copy-unowned-image\n\nChange-Id: Icf3aded7975d99dbc01b9378905f5e9dece07f04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/363a78ddfb15ac07ad7d712dddbc7e7cda734973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/363a78ddfb15ac07ad7d712dddbc7e7cda734973"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
