)]}'
{"id":"openstack%2Fglance~741992","triplet_id":"openstack%2Fglance~stable%2Fussuri~I93dc7c999ff7a180e6b3ff760fa65328b9a883f3","project":"openstack/glance","branch":"stable/ussuri","topic":"fix_adm_doc","hashtags":[],"change_id":"I93dc7c999ff7a180e6b3ff760fa65328b9a883f3","subject":"Fix admin docs deploying under HTTPD","status":"MERGED","created":"2020-07-20 17:08:54.000000000","updated":"2020-07-20 21:14:15.000000000","submitted":"2020-07-20 21:12:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741992-1595279567838-8f1d0fce","meta_rev_id":"b38ab462e547230acd26eec88f26fe86668be6ff","_number":741992,"virtual_id_number":741992,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-20 21:12:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2020-07-20 18:41:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2020-07-20 18:41:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 18:41:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-20 21:12:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cbc8867433c5ab2717665917b37ec2e4cdb3475c","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-20 17:08:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dd176cd5f2caa919e9a833c82791267a1cabc47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 17:56:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d71ec6f780ee405bbbb25f89d95a680e : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf43a413c14248aa8e941b72d1ed164f : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85e053b6ba824cfa9c2b0cab0c3cc715 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"77b55d07e4a7e9f536651317924ee3445861b8ec","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-20 18:41:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1c1aaf4eefb0cfd6a9df5f9ac9c5be7a5dd1fd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 18:41:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"153ecced2f69517ec249aee704436394972f240d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 21:12:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a687436eccd4719a1fcb4bd4b16275b : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7a29d4792f74a3c8c2c01b55f115ac7 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba6f482c4e774c2ab558a9032cb68592 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f29712b24d6f544fe51c4ecd23ae6bde1ff6b4df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 21:12:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5eb4f9b337738f8ff6a4dbc8fcb31abafec9474e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 21:14:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e34f322a524a739524da102abe585d : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37772f430f93443189ab54d8582a4bbe : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a8fc940e271edfbd87369bb3920196c4dd5cf7e","revisions":{"3a8fc940e271edfbd87369bb3920196c4dd5cf7e":{"kind":"REWORK","_number":1,"created":"2020-07-20 17:08:54.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/92/741992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/741992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/741992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/741992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/741992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/741992/1"}}},"commit":{"parents":[{"commit":"9a104acbae328788451442f71aefbc5c20a98f3c","subject":"Improve lazy loading mechanism for multiple stores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a104acbae328788451442f71aefbc5c20a98f3c"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-17 17:52:50.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-20 17:08:32.000000000","tz":60},"subject":"Fix admin docs deploying under HTTPD","message":"Fix admin docs deploying under HTTPD\n\nDuring Pike cycle there was efforts to deploy services under\nApache HTTPD. It became clear very quickly that Glance did not\noperate properly when deployed as bare wsgi app under web\nserver but admin documentation was merged that indicated that\nbeing somehow the preferred method. It was added to the doc\nthat in Pike release there was issues in these models.\n\nThere was never interest nor resources to fix the underlying\nissues but the doc stayed in place indicating that those\nissues could be only Pike related when they in matter of\nfact has got even worse over the time. Due to the fact that\nGlance is even more relying on eventlet than back then it\u0027s\ntime to clarify the docs and make clear it\u0027s not adviced\ndeployment model, it won\u0027t work and there is no itention\nto change that.\n\nChange-Id: I93dc7c999ff7a180e6b3ff760fa65328b9a883f3\nCloses-Bug: #1887994\n(cherry picked from commit 785eefcf7853ccccfa9bec392e2271c3589a3927)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a8fc940e271edfbd87369bb3920196c4dd5cf7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a8fc940e271edfbd87369bb3920196c4dd5cf7e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
