)]}'
{"id":"openstack%2Fglance~742309","triplet_id":"openstack%2Fglance~master~Ie4a9171f002b42a13c1786268057bdc0ab3804d0","project":"openstack/glance","branch":"master","topic":"glance-import-job","hashtags":[],"change_id":"Ie4a9171f002b42a13c1786268057bdc0ab3804d0","subject":"Make our import-workflow job also convert images to raw","status":"MERGED","created":"2020-07-21 23:06:05.000000000","updated":"2020-09-02 21:49:58.000000000","submitted":"2020-09-02 21:43:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742309-1599082986736-941d6ed0","meta_rev_id":"b50d9068a861797ed4663e53b26da2f98fb844cd","_number":742309,"virtual_id_number":742309,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-27 14:32:13.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-02 21:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-28 14:50:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-02 14:06:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-28 14:50:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 00:22:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-27 14:32:13.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-02 14:06:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-09-02 21:43:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91546d32a0b6964a7d403f4d327af0487d82c8bd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-21 23:06:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1620209efea9809b41f0648f47716c5bb03350c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-21 23:06:32.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"df03c69c363926a6e2ff95f8c5dd5ae2907034f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:27:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f0e8dcfddbd4d9f8c25e773634ba794 : SUCCESS in 1h 09m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/24900faa4ff848ec98c27e679018aeac : SUCCESS in 1h 14m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3841247151cb456a9abc08415de24300 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd764481a6e42d9abc576bbe98e718a : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/754fae0eb88e4d8b91fb72750723f0a6 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7e48c7cdb454738906a918f6f2a4828 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4012a4ddb04d55af7d430c790a7448 : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e3a5c9d72fc54d51a8991282d9541f72 : SUCCESS in 8m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/54c26305db76408f8c9f218bc81fdcbd : SUCCESS in 8m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8ab56e1cae3645289b5a113a8964042f : SUCCESS in 5m 10s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8eddc46e9d3344ea987c49e9627beca8 : SUCCESS in 1h 05m 14s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/a15a5163ff6046a199dc2c09f5ebbb45 : SUCCESS in 1h 14m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f8536de7ad9413b8c00ec929341c148 : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":1},{"id":"535439c9ec9f12ab507753b6f76349564f348af4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-22 06:59:51.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\nLooks good to me but now IMO we should change the name of playbook from post-check-metadata-injection.yaml to post-check-import-workflow.yaml","accounts_in_message":[],"_revision_number":1},{"id":"f22230df966b087d9bfbd99607668b18dc1e5905","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-27 14:31:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f3e4d85ea4dfcfdcdb85126810aa64be2068a493","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-27 14:32:13.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"cfca9e2a7de266c0d4883de1b732f9d22ef35783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 17:59:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/71270a5c83164738aaf88d7ec179e654 : SUCCESS in 1h 16m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0aebf749f176466682b06c27304288d0 : SUCCESS in 1h 18m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9dcf9833ff8411792904f767943909f : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f2ded890b542008bd6fd7bfa2e4eff : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d35ce8668984d899cc475cb91362983 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcf8ba0b25984fd3b748f39ddf883a1c : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c655f2b34c2464b8352155742979dd0 : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6183804c1c0847e696a6f41c386451ab : SUCCESS in 8m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/879c9efd5b3c4058b561046295db219a : SUCCESS in 9m 04s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1c703620e42742e7992d8886035bcded : SUCCESS in 5m 51s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9f31430701694f60b03a4738ead48f70 : SUCCESS in 1h 05m 57s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d15448b343b146db9d03f4e882d45eeb : SUCCESS in 1h 05m 01s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/194a8b93c7f848d0bc74507cbaabec4b : SUCCESS in 1h 32m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/891a0a74dc3e4fe48f3b89f7be7d34c7 : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":2},{"id":"33ec61cff1ecdf12c561a55cdaad9f7aed0adb5e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-28 14:50:56.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"1a177431de549d74b3f74036c4a0dd59b9806cbe","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-02 14:06:12.000000000","message":"Patch Set 2: Workflow+1\n\nNinja approving, helpful job to cover our conversion plugin work!!!","accounts_in_message":[],"_revision_number":2},{"id":"daeb67ade70ba6640ea11878140fdd98e3d4630b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 14:06:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b3c9befaf25e73de0aee2f3a798812d62c115ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 21:43:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c492449acfa52bdf7df476a31443d50f235248a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 21:43:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0407e18bb63f488d84621f2ed886d971 : SUCCESS in 1h 04m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04c8d14893ef4846b30b34376bff07f7 : SUCCESS in 1h 36m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c656414bcdb145458465ace883efbe8c : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebeea7831b4540cfa0c716916d7f5fa4 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57ce87dd56574e97921093d02abeef29 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9292cefe998f4e74ab88068c390677e5 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c8ca2efe2984ed988306f9591af20ed : SUCCESS in 8m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b7b054ba71f487dbdc30f059dac4c01 : SUCCESS in 8m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9360b43b468b4953b302c5263d87e5b7 : SUCCESS in 8m 38s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/0f546de11f594d9b817f7af69853753a : SUCCESS in 54m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e97907494e864195a0d957fe4353fd6c : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":2},{"id":"1db02d06ae0112852a05311b83d6436f76105799","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 21:49:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3468709ca0cf400cbeec0056dadf63ae : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c023445d051d0411998743df7fbdbefaaad1bd6c","revisions":{"dc0ddba9b5a99bd874f1169150fdbafd5479338c":{"kind":"REWORK","_number":1,"created":"2020-07-21 23:06:05.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/09/742309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/742309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/742309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/742309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/742309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/742309/1"}}},"commit":{"parents":[{"commit":"7d49b7ffc22d8835610c7670acbac0b910afcc63","subject":"Make our tempest job use import, standalone, and inject_metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d49b7ffc22d8835610c7670acbac0b910afcc63"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 22:54:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:05:32.000000000","tz":-420},"subject":"Make our import-workflow job also convert images to raw","message":"Make our import-workflow job also convert images to raw\n\nThis enables image conversion on the import-workflow job so that we at\nleast run those code paths somewhere in CI.\n\nChange-Id: Ie4a9171f002b42a13c1786268057bdc0ab3804d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc0ddba9b5a99bd874f1169150fdbafd5479338c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc0ddba9b5a99bd874f1169150fdbafd5479338c"}]},"branch":"refs/heads/master"},"c023445d051d0411998743df7fbdbefaaad1bd6c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-27 14:31:51.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/09/742309/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/742309/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/742309/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/742309/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/742309/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/742309/2"}}},"commit":{"parents":[{"commit":"1b006c4f44e56d08d5ca1fb9090dc83990395d87","subject":"Add functional test for task status updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b006c4f44e56d08d5ca1fb9090dc83990395d87"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 22:54:24.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-27 14:29:12.000000000","tz":-420},"subject":"Make our import-workflow job also convert images to raw","message":"Make our import-workflow job also convert images to raw\n\nThis enables image conversion on the import-workflow job so that we at\nleast run those code paths somewhere in CI.\n\nChange-Id: Ie4a9171f002b42a13c1786268057bdc0ab3804d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c023445d051d0411998743df7fbdbefaaad1bd6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c023445d051d0411998743df7fbdbefaaad1bd6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
