)]}'
{"id":"openstack%2Fglance~742314","triplet_id":"openstack%2Fglance~master~I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6","project":"openstack/glance","branch":"master","topic":"async-native-threads","hashtags":[],"change_id":"I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6","subject":"Make image conversion use a proper python interpreter for prlimit","status":"MERGED","created":"2020-07-21 23:30:02.000000000","updated":"2020-07-27 22:04:24.000000000","submitted":"2020-07-27 22:01:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742314-1595887306124-7a5aba1f","meta_rev_id":"e744e7b68ffe9b0b8392e6e5669476827e019f96","_number":742314,"virtual_id_number":742314,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-27 16:57:52.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-27 22:01:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-27 17:35:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-27 17:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-27 17:35:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-27 16:57:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 16:55:24.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-07-27 16:57:52.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-27 17:35:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-27 22:01:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ed3a0846f763f059ca7e1adc6bb3237a0ea8643","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-21 23:30:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"573647ca2517ff69e0b6efa56ff442a22ab7adaa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-21 23:30:26.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"7998efc691d435b847c5b21133d3ef00862c6167","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-21 23:35:21.000000000","message":"Patch Set 1:\n\nAbhi, here\u0027s a stab at that fix.\n\nI\u0027m getting a little crazy with all the jobs because it\u0027s taking so long to get a result from CI that I\u0027m trying to hedge my bets.\n\nI think the order of landing should probably be:\n\n1. The existing import workflow job that we have\n2. The patch to make the above also do image conversion\n3. The wsgi async patch\n4. This patch\n5. Change devstack default back to wsgi mode\n6. Change the glance jobs to let the tempest-integrated-storage job run as wsgi, the import-workflow job as standalone with everything enabled\n7. Once my nova job lands, add that to glance gate, which will cover ceph, multistore, import\n\nI think gmann is planning to get the tempest stuff landed soon, so maybe tomorrow before any of this can happen anyway...","accounts_in_message":[],"_revision_number":1},{"id":"256477be9b0053bcf9a0317b8262c9fe6d207f71","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-22 02:44:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"066df64ff665287edb380e429e67ff8a99f3166a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 02:44:44.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"46b858cd05402a723c1e667d93ef3a9136e44e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:02:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3a78f107671f4d9789150b6b1204cc8e : SUCCESS in 1h 11m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7b9fcf8e32d9406395def495b4152579 : SUCCESS in 1h 07m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f75a3c030ad4952b90ef75d05bdc112 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f9117facf6b457b8c9a2f52ca5c7998 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb8f89511ca1412183caa64a4acbca5d : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95bbf8b0773848b99140bf56588647ee : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1264e73d3c0f47b88f5e7519c3ccfda8 : FAILURE in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8844b7099c6a4e2b97865e685279f487 : TIMED_OUT in 30m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f299a73fd5744234affe88a54f2b8d0e : SUCCESS in 8m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8f5b6efc922141448ca9d52398d6de6c : SUCCESS in 5m 55s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/dc5097d149d246a9893188d259c70548 : SUCCESS in 1h 00m 06s (non-voting)\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/4fa0f4be7ba3456196b37a17b50609b5 : FAILURE in 30m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2df88b9ad14449dc889312bdba4544a4 : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":2},{"id":"f8b71aadbd805db61cc33ac705ecc60853d8c7a4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-22 13:31:58.000000000","message":"Patch Set 2:\n\nHah, it would help if I set the config I\u0027m adding...","accounts_in_message":[],"_revision_number":2},{"id":"56c46776f6c66a83394240e1fae214f6befdd991","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-22 13:32:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"52e8e01af62d929761f3ac175a4d847e6c67b9ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:33:09.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"76a36bca5c3e2394cb4f0aeec3d4b739647aeb24","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 13:33:25.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"6182b70ec8c05f259c1761044d1fac8808b8f419","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-22 13:38:38.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"427d6857642ae6209d4922df5ec704830da4541e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 13:39:17.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"38b61d04672b6a9048d6fc687fc9fbb18d136378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:39:38.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  while scanning a simple key\n  in \"openstack/glance/.zuul.yaml@master\", line 245, column 13\ncould not find expected \u0027:\u0027\n  in \"openstack/glance/.zuul.yaml@master\", line 246, column 33","accounts_in_message":[],"_revision_number":4},{"id":"643d1a8375bfe0f1d267a843a8a8a7be06a92526","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-22 13:45:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1f840bf47cf8aee330f828a43087c59f0b81f283","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 13:45:41.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"249ac7048814669f9f836ce3b645dd39d65c9f50","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-22 14:35:29.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"c48caa95f81b1d18249698d656a1ae3750b18a18","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 14:36:30.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"d145b96dfef7ba3ab5bd9ccd9c84453ce4a5f3c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 14:52:45.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"e4be1d9eab0864d2a47a607262abeed5efbf4544","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-22 14:53:37.000000000","message":"Patch Set 6:\n\nrecheck dep changed","accounts_in_message":[],"_revision_number":6},{"id":"14cff81dfec5a9579b25e968b4a767bd6c630e65","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 14:54:01.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"9e4524a5c0327d73a5e60d298c5c6bad892294f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:00:29.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/fbd145e0b8ef49879800e7ab22067f5f : SUCCESS in 1h 10m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3af74c47bf99459eaf57423fb8b11250 : FAILURE in 1h 36m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f6a7275c5124b35a662e0e9f325a3a2 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb678852255b448f8e576ceac0f669a0 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a0462d7c084461e882b3c96fec71e56 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9df910e351144ce6ab1421d4cb6873bb : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb8aee0ca894bb19b5448d4148b5f80 : SUCCESS in 7m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/38137155034b4952bc7bf162bad61237 : TIMED_OUT in 33m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/886a3bf4e4a4463d99af6ef0b4035353 : SUCCESS in 9m 05s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ca8f1fc136bc4a9abefc38229a029711 : SUCCESS in 6m 09s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b31bc692c645462c99726a58fbb0ec63 : SUCCESS in 1h 05m 57s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f0b5d4a536a94c6f9aee7742c29f9134 : SUCCESS in 1h 01m 49s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/9d4a41be0bbd443cb87367fc2ac68e7e : SUCCESS in 1h 03m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a270776e6b54acf8cd665a571625fd2 : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":6},{"id":"2e813f7be115f6714e6a02e34cf53e43930d2f20","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-22 19:15:53.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"2e1a41f8f982b184f72ab878cc4cb3fa2d4156f6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 19:16:20.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"8b19b2a1fa8ee75653545e329e257e757eb5a815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 21:22:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1e3af9922fea424389b0e720ae8ff58f : SUCCESS in 1h 09m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17833276cf9f4048953f82fc03b38590 : SUCCESS in 1h 13m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb23aaa6260c41f49d9f84f01908486c : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a252bcccfdf04d3fabbee90315d95108 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b76735ebdf8b4ac1ba4222f5da8b95fd : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ca37b84aa1b4085911dafb1249217d9 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aac122b2e5746fea96d6f20653157e9 : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d03e701b51946f6856a952206a0ae5e : SUCCESS in 7m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/08dcff6736b34a19819a0b8f12612db5 : SUCCESS in 8m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0997e2d9be574f44ade09e78b4afca25 : SUCCESS in 4m 24s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f95ddb39a6ce44ba93aef8869b0759c3 : SUCCESS in 56m 06s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/26f219eea8414fb69e0130f05a796f47 : SUCCESS in 1h 01m 06s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/30efad480abc4bfab1027342c3f8b455 : SUCCESS in 1h 03m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5740de3ce9f46e1be8e34f8ecfa3056 : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":6},{"id":"deff7799589ba25cff75d50b6d9d85f76653a3da","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-23 17:10:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ac4804597fbbd2ad81ef70f7ff7c636bf1c14c1d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-23 17:11:00.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"f8f6d55882369d3a9adfbe247d169729e6b2d2e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 18:53:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a07e837770fc48c996570dc0c64fe68b : SUCCESS in 1h 07m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/05210f82bae6417e80f5ff90106e7c79 : SUCCESS in 1h 30m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dc8854c524a4806b6954de063b62772 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0b3cc9f18344b895c925c0112d0285 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddaebf820cc64a698912434a41c0e190 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de9ba1672c1f42b8851f992e2497316e : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/703413bb8b5e4b0085b72e505f08338b : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2746a8b1e92d4ca8b4ac912afc0a3488 : SUCCESS in 11m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/41dd68fa52cd4541b44275b87092c052 : SUCCESS in 11m 23s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/27fc87da5755405883928b4e3226effe : SUCCESS in 6m 08s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ce24459c2e8748a2a6bbf20330cc67ab : SUCCESS in 48m 05s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/bff3062626ff4c9ea9f6bfcd375ede0e : SUCCESS in 1h 33m 05s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/b5e7d0819e8448ebb5eae7d08efb61bc : FAILURE in 1h 30m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e49c26255ca74df986ede1a6220993d3 : SUCCESS in 1h 11m 40s","accounts_in_message":[],"_revision_number":7},{"id":"c87cad21e72ece4bdc4a92cf91b0a7266cd39668","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-23 19:28:16.000000000","message":"Patch Set 7:\n\nrecheck cinder fails","accounts_in_message":[],"_revision_number":7},{"id":"ce9cf3dbb7fce46fc3375542a9d06b5c40a8a774","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-23 19:28:39.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"03cacb22612f727dfb645044bd407e2f52c6caa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 21:09:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/627d50ad06684d48a65b275a319e5726 : SUCCESS in 1h 24m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/36800db547ad4c4abb644d9cf21e55f7 : SUCCESS in 49m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f0f02dc17c44702ba0b63d0e3a93b2d : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bf1104c455e4e01a5c9faab284dfe18 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/091b4ee30f384c62b19a17c16d61b852 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7460f1a80ee46ef9fa6d63a006e8f00 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af9980e239245a1b0a9be5c264d9cb2 : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d618758edac47f1b0a9f4ae6a994d31 : SUCCESS in 9m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b7288035ed4a44af9bc75c9085da6f5c : SUCCESS in 8m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ed636a2b425a4a1d8c4f9f22bbccd211 : SUCCESS in 6m 10s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f7130a2d8da84037870357355aaaca3f : SUCCESS in 1h 07m 18s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/ce53bc51f00d442793355028a70d9534 : SUCCESS in 1h 11m 13s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/16831232a73a405981c9b9b8eef5d75c : SUCCESS in 1h 12m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa55091a3a6a4c7ba801c84869d66089 : SUCCESS in 51m 29s","accounts_in_message":[],"_revision_number":7},{"id":"78d4a8d0c42fa7107724a612d4f9502980d7dc9d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-23 23:50:07.000000000","message":"Patch Set 7: Code-Review+2\n\nLooks safe with setting the default to sys.executable.","accounts_in_message":[],"_revision_number":7},{"id":"56ffed9185659fac836086185abd1c8a4712e377","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-24 16:41:42.000000000","message":"Patch Set 7: Code-Review+1\n\nI guess we should mention about new config option in our doc.","accounts_in_message":[],"_revision_number":7},{"id":"f5df7a9c5901f984d5936ced0dcbb617c9616c09","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-24 16:55:24.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+1\n\nLooks good to me, will approve once partent patch is merged. This will also give time to others if they want to test it locally.","accounts_in_message":[],"_revision_number":7},{"id":"79d408063d2ba445909d0070f6e0ba512791f6c4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-24 17:49:58.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"9c41f92b9f87d6607bd7f393a1e5690d518e0a7e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-24 17:50:33.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":8},{"id":"40f353543060900c30c3f4c5feee9d019a049178","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-24 18:13:53.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"681c9fd222b928a958111405052b4dc2e33ba0da","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-24 18:14:28.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":9},{"id":"3005019635a0f4a502f5a7f77fe0b26ee12e2b41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 19:28:53.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7f8a626eb6104eb498f2cdba825336cb : SUCCESS in 1h 03m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ae204127d9554119ad7154d0e4bb4053 : SUCCESS in 1h 06m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e0be089b22fe4db9ae9c1192c0ac1cee : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfa317172fce4cbda3e8b27c95ad3ad9 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60b884b901324af5ba5fb8602aabe8aa : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62725ded2fd3435985eb5f87213a46c5 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e826443c2c48d7ab69415eca97a96f : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c68c7c6ffb794e908f807ae9fae82f8c : SUCCESS in 7m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6388fcf9b4494d9eb1bd42bcaf0bffc4 : SUCCESS in 7m 54s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4f899bc40ea047e0a7bd63e2a98f2d63 : SUCCESS in 7m 05s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3b23f6fb92ad4bae8b528bb0ddf8b62f : SUCCESS in 55m 39s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d444d8824c2743708c5eb0a138b82c62 : SUCCESS in 1h 07m 53s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/cf90182754f94d6c98bdb2469e4a676e : SUCCESS in 1h 04m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7973880bba4941f0949354f6479badd1 : SUCCESS in 38m 32s","accounts_in_message":[],"_revision_number":9},{"id":"810ce141e6c6262a3033bade1688b5f531334e61","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-27 15:39:26.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"f17d0936aa4f09694ef220b1d54daab238103330","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-27 16:31:19.000000000","message":"Patch Set 9:\n\nrecheck kick","accounts_in_message":[],"_revision_number":9},{"id":"c38aaf9f1f736bab5c95b5993d7e0368aca762c0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-27 16:31:39.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":9},{"id":"c51c3b7b9e7b94ee1cd558085fc984fcf1867869","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-27 16:57:28.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"3009b9121e01882e34ddd9fe300eb5e28e751c16","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-27 16:57:52.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":10},{"id":"71e5b93415645b7f237d9a47407d8f3375feb305","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-27 17:02:14.000000000","message":"Patch Set 10: Code-Review+2\n\nJust updated commit message to remove depends-on tag, no change in actual code.\n\nWill ninja approve once check job finishes.","accounts_in_message":[],"_revision_number":10},{"id":"25b451ab90f0785bae37857abe6baae8551feb4a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-27 17:35:42.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"2b330f39d5e6eccf4ffae4d9897e8ddec38b223a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 19:55:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f477b78141be40c5a6e704a8fd5aa4e4 : SUCCESS in 1h 05m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5d266dccaf847ca8940119598bfde36 : SUCCESS in 1h 05m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e5d1a5dce574c2ab0a555fe926dbf63 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e6cf8d59b244f278fea37851e8592cd : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0410228f79b94f91a00f49c22b1cb598 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70797ed5195c46acb9ee3b1be6a54a1c : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0faee3402364244b4c4ccba9c3d6739 : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f9a230ca6ad4af0bb75235dc8c8a98b : SUCCESS in 7m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6a1bd5dc00e047f3b7c58d5228d55e45 : SUCCESS in 7m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1c394d6efb1541bfb92e458b92753f7d : SUCCESS in 7m 23s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b4cd19a704334b55a8793a0a8fd851b0 : SUCCESS in 1h 05m 05s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f8558648c4c547b2830d447bad522ba6 : SUCCESS in 1h 26m 08s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/5488c61b01be47ab852577d4560b2a4f : SUCCESS in 1h 07m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/917640095d464dec93ed15bb7af6f798 : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":10},{"id":"84b8ac13c59c9308410b09729ed5ca78ec954694","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 19:56:19.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"4735a77f5d33e5613fd4b4d6402edebcb15a9bfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 22:01:45.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a2cc5557716a4501816ae6af2d48e84d : SUCCESS in 1h 09m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d901c5f7c2034d63b5ecd6f4d025b16d : SUCCESS in 1h 05m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0b756ee7a3b430780edeb8dfce8c5c0 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9532cfb75bf34389b504a86ccba7978f : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8c0e90ba9d74a07a339a70b47dfa6cb : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d8442588c7d483cbd7e4efc7ec8805d : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5667dadacaa484d9a0197890bd06c70 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f321a19c7d264c0699f75931e049905a : SUCCESS in 7m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cffb673dc3244abc8ea530eb79891b05 : SUCCESS in 8m 13s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/7b4ad2d4c8844e949e8b35c6c96fde2a : SUCCESS in 1h 04m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41b30edf8ee54352950e3ca7c9312e47 : SUCCESS in 1h 10m 30s","accounts_in_message":[],"_revision_number":10},{"id":"e635c48442b4fed0ef919a60a0b458d1c1ecf021","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 22:01:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"cee383c070538ea0bfc6b5fbcf6754b85bac6b61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 22:04:24.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9265d1e608ea4f87af1677b4905b7827 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"783fa72f481d68962509e708585fea4c163d5bf4","revisions":{"2921d17f8eae48df1cc2493b8e7da9a953089787":{"kind":"REWORK","_number":1,"created":"2020-07-21 23:30:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/742314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/742314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/742314/1"}}},"commit":{"parents":[{"commit":"f112f9e2e300fdcaa3415ec8bf0cb1c06b6fbdbd","subject":"Make glance-api able to do async tasks in WSGI mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f112f9e2e300fdcaa3415ec8bf0cb1c06b6fbdbd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:25:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:30:09.000000000","tz":-420},"subject":"Make image conversion use a proper python interpreter for prlimit","message":"Make image conversion use a proper python interpreter for prlimit\n\nThe image conversion plugin does a processutils exec(), which needs\nto spawn python for prlimit support. Under uwsgi, sys.executable\npoints to uwsgi itself, which won\u0027t work in this case. This introduces\na [wsgi]/python_interpreter config option (because I don\u0027t think\nthere is any way to get this from uwsgi itself) which we use for\nthe exec. By default, it\u0027s sys.executable, which is what is used\nright now so nobody should notice a change unless they need it.\n\nChange-Id: I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2921d17f8eae48df1cc2493b8e7da9a953089787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2921d17f8eae48df1cc2493b8e7da9a953089787"}]},"branch":"refs/heads/master"},"cc983bc43ae53772a7bd1daf342ddfcbf340a54b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-22 02:44:17.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/742314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/742314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/742314/2"}}},"commit":{"parents":[{"commit":"f112f9e2e300fdcaa3415ec8bf0cb1c06b6fbdbd","subject":"Make glance-api able to do async tasks in WSGI mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f112f9e2e300fdcaa3415ec8bf0cb1c06b6fbdbd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:25:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-22 02:43:52.000000000","tz":-420},"subject":"Make image conversion use a proper python interpreter for prlimit","message":"Make image conversion use a proper python interpreter for prlimit\n\nThe image conversion plugin does a processutils exec(), which needs\nto spawn python for prlimit support. Under uwsgi, sys.executable\npoints to uwsgi itself, which won\u0027t work in this case. This introduces\na [wsgi]/python_interpreter config option (because I don\u0027t think\nthere is any way to get this from uwsgi itself) which we use for\nthe exec. By default, it\u0027s sys.executable, which is what is used\nright now so nobody should notice a change unless they need it.\n\nNote: Making this depend on the devstack change to remove the wsgi\nimport restriction so we can get a test on it.\n\nDepends-On: https://review.opendev.org/#/c/742332/\nChange-Id: I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc983bc43ae53772a7bd1daf342ddfcbf340a54b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cc983bc43ae53772a7bd1daf342ddfcbf340a54b"}]},"branch":"refs/heads/master"},"b362f303e9e776903182751021dc430c30a39c58":{"kind":"REWORK","_number":3,"created":"2020-07-22 13:32:57.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/742314/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/742314/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/742314/3"}}},"commit":{"parents":[{"commit":"f112f9e2e300fdcaa3415ec8bf0cb1c06b6fbdbd","subject":"Make glance-api able to do async tasks in WSGI mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f112f9e2e300fdcaa3415ec8bf0cb1c06b6fbdbd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:25:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-22 13:33:08.000000000","tz":-420},"subject":"Make image conversion use a proper python interpreter for prlimit","message":"Make image conversion use a proper python interpreter for prlimit\n\nThe image conversion plugin does a processutils exec(), which needs\nto spawn python for prlimit support. Under uwsgi, sys.executable\npoints to uwsgi itself, which won\u0027t work in this case. This introduces\na [wsgi]/python_interpreter config option (because I don\u0027t think\nthere is any way to get this from uwsgi itself) which we use for\nthe exec. By default, it\u0027s sys.executable, which is what is used\nright now so nobody should notice a change unless they need it.\n\nNote: Making this depend on the devstack change to remove the wsgi\nimport restriction so we can get a test on it.\n\nDepends-On: https://review.opendev.org/#/c/742332/\nChange-Id: I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b362f303e9e776903182751021dc430c30a39c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b362f303e9e776903182751021dc430c30a39c58"}]},"branch":"refs/heads/master"},"d5132af97a00c51b5563c97856666f6b36c0dc51":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-22 13:38:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/742314/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/742314/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/742314/4"}}},"commit":{"parents":[{"commit":"0e68345521c43d28570385f4599c45a3df7d0a24","subject":"Make glance-api able to do async tasks in WSGI mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e68345521c43d28570385f4599c45a3df7d0a24"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:25:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-22 13:38:26.000000000","tz":-420},"subject":"Make image conversion use a proper python interpreter for prlimit","message":"Make image conversion use a proper python interpreter for prlimit\n\nThe image conversion plugin does a processutils exec(), which needs\nto spawn python for prlimit support. Under uwsgi, sys.executable\npoints to uwsgi itself, which won\u0027t work in this case. This introduces\na [wsgi]/python_interpreter config option (because I don\u0027t think\nthere is any way to get this from uwsgi itself) which we use for\nthe exec. By default, it\u0027s sys.executable, which is what is used\nright now so nobody should notice a change unless they need it.\n\nNote: Making this depend on the devstack change to remove the wsgi\nimport restriction so we can get a test on it.\n\nDepends-On: https://review.opendev.org/#/c/742332/\nChange-Id: I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5132af97a00c51b5563c97856666f6b36c0dc51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d5132af97a00c51b5563c97856666f6b36c0dc51"}]},"branch":"refs/heads/master"},"476a82e513df9e7bb40e74fd6f4826c8f5f3b08d":{"kind":"REWORK","_number":5,"created":"2020-07-22 13:45:17.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/742314/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/742314/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/742314/5"}}},"commit":{"parents":[{"commit":"0e68345521c43d28570385f4599c45a3df7d0a24","subject":"Make glance-api able to do async tasks in WSGI mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0e68345521c43d28570385f4599c45a3df7d0a24"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:25:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-22 13:45:27.000000000","tz":-420},"subject":"Make image conversion use a proper python interpreter for prlimit","message":"Make image conversion use a proper python interpreter for prlimit\n\nThe image conversion plugin does a processutils exec(), which needs\nto spawn python for prlimit support. Under uwsgi, sys.executable\npoints to uwsgi itself, which won\u0027t work in this case. This introduces\na [wsgi]/python_interpreter config option (because I don\u0027t think\nthere is any way to get this from uwsgi itself) which we use for\nthe exec. By default, it\u0027s sys.executable, which is what is used\nright now so nobody should notice a change unless they need it.\n\nNote: Making this depend on the devstack change to remove the wsgi\nimport restriction so we can get a test on it.\n\nDepends-On: https://review.opendev.org/#/c/742332/\nChange-Id: I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/476a82e513df9e7bb40e74fd6f4826c8f5f3b08d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/476a82e513df9e7bb40e74fd6f4826c8f5f3b08d"}]},"branch":"refs/heads/master"},"3d13e77b34644c03506f08cc88431db95a294d33":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-07-22 14:35:29.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/742314/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/742314/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/742314/6"}}},"commit":{"parents":[{"commit":"53819ba9beb80ea3a1ce9879419ae198af699189","subject":"Make glance-api able to do async tasks in WSGI mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53819ba9beb80ea3a1ce9879419ae198af699189"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:25:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-22 14:35:42.000000000","tz":-420},"subject":"Make image conversion use a proper python interpreter for prlimit","message":"Make image conversion use a proper python interpreter for prlimit\n\nThe image conversion plugin does a processutils exec(), which needs\nto spawn python for prlimit support. Under uwsgi, sys.executable\npoints to uwsgi itself, which won\u0027t work in this case. This introduces\na [wsgi]/python_interpreter config option (because I don\u0027t think\nthere is any way to get this from uwsgi itself) which we use for\nthe exec. By default, it\u0027s sys.executable, which is what is used\nright now so nobody should notice a change unless they need it.\n\nNote: Making this depend on the devstack change to remove the wsgi\nimport restriction so we can get a test on it.\n\nDepends-On: https://review.opendev.org/#/c/742332/\nChange-Id: I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d13e77b34644c03506f08cc88431db95a294d33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d13e77b34644c03506f08cc88431db95a294d33"}]},"branch":"refs/heads/master"},"1de3a1b281a1def8cf77f98fe3fde45052cda376":{"kind":"REWORK","_number":7,"created":"2020-07-23 17:10:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/742314/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/742314/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/742314/7"}}},"commit":{"parents":[{"commit":"6a063a17a08e5a819483479fffd50e8cff6f6b2d","subject":"Make glance-api able to do async tasks in WSGI mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a063a17a08e5a819483479fffd50e8cff6f6b2d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:25:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-23 17:09:12.000000000","tz":-420},"subject":"Make image conversion use a proper python interpreter for prlimit","message":"Make image conversion use a proper python interpreter for prlimit\n\nThe image conversion plugin does a processutils exec(), which needs\nto spawn python for prlimit support. Under uwsgi, sys.executable\npoints to uwsgi itself, which won\u0027t work in this case. This introduces\na [wsgi]/python_interpreter config option (because I don\u0027t think\nthere is any way to get this from uwsgi itself) which we use for\nthe exec. By default, it\u0027s sys.executable, which is what is used\nright now so nobody should notice a change unless they need it.\n\nNote: Making this depend on the devstack change to remove the wsgi\nimport restriction so we can get a test on it.\n\nPartial-Bug: #1888713\nDepends-On: https://review.opendev.org/#/c/742332/\nChange-Id: I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1de3a1b281a1def8cf77f98fe3fde45052cda376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1de3a1b281a1def8cf77f98fe3fde45052cda376"}]},"branch":"refs/heads/master"},"dd41dca4327dea45c43addf06b51e397f0f31a07":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-07-24 17:49:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/742314/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/742314/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/742314/8"}}},"commit":{"parents":[{"commit":"86ca2bd6442e7b0a21193f2d39e7f5516aa50ef9","subject":"Make glance-api able to do async tasks in WSGI mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/86ca2bd6442e7b0a21193f2d39e7f5516aa50ef9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:25:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 17:49:46.000000000","tz":-420},"subject":"Make image conversion use a proper python interpreter for prlimit","message":"Make image conversion use a proper python interpreter for prlimit\n\nThe image conversion plugin does a processutils exec(), which needs\nto spawn python for prlimit support. Under uwsgi, sys.executable\npoints to uwsgi itself, which won\u0027t work in this case. This introduces\na [wsgi]/python_interpreter config option (because I don\u0027t think\nthere is any way to get this from uwsgi itself) which we use for\nthe exec. By default, it\u0027s sys.executable, which is what is used\nright now so nobody should notice a change unless they need it.\n\nNote: Making this depend on the devstack change to remove the wsgi\nimport restriction so we can get a test on it.\n\nPartial-Bug: #1888713\nDepends-On: https://review.opendev.org/#/c/742332/\nChange-Id: I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd41dca4327dea45c43addf06b51e397f0f31a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd41dca4327dea45c43addf06b51e397f0f31a07"}]},"branch":"refs/heads/master"},"708c36b1ec1f05dc197fb9751b712fd3aaac5d1c":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-07-24 18:13:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/742314/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/742314/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/742314/9"}}},"commit":{"parents":[{"commit":"16a5431c665adadbfcb9909cf3a73334d796c19a","subject":"Make glance-api able to do async tasks in WSGI mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/16a5431c665adadbfcb9909cf3a73334d796c19a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:25:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 18:13:45.000000000","tz":-420},"subject":"Make image conversion use a proper python interpreter for prlimit","message":"Make image conversion use a proper python interpreter for prlimit\n\nThe image conversion plugin does a processutils exec(), which needs\nto spawn python for prlimit support. Under uwsgi, sys.executable\npoints to uwsgi itself, which won\u0027t work in this case. This introduces\na [wsgi]/python_interpreter config option (because I don\u0027t think\nthere is any way to get this from uwsgi itself) which we use for\nthe exec. By default, it\u0027s sys.executable, which is what is used\nright now so nobody should notice a change unless they need it.\n\nNote: Making this depend on the devstack change to remove the wsgi\nimport restriction so we can get a test on it.\n\nPartial-Bug: #1888713\nDepends-On: https://review.opendev.org/#/c/742332/\nChange-Id: I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/708c36b1ec1f05dc197fb9751b712fd3aaac5d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/708c36b1ec1f05dc197fb9751b712fd3aaac5d1c"}]},"branch":"refs/heads/master"},"783fa72f481d68962509e708585fea4c163d5bf4":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2020-07-27 16:57:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/742314/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/742314/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/742314/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/742314/10"}}},"commit":{"parents":[{"commit":"16a5431c665adadbfcb9909cf3a73334d796c19a","subject":"Make glance-api able to do async tasks in WSGI mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/16a5431c665adadbfcb9909cf3a73334d796c19a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-21 23:25:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 16:57:16.000000000","tz":-420},"subject":"Make image conversion use a proper python interpreter for prlimit","message":"Make image conversion use a proper python interpreter for prlimit\n\nThe image conversion plugin does a processutils exec(), which needs\nto spawn python for prlimit support. Under uwsgi, sys.executable\npoints to uwsgi itself, which won\u0027t work in this case. This introduces\na [wsgi]/python_interpreter config option (because I don\u0027t think\nthere is any way to get this from uwsgi itself) which we use for\nthe exec. By default, it\u0027s sys.executable, which is what is used\nright now so nobody should notice a change unless they need it.\n\nNote: Making this depend on the devstack change to remove the wsgi\nimport restriction so we can get a test on it.\n\nPartial-Bug: #1888713\nChange-Id: I7cb2e135d6ea2cb21de55060df3f7bf40b3e64b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/783fa72f481d68962509e708585fea4c163d5bf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/783fa72f481d68962509e708585fea4c163d5bf4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
