)]}'
{"id":"openstack%2Fglance~742373","triplet_id":"openstack%2Fglance~master~I89fc8ef71ad58981238a4733687acd9bd0ae71d3","project":"openstack/glance","branch":"master","topic":"poc/lazy-loading-cinder","hashtags":[],"change_id":"I89fc8ef71ad58981238a4733687acd9bd0ae71d3","subject":"PoC for lazy loading cinder store","status":"ABANDONED","created":"2020-07-22 08:49:30.000000000","updated":"2020-09-09 14:07:20.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6f47cb15f1df0805879565640d76090cbc4628b6","_number":742373,"virtual_id_number":742373,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-12 06:54:55.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-11 20:46:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":-1,"date":"2020-08-19 01:17:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-08-11 19:57:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 16:59:03.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-08-11 20:46:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-12 06:54:55.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-19 01:17:35.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"}],"messages":[{"id":"fd8ff1cf64cf40659954c104c98306b5f0ddc390","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-22 08:49:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74515b6c08d4b339256714d1cae7861dc09ee0d0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-22 08:49:50.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"5c84339b8af8dc1fcb965c50f72b54c597f9ef6e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-22 08:49:55.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1c2350743f23a281477f13258277fdc3f631ebee","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-22 08:58:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b63edc2bbbbf923f65345f265bf5ae579139c362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 10:50:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/89b6af4c5298458ab35dda65fe738b51 : SUCCESS in 1h 06m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f179208f5f3b4d5ab1b38ed835a21a20 : SUCCESS in 1h 17m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be8df5115b5241968e71a10cf0d32a81 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f6ce0c24a84e8199638030b0de8fd6 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/301a29993ef24646804acf054396f740 : SUCCESS in 9m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e8fa8eba05b4c1aa870a27163861762 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/675862f7a61b4df399ddc08c077817c3 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c6a49457a9324276870cc9f670675965 : SUCCESS in 14m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/08d491a0cbbc480587857c1cfaa5cbb5 : SUCCESS in 9m 23s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/65fb7659a0f14553a8306ae8b4722b86 : SUCCESS in 5m 45s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/cb8be6946ccc4ef79308332fdaa1e8df : SUCCESS in 58m 07s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/2ef67e2e4172411a94e5dcca62050026 : SUCCESS in 1h 03m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45f6b8e571e94a6db57d2401fffcada9 : SUCCESS in 52m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c0667badf24c5863d338588f3ccd5b687fb3f7f9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-29 16:54:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e3eed7c513a6de3c8afb6f10b41050f8b43d32eb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-29 16:55:05.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"523ab277fd9c118aac0eca040fab3368caf458cb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-29 16:55:36.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"1998d5583054ebb72e88e7bae51a402e9f0cf9cd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-29 16:57:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"40697aa4fbb0b44a460b7354c748d33087663bdb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-29 16:58:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d617923bf52f62c97580a9e464c225a726f7b1ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 16:58:32.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  Job tempest-multistore-cinder-import-workflow not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - integrated-gate-storage\n        - openstack-lower-constraints-jobs\n        - openstack-python3-victoria-jobs\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n      check:\n        jobs:\n          - openstack-tox-functional-py36\n          - openstack-tox-functional-py38\n          - glance-code-constants-check\n          - devstack-plugin-ceph-tempest-py3:\n              voting: false\n              irrelevant-files: \u0026tempest-irrelevant-files\n                - ^(test-|)requirements.txt$\n                - ^lower-constraints.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^glance/hacking/.*$\n                - ^glance/locale/.*$\n                - ^glance/tests/.*$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^tools/.*$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n          - tempest-integrated-storage:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-integrated-storage-import-workflow:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-integrated-storage-wsgi-import:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-multistore-cinder-import-workflow:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-ipv6-only:\n              irrelevant-files: *tempest-irrelevant-files\n  \n      gate:\n        jobs:\n          - openstack-tox-functional-py36\n          - openstack-tox-functional-py38\n          - tempest-integrated-storage:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-integrated-storage-import-workflow:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-multistore-cinder-import-workflow:\n              irrelevant-files: *tempest-irrelevant-files\n          - grenade:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-ipv6-only:\n              irrelevant-files: *tempest-irrelevant-files\n      experimental:\n        jobs:\n          - glance-tox-py38-glance_store-tips\n          - glance-tox-py36-glance_store-tips\n          - glance-tox-functional-py38-glance_store-tips\n          - glance-tox-functional-py36-glance_store-tips\n          - barbican-simple-crypto-devstack-tempest\n          - grenade-multinode\n          - tempest-pg-full:\n              irrelevant-files: *tempest-irrelevant-files\n          - tempest-full-py3-opensuse15:\n              irrelevant-files: *tempest-irrelevant-files\n      periodic:\n        jobs:\n          # NOTE(rosmaita): we only want the \"tips\" jobs to be run against\n          # master, hence the \u0027branches\u0027 qualifiers below.  Without them, when\n          # a stable branch is cut, the tests would be run against the stable\n          # branch as well, which is pointless because these libraries are\n          # frozen (more or less) in the stable branches.\n          #\n          # The \"tips\" jobs can be removed from the stable branch .zuul.yaml\n          # files if someone is so inclined, but that would require manual\n          # maintenance, so we do not do it by default.  Another option is\n          # to define these jobs in the openstack/project-config repo.\n          # That would make us less agile in adjusting these tests, so we\n          # aren\u0027t doing that either.\n          - glance-tox-functional-py38-oslo-tips:\n              branches: master\n          - glance-tox-functional-py36-oslo-tips:\n              branches: master\n          - glance-tox-py38-keystone-tips:\n              branches: master\n          - glance-tox-py36-keystone-tips:\n              branches: master\n          - glance-tox-functional-py38-keystone-tips:\n              branches: master\n          - glance-tox-functional-py36-keystone-tips:\n              branches: master\n          - glance-tox-py38-glance_store-tips:\n              branches: master\n          - glance-tox-py36-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py38-glance_store-tips:\n              branches: master\n          - glance-tox-functional-py36-glance_store-tips:\n              branches: master\n          - glance-tox-py38-cursive-tips:\n              branches: master\n          - glance-tox-py36-cursive-tips:\n              branches: master\n          - glance-tox-functional-py38-cursive-tips:\n              branches: master\n          - glance-tox-functional-py36-cursive-tips:\n              branches: master\n\n  in \"openstack/glance/.zuul.yaml@master\", line 268, column 3","accounts_in_message":[],"_revision_number":3},{"id":"2a87898660d57b20a1a962c07e4c6b35e9e94dc0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-29 16:58:33.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"1ff43d9a93a360ba4f5d79e907606d080b3d292b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-29 16:59:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a00f57800aed9a5a6bf8bb87ed2f0a21fd536f0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-29 17:03:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"19aac17449011bf543eeca49d133056cc7c404ac","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-29 17:03:28.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"1cba276ea8747d891e14d937fee28cf6c39eb3b7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-29 17:03:31.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThank you for review GMann!","accounts_in_message":[],"_revision_number":3},{"id":"a0c5c06a9567889aab4acaed738695b4a386349c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-29 17:03:40.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"040c5c6c252020e1460619ab963b5244ace4fcd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 23:48:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/49608d02c7234d29b2aa455f061a7859 : SUCCESS in 1h 03m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/93a05c05eb3f460f8b8c54320f7b86c6 : SUCCESS in 48m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9923a49f42047f4a05c65a2e4cc6f03 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b622b62876e4287b8983a16393ecedf : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f737839334a84ef08b63cc49131c9ad2 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08baccc3beef454399231d41ac11a9ea : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f667c410a0c14979ae982d2aa86fdb2b : SUCCESS in 8m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a7d8532f2194b319be05c38c495d75a : SUCCESS in 6m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/05e806061f354d36af6dc19d7974b3c7 : SUCCESS in 8m 24s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fa8e102bf5b54ae89d31af491501e523 : FAILURE in 5m 46s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/14fffbd9fb5243c2b7444ff37b6d79a8 : SUCCESS in 57m 53s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/43fe4211646a43c1a225bf02356e0a53 : SUCCESS in 1h 11m 49s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/780078ab6c40421681da1ee616acf84e : SUCCESS in 1h 10m 42s\n- tempest-multistore-cinder-import-workflow https://zuul.opendev.org/t/openstack/build/d7a5f7724e8c48f09f884c0ebe139a05 : SUCCESS in 1h 14m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6b938a463e4474a9c7aea674102df03 : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":4},{"id":"7da69f22347702b880f443193769d0c060a4ee78","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-31 16:53:50.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6572f06d15dd2382111adc0e4d38927bcaeb933a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-31 16:54:15.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"6684fd2623d63ec6f02c5d63ed04197ecc6b7448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 23:35:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/730989a760f746aa8ef9554e0ba43461 : SUCCESS in 55m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8289d74559d64a3f8e6b7921b2b17a99 : SUCCESS in 46m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a42de49557b40fa9806c89b6324cf97 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759af4d3ce2b46d7af77e7eb3f30b096 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66f1244b1efb482eb5537e3a650c6e79 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0484aea356c5402e8d189479c3b28150 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e92ef7cb1a94a8785de4666c31a85e2 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f34d6074cd6948d8a480b8d6c4d1efbb : SUCCESS in 7m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/70dbe61e18ff4ed6a3a8f02bad56bfa0 : SUCCESS in 7m 43s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/39be72401bf04f34a72a8721ebf0a8fe : SUCCESS in 4m 42s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/102286104e5845fdaa7a6be07b3ccc4d : SUCCESS in 56m 42s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/56ba7e6a4fc045cb9e075be714a98914 : SUCCESS in 1h 05m 10s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/13240d9d3ccb46ff9704eedf1d1ddee7 : SUCCESS in 1h 04m 37s\n- tempest-multistore-cinder-import-workflow https://zuul.opendev.org/t/openstack/build/4d270ea7ca9943be81d5a2f7b5f9aae1 : SUCCESS in 1h 14m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70a09a40eeef437e9c52b6d3ca8aa1ee : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":4},{"id":"720232e7b2dfe326ad78c6754620db7fcc46c951","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-11 17:02:52.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"cb4af85e69d0321ae4e823885e697c5f20635403","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-11 17:03:15.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"7170104441241f0fdcfb00bc6c9466799bd55eed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 18:20:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bd36747e60f04a9ea23fb1b0bb2267e0 : SUCCESS in 56m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c57726cd2b0472c9ac1d4874c5ea986 : SUCCESS in 1h 05m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0eb707793c73421fbb73e2ca69d2359c : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/867a963294c14af8aea32eb421d14ea0 : SUCCESS in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7ecd59132fc467caa2c2629dac3f66f : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07ff73fe670448a1b1bd40fb5f04cdd2 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c4c9954fc3d493a817b2c8d0bd5c3cf : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22746667bbf646f2a09c341f6380c57b : TIMED_OUT in 31m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/30da1b056630462d874d0b79c03d8565 : SUCCESS in 7m 36s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/97aa68f0be054bc9a8085b9b972fe970 : SUCCESS in 6m 31s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3bdc8d9f517149f2a70113f57921216a : SUCCESS in 1h 08m 41s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d5b5270814ad4ab3813ae543331987fb : SUCCESS in 1h 11m 49s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/92ab33912d21452f9dd743f24ad32da1 : SUCCESS in 1h 03m 29s\n- tempest-multistore-cinder-import-workflow https://zuul.opendev.org/t/openstack/build/6800a6db36724007b402683bfe865495 : FAILURE in 31m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1c63788f9ef449db361187563559bdb : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":4},{"id":"52a719b0efff5d59e92d0c3d25c57e532ea53161","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-11 19:53:35.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9ddb935116cc7a4ce8915382e2180844eab5c38a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-11 19:57:10.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"a144da1e97d6d5621b03ab7e53c8bf72cdc92b01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 20:46:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/cb17f86b1fce4b2ca1e35dfe158a36d2 : FAILURE in 47m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30723bdf22b54b5197c8462dcaa29d9b : FAILURE in 15m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8c5fa5656314f99b9f456776f631ac8 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f649a84f885e49d9b7ec43f7c2929dcb : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34ddc499ef6d40c0b5cd07b6035acfd2 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22c78988ee4040e9b3be345bbaf1385d : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbfe331a5bb6476bb5b3f278a9ed4476 : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/382bae02b8ac417d996c4aa3cad8981f : SUCCESS in 6m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/de7e1751574645a2a229430769419847 : SUCCESS in 8m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8093bdcf7d6849e7bcf959270ea28517 : SUCCESS in 5m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/864e075ef6424805976c0d8c1e4734ee : FAILURE in 17m 01s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d6b23d45bcfc44368242a3018a1e0a14 : POST_FAILURE in 10m 40s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/df8eb686a5004785ad4b2a0e6d829ff8 : FAILURE in 9m 13s\n- tempest-multistore-cinder-import-workflow https://zuul.opendev.org/t/openstack/build/9697d90bc5b547b58b7c4720a97603a6 : POST_FAILURE in 16m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff72f66e9c194522bcd036945e766dff : FAILURE in 15m 07s","accounts_in_message":[],"_revision_number":5},{"id":"169313c326be9ac6b3ca7a249b47778322a46813","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-11 21:14:35.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"3fffb6810a3585cb149f1ff7932a97e227635ad0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 22:07:32.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/fc04ee74fcd7403f8d0f43005d98d410 : FAILURE in 41m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e5e22a105b74d46b357083cd1a7bcdf : FAILURE in 15m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19b937f68cb54f1387cb6d7eb7f8a1fb : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1bd79f5881d40e0b4c00aff86a4c553 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ff00bd1cca14d4d8293cb48028e7f34 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/174e8469ed2e47c582b8c874f57f8a23 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c1bab23b4648679e63b72952d00014 : SUCCESS in 4m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb2d9965ebd94c358dc9407570ffb27a : SUCCESS in 6m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/08d9e8d4624d404993d54c2ca794e9f0 : SUCCESS in 9m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9ba02b9064dc4171862aefb48e23c4eb : SUCCESS in 4m 34s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/35f765088b1341dbb36c8cb87aabe174 : FAILURE in 12m 57s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/db74e4aac5914b27a8782a11b2e94812 : POST_FAILURE in 10m 41s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/685e8dba5d84407e8f52014cddea4b79 : FAILURE in 15m 47s\n- tempest-multistore-cinder-import-workflow https://zuul.opendev.org/t/openstack/build/ea46fcda67104dd184c32ae407eac639 : POST_FAILURE in 10m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9ab79b082384f73be8d2485cc966f04 : FAILURE in 9m 28s","accounts_in_message":[],"_revision_number":5},{"id":"7cc706e20b6b29913a8259825d0bdd3d79b7ff8a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-12 06:54:31.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"0e78ac51bb94e17fc82be08458eefe3a871370d0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-12 06:54:55.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"38c49e902471f24170125f5dad34ff52ae8d100f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 08:27:10.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/289144a3267443e6b1a72307e365536e : FAILURE in 51m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ecd6e172e984dfda26a641439e38e8a : FAILURE in 16m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b619028de07466890d5fdb4064d5515 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af244b7897c944488dc3f22cad73c11b : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fc84cf94b2547b3bea80b03723cd047 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/412d0610d6d446599d6008d923a6f7dd : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d5005875134483ab9b0c599c1f2c3f : SUCCESS in 10m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0e05781a255422dacb1922aba11889b : SUCCESS in 7m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/63c669312e4341d38d6af84082a94a2e : SUCCESS in 7m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4f956a999db44120a49b0b9b9e92c147 : SUCCESS in 6m 37s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/703409f5c2a041f190ef5f3b60a6de4e : FAILURE in 14m 51s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/a82ae55e9e714110a293330de3a6dca4 : SUCCESS in 1h 10m 34s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/76af68e0c797430e93d5af858daa3dd6 : FAILURE in 15m 19s\n- tempest-multistore-cinder-import-workflow https://zuul.opendev.org/t/openstack/build/24d3dca407eb4fee82ed52f02e833c10 : SUCCESS in 1h 10m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3efcb8c919da448fb6bd05d12f7689e3 : FAILURE in 16m 11s","accounts_in_message":[],"_revision_number":5},{"id":"3e771fb24f62157ea0955a192127199f16883ac3","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-19 01:17:35.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d96468bbaf2c4e980d71c7b053798e5ee1a9861e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-20 13:48:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"85056e1625fb84ba1e8755908d1dbc1dcec010cb","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-25 20:17:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bcffa50ffbe849ca5729f1a4158667cff99bc8f1","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-08-25 20:32:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0a42ebe86c47423051e9971cf012ad80b1d3a480","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-09-09 14:07:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1cff883b71f534675b0fab7f1d11c1aca927417a","revisions":{"eebdfe9123f74da91e78c44d47bbfc77a8858fec":{"kind":"REWORK","_number":1,"created":"2020-07-22 08:49:30.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/73/742373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/742373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/742373/1"}}},"commit":{"parents":[{"commit":"5fed61d100955b9e7c853942d1e866ebd7945639","subject":"Merge \"Make our tempest job use import, standalone, and inject_metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5fed61d100955b9e7c853942d1e866ebd7945639"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-03 20:00:24.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-22 08:49:06.000000000","tz":0},"subject":"PoC for lazy loading cinder store","message":"PoC for lazy loading cinder store\n\nChange-Id: I89fc8ef71ad58981238a4733687acd9bd0ae71d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eebdfe9123f74da91e78c44d47bbfc77a8858fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eebdfe9123f74da91e78c44d47bbfc77a8858fec"}]},"branch":"refs/heads/master"},"f652d88a074ac74fc49153350b2a39a8cf4e8f1e":{"kind":"REWORK","_number":2,"created":"2020-07-29 16:54:44.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/73/742373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/742373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/742373/2"}}},"commit":{"parents":[{"commit":"cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82","subject":"Merge \"Refresh Glance example configs for Victoria milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-03 20:00:24.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-29 16:54:34.000000000","tz":0},"subject":"PoC for lazy loading cinder store","message":"PoC for lazy loading cinder store\n\nDepends-On: https://review.opendev.org/742371\nDepends-On: https://review.opendev.org/743800\n\nChange-Id: I89fc8ef71ad58981238a4733687acd9bd0ae71d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f652d88a074ac74fc49153350b2a39a8cf4e8f1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f652d88a074ac74fc49153350b2a39a8cf4e8f1e"}]},"branch":"refs/heads/master"},"f0485671be21d79dad1a093facd550ddce13bb9a":{"kind":"REWORK","_number":3,"created":"2020-07-29 16:58:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/73/742373/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/742373/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/742373/3"}}},"commit":{"parents":[{"commit":"cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82","subject":"Merge \"Refresh Glance example configs for Victoria milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-03 20:00:24.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-29 16:58:03.000000000","tz":0},"subject":"PoC for lazy loading cinder store","message":"PoC for lazy loading cinder store\n\nDepends-On: https://review.opendev.org/742371\nDepends-On: https://review.opendev.org/743800\n\nChange-Id: I89fc8ef71ad58981238a4733687acd9bd0ae71d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0485671be21d79dad1a093facd550ddce13bb9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0485671be21d79dad1a093facd550ddce13bb9a"}]},"branch":"refs/heads/master"},"261071b798f57d8a9e9845c93bc46ca00df7ab8d":{"kind":"REWORK","_number":4,"created":"2020-07-29 17:03:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/73/742373/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/742373/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/742373/4"}}},"commit":{"parents":[{"commit":"cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82","subject":"Merge \"Refresh Glance example configs for Victoria milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-03 20:00:24.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-29 17:02:58.000000000","tz":0},"subject":"PoC for lazy loading cinder store","message":"PoC for lazy loading cinder store\n\nDepends-On: https://review.opendev.org/742371\nDepends-On: https://review.opendev.org/743800\n\nChange-Id: I89fc8ef71ad58981238a4733687acd9bd0ae71d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/261071b798f57d8a9e9845c93bc46ca00df7ab8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/261071b798f57d8a9e9845c93bc46ca00df7ab8d"}]},"branch":"refs/heads/master"},"1cff883b71f534675b0fab7f1d11c1aca927417a":{"kind":"REWORK","_number":5,"created":"2020-08-11 19:53:35.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/73/742373/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/73/742373/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/73/742373/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/73/742373/5"}}},"commit":{"parents":[{"commit":"cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82","subject":"Merge \"Refresh Glance example configs for Victoria milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-03 20:00:24.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-08-11 19:43:12.000000000","tz":0},"subject":"PoC for lazy loading cinder store","message":"PoC for lazy loading cinder store\n\nDepends-On: https://review.opendev.org/742371\nDepends-On: https://review.opendev.org/743800\n\nChange-Id: I89fc8ef71ad58981238a4733687acd9bd0ae71d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1cff883b71f534675b0fab7f1d11c1aca927417a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1cff883b71f534675b0fab7f1d11c1aca927417a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
