)]}'
{"id":"openstack%2Fglance~742742","triplet_id":"openstack%2Fglance~stable%2Fussuri~I57d473572ca0a341082bacc3883cd9f763d77fa7","project":"openstack/glance","branch":"stable/ussuri","topic":"bug/1888349-stable/ussuri","hashtags":[],"change_id":"I57d473572ca0a341082bacc3883cd9f763d77fa7","subject":"Fix broken glance-cache-manage utility","status":"MERGED","created":"2020-07-23 20:35:29.000000000","updated":"2020-07-24 14:10:47.000000000","submitted":"2020-07-24 14:01:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742742-1595599282062-d04e827d","meta_rev_id":"a16477914d15bf9027f78e044449f62de2f33c34","_number":742742,"virtual_id_number":742742,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 14:01:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-23 20:41:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-24 12:30:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-24 12:30:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 20:41:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-24 12:30:04.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-24 14:01:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3466d8f3f7a71f783fb046acbaa2eaa2e202276e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-23 20:35:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3a4ce286e9569475501823f386ef2395ff793dfc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-23 20:41:45.000000000","message":"Patch Set 1: Code-Review+2\n\nGood to backport.","accounts_in_message":[],"_revision_number":1},{"id":"5e60c5c77100586732464fe2b4bbc0dd7fd9a778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 21:46:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3d2a9f63e85a41bcafb138c4b2b7f7b7 : SUCCESS in 1h 10m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85133d272a0a46f5a5c5257e7e97a6ac : SUCCESS in 1h 07m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35643898aa434225b78ffdfded06467c : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975be18bc57e4e089544c59062014308 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83bd594bc42f43ec889cf86784e593a4 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/632702742b874307a09eae54345ed3c1 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef03aed73214378a250000a85eba7f5 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f68a3bcddeb436094b85eb0c33a7afe : SUCCESS in 7m 17s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/130c56b76bb0460a85fa15d2bbdcedfe : SUCCESS in 7m 11s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f159fe450e274aba8524a7dba4cf756e : SUCCESS in 6m 13s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ff22db1475d64f65867ff983ee752f76 : SUCCESS in 44m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/471bc8735e6a498c95ba3e2148f65135 : SUCCESS in 49m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ad6e894e6b2d637ca72e0e70d69e0421a30b29c1","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-24 12:30:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86b20142fb2332b820ba95dd4975a9fa019f32d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 12:30:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c67724c6efc1866b7202a3f8e2b1bcbd2a4db5ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 14:01:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c9d47e210512479999f98dfd7d01787d : SUCCESS in 1h 22m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6ba452106804d9c978ad66bcbcabc79 : SUCCESS in 1h 28m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33d85fe5bf894c65b76206b858c5f213 : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbaf27daf82a4ad487b33c36163ddae5 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c4ac2bdff9b479c8b51a4a9b9273365 : SUCCESS in 7m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7f673061c51410498359169034e6787 : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21470061125346519fd195e926988065 : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/91b0e91ca8764e10a387c0b0a0eab0fc : SUCCESS in 13m 33s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8c73f61629fc4b6bb4e74f4b20d53b77 : SUCCESS in 12m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a3d72dd581145f1abbd08f6f04f0ae7 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":1},{"id":"06c07661aafe260c274786d5b37308fc60e12f8e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 14:01:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a8790112819c7d70aea609ceb6617ccae22463e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 14:02:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43a583477994d5b8ca27c2400fc944c : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1},{"id":"37ed6fa0764b094ecdd2a9b98124075db6eed90f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-24 14:10:47.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit a32a3f9cb594858cf643a5531edd2e98f98410b9","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1bb6a32bad7b84f43e54d1f6d7c68e062fcc673a","revisions":{"1bb6a32bad7b84f43e54d1f6d7c68e062fcc673a":{"kind":"REWORK","_number":1,"created":"2020-07-23 20:35:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/42/742742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/742742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/742742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/742742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/742742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/742742/1"}}},"commit":{"parents":[{"commit":"3a8fc940e271edfbd87369bb3920196c4dd5cf7e","subject":"Fix admin docs deploying under HTTPD","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a8fc940e271edfbd87369bb3920196c4dd5cf7e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-21 09:18:31.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-23 20:35:29.000000000","tz":0},"subject":"Fix broken glance-cache-manage utility","message":"Fix broken glance-cache-manage utility\n\nglance-cahce-manage is broken due to circilar imports.\nFixed the same by importing \u0027prefetcher\u0027 module right before\n\u0027Prefetcher\u0027 object is initialized.\n\nCloses-Bug: #1888349\nChange-Id: I57d473572ca0a341082bacc3883cd9f763d77fa7\n(cherry picked from commit ffd2f5e63f8b7aebc8986c05aea249c75a8a692b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1bb6a32bad7b84f43e54d1f6d7c68e062fcc673a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1bb6a32bad7b84f43e54d1f6d7c68e062fcc673a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
