)]}'
{"id":"openstack%2Fglance~742810","triplet_id":"openstack%2Fglance~master~If0fd74d9f2eecb21153493457c58d767f12ffdeb","project":"openstack/glance","branch":"master","topic":"docs/unowned-images","hashtags":[],"change_id":"If0fd74d9f2eecb21153493457c58d767f12ffdeb","subject":"[Doc] Policy support to copy unowned images","status":"MERGED","created":"2020-07-24 07:52:20.000000000","updated":"2020-07-28 18:03:08.000000000","submitted":"2020-07-28 18:01:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742810-1595959291089-fddd1e41","meta_rev_id":"9ad7b14c715d35bce0c02d64fb5a4e4cc70199f3","_number":742810,"virtual_id_number":742810,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-27 16:54:31.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 18:01:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-28 16:11:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2020-07-28 14:47:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-28 16:11:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 16:54:31.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-28 14:47:58.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-07-28 16:11:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-28 18:01:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b7aa37137d98f8bd9e788659a1ec047de6e67cd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-24 07:52:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d43c77d4d2cd105fe618ad6ed04412a1ca1cf54","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-24 07:52:47.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"bbb1abb774b3cd48a2bed78466980dd3bbd1b220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 08:48:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/287e378e5c6d416c8403273283b874e3 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bd6eb4dd20f4a029b71fae8b4a0f18f : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f856206cf24a5097547eb473dfecb7 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b51f1415be0d806415cf65a7243c09e4acf64f5f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-24 13:34:48.000000000","message":"Patch Set 1:\n\n(8 comments)\n\nThanks for writing this!","accounts_in_message":[],"_revision_number":1},{"id":"219923c4d60838bcbe0e345ad7d605f25edba29c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-24 15:07:45.000000000","message":"Patch Set 1:\n\n(8 comments)\n\nThank you for review Dan!","accounts_in_message":[],"_revision_number":1},{"id":"a41195c2913a5511cf4ef4411b337e4a9acf3b0d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-24 15:21:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89cf405b67995d25632cd8f27c8db39be3ed73b4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-24 15:21:54.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"6413746a1b3eca2d736b8aee837b83837fff3c41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 16:23:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a7a40dc1d36f4434b7024eb53320e929 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f338937f9cb4d9787d93e88ad493bc9 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/614415d183924a4493d0d6a533170e2f : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":2},{"id":"9150ca2c43ec1cf4428cf54af88770e4e561ee7f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-24 16:57:11.000000000","message":"Patch Set 2: Code-Review+1\n\nLooks good now, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"5f41b7c2c9071c615fd92ad615a93cffd2fc07fe","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-27 13:52:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nContent looks good.  I think you need to change the api-ref text, though, because api consumers don\u0027t have access to the policy file.","accounts_in_message":[],"_revision_number":2},{"id":"cdcc225a61a83f26d58dcb5c1a59bb2caa4d07b9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-27 14:01:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90e9783102a50d65607dd8b58b8ceb48fe74d645","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-27 16:54:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a67d58741ad088bcbe941438950028403a723c68","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-27 16:54:31.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"774994f2786310bf82384fe777040fbabfd1af15","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-27 16:54:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"de340e861a8020510c93fffa7b6d0d3a52143f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 18:24:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/66ad3720e87b467fb9b2d0367a4d6ec7 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9c01c3c70dd47ea86c6534bfb2fd52f : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5f617664a345a68f2d56f997f50a9c : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f9a55bef8176669a11bf23359707ccf824cbea84","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-28 14:47:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0b2bfac2d83b558732e622fd05c2ab099b9dec25","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-28 16:11:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nNinja-approving a docs change per glance team policy.","accounts_in_message":[],"_revision_number":3},{"id":"5066296ddb987685348c0acf50e10c7d89bc9adb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:11:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ab9ae884f6c197d6f2d3740cc65b05dc3f0dc928","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 18:01:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/525e0f04826949c093fea1ded2eb8ffa : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5909b8b31a994aa19e5a7f66b6c2bbaf : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58bebc00a554480c807cf43c1c32fb89 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":3},{"id":"765f345ae2bf43292301499e47a58eddedbf8b67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 18:01:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6f47d8cad2a25a926af32abce200c8984a807f55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 18:03:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/666920d466f94264800b7b25101a04d9 : SUCCESS in 58s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe5d75f7e34480e842a91608277d675 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"52eaa56e40a5649c6c8da1c4e5204331d07b63aa","revisions":{"9566d6acaf870ef6bf6c4ffab80c60d040886b17":{"kind":"REWORK","_number":1,"created":"2020-07-24 07:52:20.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/10/742810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/742810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/742810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/742810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/742810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/742810/1"}}},"commit":{"parents":[{"commit":"ffd2f5e63f8b7aebc8986c05aea249c75a8a692b","subject":"Fix broken glance-cache-manage utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffd2f5e63f8b7aebc8986c05aea249c75a8a692b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-24 07:46:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-24 07:46:53.000000000","tz":0},"subject":"[Doc] Policy support to copy unowned images","message":"[Doc] Policy support to copy unowned images\n\nChange-Id: If0fd74d9f2eecb21153493457c58d767f12ffdeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9566d6acaf870ef6bf6c4ffab80c60d040886b17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9566d6acaf870ef6bf6c4ffab80c60d040886b17"}]},"branch":"refs/heads/master"},"1d47a6800ac6e5fb011aa5356cbc75b62489af94":{"kind":"REWORK","_number":2,"created":"2020-07-24 15:21:34.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/10/742810/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/742810/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/742810/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/742810/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/742810/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/742810/2"}}},"commit":{"parents":[{"commit":"ffd2f5e63f8b7aebc8986c05aea249c75a8a692b","subject":"Fix broken glance-cache-manage utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ffd2f5e63f8b7aebc8986c05aea249c75a8a692b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-24 07:46:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-24 15:21:22.000000000","tz":0},"subject":"[Doc] Policy support to copy unowned images","message":"[Doc] Policy support to copy unowned images\n\nChange-Id: If0fd74d9f2eecb21153493457c58d767f12ffdeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d47a6800ac6e5fb011aa5356cbc75b62489af94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d47a6800ac6e5fb011aa5356cbc75b62489af94"}]},"branch":"refs/heads/master"},"52eaa56e40a5649c6c8da1c4e5204331d07b63aa":{"kind":"REWORK","_number":3,"created":"2020-07-27 16:54:10.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/10/742810/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/742810/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/742810/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/742810/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/742810/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/742810/3"}}},"commit":{"parents":[{"commit":"16a5431c665adadbfcb9909cf3a73334d796c19a","subject":"Make glance-api able to do async tasks in WSGI mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/16a5431c665adadbfcb9909cf3a73334d796c19a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-24 07:46:53.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-27 16:54:01.000000000","tz":0},"subject":"[Doc] Policy support to copy unowned images","message":"[Doc] Policy support to copy unowned images\n\nChange-Id: If0fd74d9f2eecb21153493457c58d767f12ffdeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/52eaa56e40a5649c6c8da1c4e5204331d07b63aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/52eaa56e40a5649c6c8da1c4e5204331d07b63aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
