)]}'
{"id":"openstack%2Fglance~742883","triplet_id":"openstack%2Fglance~master~I9b5e6cb162d5f6aa44330e783aa2a43d0e16a8cc","project":"openstack/glance","branch":"master","topic":"async-native-threads","hashtags":[],"change_id":"I9b5e6cb162d5f6aa44330e783aa2a43d0e16a8cc","subject":"DNM: Run tempest tests on WSGI stack","status":"ABANDONED","created":"2020-07-24 13:39:26.000000000","updated":"2020-08-10 18:42:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4101eb00329eb12ba20b76ab07987ccb08e1bce7","_number":742883,"virtual_id_number":742883,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-07-27 17:53:53.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-27 20:55:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 17:36:46.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-07-27 17:53:53.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-27 20:55:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44bf40c154ae80d4b74ee71e8a9c033cd0fdb629","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-24 13:39:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a55cf3c03bdf4de0244bd1d70e85423e8f9bb45","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-24 13:40:00.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"2601002676c741ba2e7100e7cf214b87d78c9bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 14:28:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ea0b874131940508024a5b59d16ff25 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad3f636526294724ab81878797c8b85b : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4ec8766fb3343e7b78ed08aa8ad20ee : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9483b383c8324f1590666129fb283e24 : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e9bade8bc0d4f83837b5040216df274 : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5fdd1979c0f94f5db879ef34fc7b09f7 : SUCCESS in 9m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45a709ba953e4da8b67128243b633ab3 : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8b15cf02e1e0df155fc333fbafceba56a3ece267","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-24 14:59:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"52a0376fc3b1bbe463157a108142206e0ff02569","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-24 15:00:22.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"5656652ff4419c5528045b554ff8c2d484472005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 17:45:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6d4c08d52a4549e4b50fea587b207ba9 : SUCCESS in 1h 10m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a29c5d9f2f77474683590202accbbc56 : SUCCESS in 1h 23m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a8ef4e00db94ade8118fa10c4888a43 : FAILURE in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afada5ea3aa64c419c3bfc86cf44f2df : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/234bc1c0306b4df9a95f67bf87e776ac : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/364c5c8eeb5e4b9e8f07a3494caf296a : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a01d97e24f7c4c1280a75f3fa65e1104 : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f7f71e50abb0469f8a0a31f6dbf71d6b : SUCCESS in 9m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2fc066dfd75e4347b4f90f9c798f338d : SUCCESS in 8m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/511d6fb0249d45ed92ac77425656f60d : SUCCESS in 7m 57s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/80f487029bbb4d23a0caefd2c9c9179d : SUCCESS in 1h 05m 54s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/a660a115406a49309f3d7f4aae590bff : SUCCESS in 1h 14m 44s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/e5c406666bf84d6a9ef20de3dbd961f3 : SUCCESS in 1h 15m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ef1c59726e54d70ac485fa26cf437ef : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":2},{"id":"6ca18f8753de2f32b7ccf538c7afffc88a930bb3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-24 17:49:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"6966d51aa7dfec7e7c968d639e0354d07781c72d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-24 17:51:11.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"9cedb0ab6e378eaa3529d54bee12c63c4d23cb1d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-24 17:58:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1d78012d88e427a7008c3bbc73b4ed4407342b2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 17:58:13.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"0cefcfb8b0b0d6fb5646a8c9865f81578013170d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-24 17:58:47.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"c454ac48d09e08392dd49becf33e797485d0a607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:04:37.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"6c87fa0e50fa1e76a07e796684976fd8660de09c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-24 18:06:46.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e26e1595da92f768f59153dc5b052291f311f942","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:07:13.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"97f1744f9b253c046ff6a65c8881dec7c3b5a2ff","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-24 18:07:22.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"a3852774a987a93670ea6820b8a945300e50d7a6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-24 18:13:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"387c40b67ca76580e341d98a822a115480f43381","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-24 18:13:40.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"53430776d7a6f1adb4a9f55b0a7536c0ba6044ac","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-24 18:13:53.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"1e7daaab65c36bfa9bd5d4ef777f5b85f3ac250c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-24 18:15:06.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"0617483051410877a57475f18288b7ac0f0e857c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 19:36:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/716941ed068f4104abcb3eeaaea168d5 : SUCCESS in 1h 10m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9bfc2af58d714a85956236fc59269f07 : SUCCESS in 1h 13m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/389e03a87720486a958b42ec4e747864 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e496526d527413387dbd792cdc97ffa : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c83298d6e52640f7b3b08ec1be85d0a3 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc7e502babc042cfbeae27677bd39588 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2743190cb6524da68c61decaf6742bdc : SUCCESS in 9m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe7f4bec8c01482bbfc46def017a8e61 : SUCCESS in 12m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a17ef624ae00419285a7d142155b9f48 : SUCCESS in 12m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a7a292005804480f99fd01e1a14ac600 : SUCCESS in 5m 33s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/314bb8eb57ee46a79cfd7a7fd7a738d2 : SUCCESS in 1h 13m 53s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/fa22e3e5245e4af4ae3bcd98005fcb5b : FAILURE in 1h 10m 12s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/d8dac4db62524344993ff44400b789a5 : SUCCESS in 1h 06m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f513c243be7345ef95425ac4365119bf : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":5},{"id":"6623916b4a72cddff9f4bcd8141223d1d357915b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-27 16:57:28.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"c9f4a45a06cf1a6362565066636264a09f6c76ad","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-27 16:58:20.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"32c673d31decb8698877a0b1b868be4794dd88de","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-27 17:36:46.000000000","message":"Patch Set 6:\n\nweb-download test are skipped in this as import method is not available  -https://zuul.opendev.org/t/openstack/build/d8dac4db62524344993ff44400b789a5/log/controller/logs/tempest_log.txt#57581\n\nis it disabled in glance by default now?","accounts_in_message":[],"_revision_number":6},{"id":"9ff081d0bf24edbb29b2141f08e6e14072bb3a5b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-27 17:53:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"142ecef5996b6e2775e4b67ea616d9e940fd773b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-27 17:53:53.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"b2cee82eebc2b8a92ba08ffbb00183618f034c7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 20:55:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/15c6716d8737405297ffe2e4597754f9 : SUCCESS in 42m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/78c03fb8f5724b95b284bcf41f5f2838 : SUCCESS in 1h 09m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98d3bd87791d465791db7bfebe8fee57 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d9c5503a8b4a1b9fdd7d594727bbb3 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4443592c8de945f4b119ccb0c36ceaa5 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a899c5b7e8b447d78acb1db6e0fa79a0 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8450f964fea462db557fb0a04490bd7 : SUCCESS in 11m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93730bcd98a040a0b7590291bbf5a83e : SUCCESS in 7m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/78a7b890c05343c39c519100348a5960 : SUCCESS in 15m 15s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3ba191dbe0da43d29710213e9a8d400f : SUCCESS in 5m 25s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bbebbfe4d5e04b0b9b5049c55e96dcf6 : SUCCESS in 1h 04m 32s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/b4aec3f38c06425cb4a4d90db1b172d6 : SUCCESS in 50m 14s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/cd04f04106be40b9a068c694ef3216c4 : SUCCESS in 1h 24m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a2bcf0bfd67422bb765e7e58490ea55 : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":7},{"id":"66193a79f04cf2e2a115e4beae3264c9a6daf3b7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-10 18:42:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"86cdf8bdcb7bd707159a93376a2e40728b5039ab","revisions":{"d7726b79e5f873c3a59df55265d0da625cc1c592":{"kind":"REWORK","_number":1,"created":"2020-07-24 13:39:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/83/742883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/742883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/742883/1"}}},"commit":{"parents":[{"commit":"a4b2cedc14c3bb5de11dfb07c25bcf4bd761d72b","subject":"Make wsgi_app support graceful shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a4b2cedc14c3bb5de11dfb07c25bcf4bd761d72b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 13:37:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 13:37:58.000000000","tz":-420},"subject":"DNM: Run tempest tests on WSGI stack","message":"DNM: Run tempest tests on WSGI stack\n\nDepends-On: https://review.opendev.org/#/c/742295\nChange-Id: I9b5e6cb162d5f6aa44330e783aa2a43d0e16a8cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7726b79e5f873c3a59df55265d0da625cc1c592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7726b79e5f873c3a59df55265d0da625cc1c592"}]},"branch":"refs/heads/master"},"8752ec794d71dfecb86ad788acd40a1c6c6de899":{"kind":"REWORK","_number":2,"created":"2020-07-24 14:59:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/83/742883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/742883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/742883/2"}}},"commit":{"parents":[{"commit":"a4b2cedc14c3bb5de11dfb07c25bcf4bd761d72b","subject":"Make wsgi_app support graceful shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a4b2cedc14c3bb5de11dfb07c25bcf4bd761d72b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 13:37:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 14:59:44.000000000","tz":-420},"subject":"DNM: Run tempest tests on WSGI stack","message":"DNM: Run tempest tests on WSGI stack\n\nDepends-On: https://review.opendev.org/#/c/742295\nChange-Id: I9b5e6cb162d5f6aa44330e783aa2a43d0e16a8cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8752ec794d71dfecb86ad788acd40a1c6c6de899"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8752ec794d71dfecb86ad788acd40a1c6c6de899"}]},"branch":"refs/heads/master"},"c6a61c62022ee033aa2297b0062f34219f451e18":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-24 17:49:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/83/742883/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/742883/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/742883/3"}}},"commit":{"parents":[{"commit":"a21f2912fe22ac37b98dfd8f685fecf07f5b3694","subject":"Make wsgi_app support graceful shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a21f2912fe22ac37b98dfd8f685fecf07f5b3694"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 13:37:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 17:49:46.000000000","tz":-420},"subject":"DNM: Run tempest tests on WSGI stack","message":"DNM: Run tempest tests on WSGI stack\n\nDepends-On: https://review.opendev.org/#/c/742295\nChange-Id: I9b5e6cb162d5f6aa44330e783aa2a43d0e16a8cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c6a61c62022ee033aa2297b0062f34219f451e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c6a61c62022ee033aa2297b0062f34219f451e18"}]},"branch":"refs/heads/master"},"9fc9203577db80615f8a96326066545930114790":{"kind":"REWORK","_number":4,"created":"2020-07-24 17:58:13.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/83/742883/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/742883/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/742883/4"}}},"commit":{"parents":[{"commit":"a21f2912fe22ac37b98dfd8f685fecf07f5b3694","subject":"Make wsgi_app support graceful shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a21f2912fe22ac37b98dfd8f685fecf07f5b3694"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 13:37:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 17:58:08.000000000","tz":-420},"subject":"DNM: Run tempest tests on WSGI stack","message":"DNM: Run tempest tests on WSGI stack\n\nDepends-On: https://review.opendev.org/#/c/742295\nChange-Id: I9b5e6cb162d5f6aa44330e783aa2a43d0e16a8cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9fc9203577db80615f8a96326066545930114790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9fc9203577db80615f8a96326066545930114790"}]},"branch":"refs/heads/master"},"59a9309b895e7a0101c617777b8a6531e137be76":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-07-24 18:13:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/83/742883/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/742883/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/742883/5"}}},"commit":{"parents":[{"commit":"41c6e79c42c08e5c04ae87cf6accb9ca4e371a75","subject":"Make wsgi_app support graceful shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41c6e79c42c08e5c04ae87cf6accb9ca4e371a75"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 13:37:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 18:13:45.000000000","tz":-420},"subject":"DNM: Run tempest tests on WSGI stack","message":"DNM: Run tempest tests on WSGI stack\n\nDepends-On: https://review.opendev.org/#/c/742295\nChange-Id: I9b5e6cb162d5f6aa44330e783aa2a43d0e16a8cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/59a9309b895e7a0101c617777b8a6531e137be76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/59a9309b895e7a0101c617777b8a6531e137be76"}]},"branch":"refs/heads/master"},"e8ec1069217cc94f97e786438418a154f03e7424":{"kind":"NO_CHANGE","_number":6,"created":"2020-07-27 16:57:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/83/742883/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/742883/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/742883/6"}}},"commit":{"parents":[{"commit":"f4b78606bafd0c118de67f6e1ccdad8e0c5f555d","subject":"Make wsgi_app support graceful shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4b78606bafd0c118de67f6e1ccdad8e0c5f555d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 13:37:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 16:57:22.000000000","tz":-420},"subject":"DNM: Run tempest tests on WSGI stack","message":"DNM: Run tempest tests on WSGI stack\n\nDepends-On: https://review.opendev.org/#/c/742295\nChange-Id: I9b5e6cb162d5f6aa44330e783aa2a43d0e16a8cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8ec1069217cc94f97e786438418a154f03e7424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8ec1069217cc94f97e786438418a154f03e7424"}]},"branch":"refs/heads/master"},"86cdf8bdcb7bd707159a93376a2e40728b5039ab":{"kind":"REWORK","_number":7,"created":"2020-07-27 17:53:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/83/742883/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/742883/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/742883/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/742883/7"}}},"commit":{"parents":[{"commit":"f4b78606bafd0c118de67f6e1ccdad8e0c5f555d","subject":"Make wsgi_app support graceful shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4b78606bafd0c118de67f6e1ccdad8e0c5f555d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-24 13:37:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:53:19.000000000","tz":-420},"subject":"DNM: Run tempest tests on WSGI stack","message":"DNM: Run tempest tests on WSGI stack\n\nDepends-On: https://review.opendev.org/#/c/742295\nChange-Id: I9b5e6cb162d5f6aa44330e783aa2a43d0e16a8cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/86cdf8bdcb7bd707159a93376a2e40728b5039ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/86cdf8bdcb7bd707159a93376a2e40728b5039ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
