)]}'
{"id":"openstack%2Fglance~743426","triplet_id":"openstack%2Fglance~master~I136ef792d601326a67266e9ea9fcadd79ddba69e","project":"openstack/glance","branch":"master","topic":"task-heartbeat","hashtags":[],"change_id":"I136ef792d601326a67266e9ea9fcadd79ddba69e","subject":"Heartbeat the actual work of the task","status":"MERGED","created":"2020-07-27 22:07:03.000000000","updated":"2020-08-11 16:36:44.000000000","submitted":"2020-08-11 16:32:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743426-1597163552029-ea02c699","meta_rev_id":"e51d1379513d7184ba751b3b734a66103301c655","_number":743426,"virtual_id_number":743426,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-11 13:30:33.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-08-11 13:30:11.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-11 16:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-10 19:12:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-08-10 17:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-10 19:12:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-08-10 17:23:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 00:26:56.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-08-10 17:23:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-10 19:12:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-11 13:30:33.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-11 16:32:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"483a5f0500d676e5a32a78bedb6b130136b6f943","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-27 22:07:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"640280b5f3313e5b1448bdf2d7ddf5b7df8dc6dd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-27 22:07:28.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"7ba7838e42bf81efdf10e2aa1cc69382f9d2209f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-27 22:12:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68065aeb0d7b4ccf9ff82fdc0838622da60e1a74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 23:42:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/370ed99ed39b4403a8c2698762446550 : SUCCESS in 43m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3525e08ddb734f9bb40c56028c204272 : SUCCESS in 1h 33m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72cd2fe4401844d68c34a711e1b6cbac : FAILURE in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5bca1a3e0d4309aa3422a07835ce8f : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2fb0071c24a48e888c848d92979b940 : FAILURE in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7da9ab623ce141069e7d455949ae05f4 : FAILURE in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302ac80278064abbbbdbbbf264a1db1a : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a213cf0f57bb4add897c9fb70fe07e0e : SUCCESS in 10m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/546cc739ddde4d508cb9610c54885998 : SUCCESS in 10m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b998465edcad483db2d70fb3e3bc9db7 : SUCCESS in 4m 27s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ee3c87c621d34d7db0f787cc99202ebc : SUCCESS in 1h 10m 27s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f78aeda1aef2469684be5aa665ca27c4 : SUCCESS in 49m 46s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/4cb37ef0d1bb4043a06195ebdbc53aed : SUCCESS in 1h 05m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2546f74ac3934329a877b1402ad375ad : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":1},{"id":"bd65fbfae345c3070cb683943a001c80d967c852","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-28 17:08:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"40463948b217a2ca6bb424cfcb7b787bfe2f8bfa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-28 17:08:57.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"f4831d2b0862d003e777fb42518dca5a3b639de8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-28 20:37:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a45635ca5176c99ba10ebb6dc0451f663182b267","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-28 20:38:17.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"6a032b0972c19038f198f6984c1e78fc1557dca2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-28 22:54:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8f331371d563f7f4d4644d5d3cc1b7c389f02057","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-28 22:55:16.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"3f17194416208f438363609ece54f9ba386b367e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 00:16:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/33d8838087834fbea1bbb6700f329925 : SUCCESS in 1h 09m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d41d5cfcd5cd4dd680d2efb0874eb4bc : FAILURE in 1h 12m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1148fd9b0ce84471bd2804205dd3ea5f : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32752a84963640ad80f63f6ffb2852cb : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46c7e3e3fde54790862fc9f3ffc08b6b : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4e33f777d334f63ae95dcf225de232f : SUCCESS in 14m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a09140c63b14ab3a414b8f4923c486a : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87acf04afa5f4bec8618a3802978f714 : TIMED_OUT in 30m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/073bcb80f9f24081805d7ea3fadea305 : SUCCESS in 7m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3fd57a3a952d4894bea3f3a7fdd2de89 : SUCCESS in 5m 12s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a6e85d0c516546ad910882b3d25734e4 : SUCCESS in 57m 24s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/66f7f28c01ff4b57983c2d11416088ac : SUCCESS in 1h 20m 03s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/f42e92ac7bbe4185926e39375e9ee6ea : SUCCESS in 1h 14m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c77ceb027db4d57b39b337ddb91422c : SUCCESS in 50m 58s","accounts_in_message":[],"_revision_number":4},{"id":"b712281a2daa22ae1e7f1fd33136b88eeaad839d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-29 00:32:44.000000000","message":"Patch Set 4:\n\nrecheck known cinder faily test","accounts_in_message":[],"_revision_number":4},{"id":"094f2e27b6bda2a564985f814392759a3f77f786","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-29 00:33:24.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"36dea774c32f70f200d735ea6671e13e03671faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 01:53:22.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0f88230fb8414a39af4cfb631c379509 : SUCCESS in 57m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7eb095dd89c4ff296175b44f5cef223 : SUCCESS in 1h 14m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98c4aa8021ed41c788c0f292feb39a9a : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b5abc3bca3a49a28750a9f55e8a7338 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0f41964831d4ac98a0c05d1f06a90ec : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a765656e8e374a909606cb7dcf808264 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee0c34e7a69248979afe4168528b1d90 : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bf018ac1fc604e66ba6fcd62b3f05ac8 : TIMED_OUT in 30m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5676ac19f0eb464f9af34cedd28fe95c : SUCCESS in 7m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cb4a3eb0de88485bb9de248867162f35 : SUCCESS in 3m 40s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b17e78d5ea7c4f8abffc96f7f07b272c : SUCCESS in 56m 16s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/8e8b191211504d1eb29776ff2769d919 : SUCCESS in 1h 01m 40s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/7ab7b5a3faf54fbe880cc08c34db44c2 : SUCCESS in 1h 06m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3096d6ab1b264d8e9e0fd57d24e0cfdf : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":4},{"id":"0a7e078c3ae5db2ffa9a8f0f1eb1a7d05ef8bdd0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-29 14:25:40.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"057510be3564ba38e53335fb627e5181a3e7028f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-29 15:06:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df970f881607906773b49fa65d57d8b901b36ec8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-29 17:07:28.000000000","message":"Patch Set 4: Review-Priority+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9ed6531951cf144737c049847e51cc7869850f84","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-29 17:08:11.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"ad16b0b5136a9f45cfe0c689d26dbe056c59798a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-29 20:09:44.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"7784fbd9c5d4a10fbc2546edf1db62165d08c66b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-29 20:11:01.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"602debeadd8239a0ce75915b9a69fd6f8af6940f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 00:59:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/50e7e739646c4ecb88160ef465957bd0 : SUCCESS in 1h 02m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/165fbe4954144c70a0420220bac1f20d : SUCCESS in 1h 12m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30d8d1198df94228b8b910da0d8ff755 : FAILURE in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9004e5481b464e449d86dd93846b4ec0 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf2cfe4bbda644918d174762f83052ef : FAILURE in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32ae54d518954d9ebf7c8ac8704e895d : FAILURE in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24640ff1661643e3bd8b03770644da0e : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5c2c12352e3b42dda90fe99526a72748 : SUCCESS in 7m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eb840b3ddb384329b7785b8873dcbe05 : SUCCESS in 8m 07s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2380140e6d1f4a20ae49aae6efd78750 : FAILURE in 5m 20s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/06391fb9ba6f4fb6b49fbcdbefd8889b : SUCCESS in 1h 15m 15s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/33229fc59cd8419981aeec0891bd63ec : SUCCESS in 48m 22s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/1bc4810943d8486da6d5a138ef14973b : SUCCESS in 1h 02m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0282e28249f46bcbfb5e55d94dbc313 : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":5},{"id":"a7e9c18180e3922ca8b6266a625ea82d799399a1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-30 13:41:40.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"146e21b4c42b43f5cf2efae156277328cef000bf","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-30 13:43:17.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"29c891e2007f6dbf934af434b67e4dbb3cc04d2d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-30 13:44:31.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"a7dfe08a85215b543e42f8c6d08dd23f9ab0356d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-30 13:46:48.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"7be9d10826c1726931611511916c268680fa00ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 19:58:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bb4653e6223f4c8fa732438f90a4cc20 : SUCCESS in 1h 00m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7461451589ed4c54802f7cec55ae2727 : SUCCESS in 1h 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2db0a3bfc0b4030abee44cc8013ac56 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3b5a64ba13241f2bdf15cb2018ced90 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da364d8663de4c2da32c2be4ad22726b : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3eead52681134dcc88fab4de6534dcfc : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6493738be6b438e8507b458fc451fe9 : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/705349b7d75049a181278b93ac8d6324 : SUCCESS in 7m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/81e4fdb641b54c9ebf1eda99cbbf516b : SUCCESS in 11m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2ed4fac82e414d8e94ad13ed85047adf : FAILURE in 4m 43s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a6979a7999fc49dcb19e26c5df3084f5 : SUCCESS in 59m 41s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/677eed1d8fd843c49a68043a864b37fb : SUCCESS in 1h 14m 14s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/cf8a77c98049412f921400b52e845c39 : SUCCESS in 1h 18m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26a5fbb9f61f43c0b82be8489e0e3eaf : SUCCESS in 1h 06m 33s","accounts_in_message":[],"_revision_number":7},{"id":"0fca8388d6c4f778e524541b3533a3a557dd53ee","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-31 00:26:56.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"328575f8eff50790f76dd97e83dac3abffe7f7fc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-31 00:28:57.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"c6e398ada5dcdf49e8cb516bc6809af427b648f3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-31 00:35:39.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"19d6e1a009ac83dd5766acdfbed3d1a78b713a8e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-31 00:38:07.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":8},{"id":"83f769cc655ccb152beb01f1ce91ba827f3e3be4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 03:14:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e1e6c2d0c6874c5b8cbc915b42a37954 : SUCCESS in 1h 06m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3b76b098568a469ea50bc8965ddf4594 : SUCCESS in 1h 08m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f9c60b25dd6437695b3370a9fe77d07 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0a648889c414522b5a23d7e78adb395 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb3f36cad96c422c961ce7e24abd7e6d : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64db1bad6dc44218a887599d300c8df1 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3243e90eae52488487b97585c9254122 : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b28010cc331b408a82010a659ec47580 : SUCCESS in 9m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/85ca30d80f5c46c48c854c30a8b945a6 : SUCCESS in 8m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/856e6654855a457bac4b8b7656e623ba : SUCCESS in 4m 28s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ff7e698a85e54ccabb8eac5041e03fdd : SUCCESS in 1h 29m 07s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/6dde9ce019d947acb2e05e064d421459 : SUCCESS in 1h 19m 58s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/5bfbc6bae99d4426b614a599dc951415 : SUCCESS in 1h 14m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d30cef8c37d41378975a9f1321ab919 : SUCCESS in 1h 06m 28s","accounts_in_message":[],"_revision_number":8},{"id":"0dc22d6a6346bacc69ccc595a1a6cc7e1ba325ac","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-04 16:59:02.000000000","message":"Patch Set 8:\n\n(1 comment)\n\njust a suggestion","accounts_in_message":[],"_revision_number":8},{"id":"4d3ab5d2a703feec349a7b3cb8d534b84643db83","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-04 17:32:39.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"95507221a057e03274d6d1b08b8629ec2429e024","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-06 17:33:38.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"49719d44d181794b6e23fd3e1dfdc770cf6c9f37","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-06 17:35:23.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":9},{"id":"272e67ec0a67f93d7cdedfdd8d2a21b0e948c4b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 19:02:07.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/85e296ef38df42369ed9428a40550f37 : SUCCESS in 52m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cec9b93d46d34127bb725003f34b4d54 : SUCCESS in 1h 17m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea171e8e1aee474d8f8826a824395976 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fe35b67eb80472b8993ecf22913bf1e : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e32e7cd8c27c4325b1b89486aa9f4329 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/224a5ac281f247fd8f647c67a717e43a : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff19449caeee472f979ee06441711a05 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c44a77e381d8464dafdc6f3de59a9088 : SUCCESS in 7m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e9981facf49245f2a0ba6306bcb25e12 : SUCCESS in 7m 49s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5465fe7b807f443da799075a11a5eefe : SUCCESS in 6m 04s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/49fc88e2b7854e458ed6f24f760bc1a4 : SUCCESS in 1h 00m 31s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/b02679aa2df94b298a9e8602e2f595df : SUCCESS in 1h 21m 09s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/9b0ba6e5fae848448edeb042690f61f8 : SUCCESS in 1h 17m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/732daa40f29c47739e7685c6671b627e : SUCCESS in 1h 02m 31s","accounts_in_message":[],"_revision_number":9},{"id":"43ce14f91f70d51eaa463e51850dda473fb9478a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-07 06:54:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5cf238c651c003f4a5519e591eec56729ea2cefd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 13:35:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"cca473069c87a41089643c79e371624f5780f04e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 16:31:03.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"33e5b196106a2b4bedcca1aff88013aacbc85dc5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 16:31:47.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":10},{"id":"0ce0c956125b7faf3deec067e5e5cbe1851cafbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 17:53:08.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4d9fe28d47ad4674bc7e39080ed60b2e : SUCCESS in 59m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fd7cd29ba8d43cdaf63c61f8ceb6966 : SUCCESS in 1h 04m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7a509e26f7c49a2b274ad7dc7661c96 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ef1a9a71f444a1787eaf867b3afc388 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f2503f01cd549728dfa4cfb193ac0b2 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0b74db8fa654b65a2c226932debf638 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5543c8ff03334560a83b461a5ecf4d12 : SUCCESS in 8m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/98f082dc11864b47afd8d5d9f5223d42 : SUCCESS in 10m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d32ab27c6265484a81a8a5b4d3439f4b : SUCCESS in 12m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/864c8a2df4074e8793571bf7d621178e : SUCCESS in 8m 44s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6508384a0d2342179f8d086e617836ac : SUCCESS in 1h 05m 05s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/003f65df66734cb0898cc86e70abc105 : SUCCESS in 1h 05m 43s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/ec555b5f0b874b2391085af53a0a26b6 : SUCCESS in 1h 12m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba56bab7e1a74cc4882ce7f71b13a332 : SUCCESS in 1h 09m 13s","accounts_in_message":[],"_revision_number":10},{"id":"94eafdd6a1c96e5a5ceede8731821110a1538d18","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 19:31:31.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c4907337a45fe84cda02a26350a1ff64c0b2cb64","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 19:36:08.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"ca6fb2ca3afdc0dd70c0f6134369ba4fa2dc95a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 21:10:46.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bd916747a3634e8d931c9dd37c0e51d4 : SUCCESS in 1h 02m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de2faf9de3b6466cb9e0a286a90ac885 : SUCCESS in 1h 06m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e1198fdc3144a84a8e401f5f1fadb70 : RETRY_LIMIT in 0s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/294e93b6e62f4af9a326d5b82319fa17 : RETRY_LIMIT in 0s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d5b0eee4cb2485d88ef2f7e501ce2cc : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a159798376049728a5b52f220da49e6 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff77c2533647400984f3bad4df081612 : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f1988bf32d554e0b960fcdca2639ea30 : FAILURE in 16m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a5bc0c70d2654c6b9c8fa5cd22ddec3e : SUCCESS in 8m 31s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b537f1ecd494448fb3d75e2352022f2e : RETRY_LIMIT in 0s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d660a556ba2241d7ad25de6ceb5f2f05 : SUCCESS in 1h 08m 23s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/dd4ef43036d84f4487b8a11a43a89ea2 : RETRY_LIMIT in 0s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/7d01c9c369f14ac0897259a2291890df : FAILURE in 1h 32m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d17aa0b4608462392468b3bc1dcff6b : RETRY_LIMIT in 0s","accounts_in_message":[],"_revision_number":11},{"id":"ea3d7dfe902fd088a097ac21aad4575eae83325b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-10 14:14:45.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"7d418e419f3ad67114e9b6fff47597628a0ae569","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-10 14:50:43.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"87006c67c2dc3b2ce9dd41c8fad24faaa45ff277","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-10 17:23:26.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+1\n\n(1 comment)\n\nJust one suggestion inline","accounts_in_message":[],"_revision_number":11},{"id":"bd494371a388b2a587e6e8180e3d663ecee284d0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-10 17:36:09.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6a3e6b92897e810db086803b1aeba85eb405f1ca","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-10 17:54:51.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nIf new PS is required otherwise let it be.","accounts_in_message":[],"_revision_number":11},{"id":"58c9040c2d769b32f7e1881768c559caff4ef23e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 17:56:45.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3f17c3f8c6da4bedb08b05b8c6233792 : SUCCESS in 1h 07m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b8d385c79bb41b8bb8473770715ec41 : SUCCESS in 1h 15m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3652f376e2b84f23bfb9520f60f66aa3 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e6aab6d735f4042a5893706c8064372 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98cbc0d5d9954d39afa130a9a5eb3688 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64ae659ebae14b2b855a701cf94e7e4e : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a9df99f9e24e44b271ba1a531b4a19 : SUCCESS in 11m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/84d675cde58b430da115507c1c58f134 : SUCCESS in 12m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ed5f5382fc034941bc572b1ac5652b82 : SUCCESS in 9m 11s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4e422ac0344b4f26b97dc8099f33d0f6 : SUCCESS in 4m 25s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c7e1cb41ac424d7b9312bcb48d03c800 : SUCCESS in 56m 04s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f22e5f42fc69434f8743053deaaa940a : SUCCESS in 1h 34m 05s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/d7ac659a9fde4f81b9e97cc4a4460a5d : SUCCESS in 1h 24m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff5766a4dd004a40bbb87088788a77c3 : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":11},{"id":"e1f1490f1552622cb6547701d8760d94fe06d944","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-10 19:12:43.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7e5b3b64482aa8e2ab1808584f2281d770007725","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 19:12:58.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"fa089d37cd0a04b1323297fae59abd3dd54233af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 21:08:06.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"ac50a2f288a9cf755f62bddffaa4a5ae48abdd78","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-10 22:00:47.000000000","message":"Patch Set 11:\n\nrecheck dep failed on cinder","accounts_in_message":[],"_revision_number":11},{"id":"ddbb002a697334bba4beccdafc3aa991026278ac","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-10 22:01:10.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"53f1f8ded53cfc888607f88447a6e4096bbb970c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 23:27:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/31b9ae3439534ea1be28fb51f19f4953 : SUCCESS in 1h 13m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c951e8a7366c4b059341bc0a132d5ee1 : SUCCESS in 1h 23m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc51e77b8989497097167010ea083f89 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc34c2f993cc4fa284f72654424f2f37 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4ccb6238c0f4510a0f477c39c35e3ca : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64ed8b80a4f44b41930bdd36690914b4 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a133e6624aa74c8ca7e9297f2f44473c : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a673d7e6d06c4b51a180e1f23355ef6b : SUCCESS in 7m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8c2f7dbe74784c019321b21423ebf5ea : SUCCESS in 7m 04s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ae10b906422945f4a44e3af707c3da25 : SUCCESS in 4m 55s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fa82a855a60143769bfcce069b1cbe06 : SUCCESS in 1h 20m 26s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f0ef90a38afe4d789f1b96c042af7d3e : SUCCESS in 1h 16m 56s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/a67df0d208a94d4d95bef92e8ee30006 : SUCCESS in 1h 09m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1b17f7b965f4564bc3a5530a7fd910f : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":11},{"id":"d68ef9557e42ff0f93b9489334786a3301d01f49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:03:00.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"223e6b07f14df8e89dde6ec795319a80cb469c3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 01:27:26.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"be013c1ba8cc44479e1622bad5a73a76137ab1cf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-11 06:02:41.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"4545190bb7867423374fb696480a50ece8970492","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-11 06:03:04.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"31ed82aed6801a9bc4400880d114af14d39b758a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 09:34:32.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7b28905ab3ea4206aff0bfac0d41f343 : SUCCESS in 1h 02m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be634d16364942c5b3fd29bc7110012e : SUCCESS in 1h 08m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bc3b11c0bd8472887557b8968450544 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ca680340aa241018c99791ab5fcd056 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af7d4a3d974f4e43979254a9649b5081 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/673f1c1726c44b7d8c9950857644aadd : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d554ddce1d84b2cb72e41f7aed4bafc : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5caedb622cbe415aa76bc4b14685e8e2 : SUCCESS in 8m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a416ba2db5f4c919f071dc96aa25c70 : SUCCESS in 8m 37s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e805e7003a404b79a028109fa4d72fd4 : SUCCESS in 5m 58s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2d706291c79c4da6a2c7a16a9eefc656 : TIMED_OUT in 2h 10m 22s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/067b162bca484d8cacb3066970adf24d : SUCCESS in 1h 09m 29s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/31136c4f9d86463dba2219947ee8d35a : SUCCESS in 1h 06m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e9efbc0997647f2b3579bca9689e7f9 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":11},{"id":"54175451e8206d9e3fcfa6bc11a18a87f55c6c6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 11:56:50.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"b943e8c8b7afc0d1e998d0107e217ba68c347b9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 13:26:01.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a8ec2d008c744a3dac7dddbf7e0ca0e8 : SUCCESS in 1h 01m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f6723ea54874d05bdffbc30168cd4ad : SUCCESS in 1h 10m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8f647887cd9434ab3bc4a0b4ff3b21f : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc6a49123d5f42fea239f9eb02dc888a : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20332300c939411fab4df4140e3c2d6e : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1063567455b34498a39d9162e311473f : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88cfd427b7104ad7a1c42cc187cf0d0a : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec38d9729e6d4c5ba996dd9eda4fa472 : TIMED_OUT in 31m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0306f0750ab44f7abb8632e25a416dcb : TIMED_OUT in 30m 52s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/584000b374bb45d2b26047c2f977aa4c : SUCCESS in 1h 22m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0c57994748948b1af5ade96a018600e : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":11},{"id":"d38f9d46659cd78d532b2846925a71c26b1e181b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-11 13:30:11.000000000","message":"Patch Set 11:\n\nrecheck known functional timeout issue","accounts_in_message":[],"_revision_number":11},{"id":"293dacba944bfb005ca56f37d66aa4623645cb90","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-11 13:30:33.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"0b60b20cc342d4df86ddfdb111da5b539af563a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 15:08:08.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9536e4bd03324c1983ff6dad26bd9aca : SUCCESS in 1h 01m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7087d28201804f83ad3a49639cd7cfda : SUCCESS in 1h 30m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d192dc3b4b5e4707a1e44cc16feae629 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c999d8cdbf46423facf657208eb20a9b : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ad55eb01df94406bf9be22b9184166e : SUCCESS in 8m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/396c965ed7744b4283064cfd136149f0 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046df0acd69e484c9c46181774c35058 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72380183f84e4767b25f2a46b45d91db : SUCCESS in 8m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/07c600ad73cb4ace97c15287f2ab8506 : SUCCESS in 7m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4eaf4c6cb1e54a8f84c8425efc62d25a : SUCCESS in 7m 02s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b6f786d8161c4630aa6c67cb8f476eae : SUCCESS in 59m 13s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/86a9bca1d9854ddea2ac094d5cfbd1bf : SUCCESS in 1h 11m 19s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/ec4de45d88774197903b87be15b99518 : SUCCESS in 1h 03m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55c297b802a04a4c8f8636c586daddbb : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":11},{"id":"2af703eb8171f69f4510ecdf9b31e27a021d4c2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 15:08:22.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"26eec4865d87c7f12a3efcbc8c64ec780d515abb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 16:32:31.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/550155d0707a4e56a7e55a5dbdefa08a : SUCCESS in 1h 04m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f91c044b7c141f19d434d76a4f210eb : SUCCESS in 1h 13m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69af7014eac94fb88e8b3cd437322623 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e635aa4fcc1a488bb391e0db87e1547c : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4459ae4d5f874efea3efef1c62e5ae70 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3f567a727a94189b9bdf39eb1e15996 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1357f1706a14af2a48c426a7f30c6b9 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3fd4f6621d2427a8eeb31ac9df076bc : SUCCESS in 8m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e796e2f10fad4f459c33f4917cdf414a : SUCCESS in 8m 45s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/98f1314dbabf4e2b95b97481398736e9 : SUCCESS in 1h 09m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f36ed5afe3e14d0e9c0ad3d636d0092a : SUCCESS in 1h 08m 49s","accounts_in_message":[],"_revision_number":11},{"id":"3cc47c58f3f36c06a2245c6c6eb0d74282a187e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 16:32:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"185a449c38bf92b928a76404bf23d5a3537ef169","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 16:36:44.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb22233816ee4d418674b2ca3a06be9d : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"e49f23c04d784b5312d7df4556e2e5064555a84f","revisions":{"122d2866d037d1dfb5f5ef0456e9b17bed787929":{"kind":"REWORK","_number":1,"created":"2020-07-27 22:07:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/743426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/743426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/743426/1"}}},"commit":{"parents":[{"commit":"e8ec1069217cc94f97e786438418a154f03e7424","subject":"DNM: Run tempest tests on WSGI stack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8ec1069217cc94f97e786438418a154f03e7424"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 21:03:31.000000000","tz":-420},"subject":"WIP: Heartbeat the actual work of the task","message":"WIP: Heartbeat the actual work of the task\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/122d2866d037d1dfb5f5ef0456e9b17bed787929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/122d2866d037d1dfb5f5ef0456e9b17bed787929"}]},"branch":"refs/heads/master"},"5d9d726585e024c39e90d6bcc778ac1f42c5854a":{"kind":"REWORK","_number":2,"created":"2020-07-28 17:08:12.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/743426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/743426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/743426/2"}}},"commit":{"parents":[{"commit":"f2cece8fe2a786018bc6dbbb1b824b4f1c9cb9e5","subject":"Add image_delete_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2cece8fe2a786018bc6dbbb1b824b4f1c9cb9e5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 16:53:35.000000000","tz":-420},"subject":"WIP: Heartbeat the actual work of the task","message":"WIP: Heartbeat the actual work of the task\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d9d726585e024c39e90d6bcc778ac1f42c5854a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d9d726585e024c39e90d6bcc778ac1f42c5854a"}]},"branch":"refs/heads/master"},"df951d07ab8fdf54c1e79a4a39d5157618402e15":{"kind":"REWORK","_number":3,"created":"2020-07-28 20:37:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/743426/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/743426/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/743426/3"}}},"commit":{"parents":[{"commit":"f2cece8fe2a786018bc6dbbb1b824b4f1c9cb9e5","subject":"Add image_delete_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2cece8fe2a786018bc6dbbb1b824b4f1c9cb9e5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 19:55:08.000000000","tz":-420},"subject":"Heartbeat the actual work of the task","message":"Heartbeat the actual work of the task\n\nThis introduces a CallbackIterator object that can sit in the data\npipeline when setting image data and provide periodic callbacks to\nallow recording status, checking for cancellation, etc. If the\ncallback raises an exception, that will bubble up to the read\noperation and cause an abort.\n\nThe heartbeat is timer-based, but cooperative, which means if the\ndata pipeline stalls, we will stop heartbeating. The goal is to\nprovide a callback about once per minute, although blocking on\nthe pipeline could make that less regular of course.\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df951d07ab8fdf54c1e79a4a39d5157618402e15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df951d07ab8fdf54c1e79a4a39d5157618402e15"}]},"branch":"refs/heads/master"},"f1a938f930a9cffe9809f5236189e7d861354fb4":{"kind":"REWORK","_number":4,"created":"2020-07-28 22:54:32.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/743426/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/743426/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/743426/4"}}},"commit":{"parents":[{"commit":"f2cece8fe2a786018bc6dbbb1b824b4f1c9cb9e5","subject":"Add image_delete_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2cece8fe2a786018bc6dbbb1b824b4f1c9cb9e5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 22:54:27.000000000","tz":-420},"subject":"Heartbeat the actual work of the task","message":"Heartbeat the actual work of the task\n\nThis introduces a CallbackIterator object that can sit in the data\npipeline when setting image data and provide periodic callbacks to\nallow recording status, checking for cancellation, etc. If the\ncallback raises an exception, that will bubble up to the read\noperation and cause an abort.\n\nThe heartbeat is timer-based, but cooperative, which means if the\ndata pipeline stalls, we will stop heartbeating. The goal is to\nprovide a callback about once per minute, although blocking on\nthe pipeline could make that less regular of course.\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1a938f930a9cffe9809f5236189e7d861354fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1a938f930a9cffe9809f5236189e7d861354fb4"}]},"branch":"refs/heads/master"},"331ee31ce405145c563ee064eeaa467a50c8a608":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-07-29 20:09:44.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/743426/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/743426/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/743426/5"}}},"commit":{"parents":[{"commit":"2dd5a3d5af6666e940d8e828963b485bb7131fc2","subject":"Add image_delete_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2dd5a3d5af6666e940d8e828963b485bb7131fc2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-29 15:16:46.000000000","tz":-420},"subject":"Heartbeat the actual work of the task","message":"Heartbeat the actual work of the task\n\nThis introduces a CallbackIterator object that can sit in the data\npipeline when setting image data and provide periodic callbacks to\nallow recording status, checking for cancellation, etc. If the\ncallback raises an exception, that will bubble up to the read\noperation and cause an abort.\n\nThe heartbeat is timer-based, but cooperative, which means if the\ndata pipeline stalls, we will stop heartbeating. The goal is to\nprovide a callback about once per minute, although blocking on\nthe pipeline could make that less regular of course.\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/331ee31ce405145c563ee064eeaa467a50c8a608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/331ee31ce405145c563ee064eeaa467a50c8a608"}]},"branch":"refs/heads/master"},"a5996cdaadf2fb80b23027360c5aa2b10ed49689":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-07-30 13:41:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/743426/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/743426/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/743426/6"}}},"commit":{"parents":[{"commit":"a0d2b1bacd69540d4635323625bd1a0915172bf1","subject":"Add image_delete_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a0d2b1bacd69540d4635323625bd1a0915172bf1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-30 13:40:55.000000000","tz":-420},"subject":"Heartbeat the actual work of the task","message":"Heartbeat the actual work of the task\n\nThis introduces a CallbackIterator object that can sit in the data\npipeline when setting image data and provide periodic callbacks to\nallow recording status, checking for cancellation, etc. If the\ncallback raises an exception, that will bubble up to the read\noperation and cause an abort.\n\nThe heartbeat is timer-based, but cooperative, which means if the\ndata pipeline stalls, we will stop heartbeating. The goal is to\nprovide a callback about once per minute, although blocking on\nthe pipeline could make that less regular of course.\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a5996cdaadf2fb80b23027360c5aa2b10ed49689"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a5996cdaadf2fb80b23027360c5aa2b10ed49689"}]},"branch":"refs/heads/master"},"ddc1ab753f111602ae4768204948c9ec87a70f12":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-07-30 13:44:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/743426/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/743426/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/743426/7"}}},"commit":{"parents":[{"commit":"c0c2d5f2e8a3720e25c91a93adc41a8c7d51ae78","subject":"Add image_delete_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c0c2d5f2e8a3720e25c91a93adc41a8c7d51ae78"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-30 13:44:24.000000000","tz":-420},"subject":"Heartbeat the actual work of the task","message":"Heartbeat the actual work of the task\n\nThis introduces a CallbackIterator object that can sit in the data\npipeline when setting image data and provide periodic callbacks to\nallow recording status, checking for cancellation, etc. If the\ncallback raises an exception, that will bubble up to the read\noperation and cause an abort.\n\nThe heartbeat is timer-based, but cooperative, which means if the\ndata pipeline stalls, we will stop heartbeating. The goal is to\nprovide a callback about once per minute, although blocking on\nthe pipeline could make that less regular of course.\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ddc1ab753f111602ae4768204948c9ec87a70f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ddc1ab753f111602ae4768204948c9ec87a70f12"}]},"branch":"refs/heads/master"},"7def80ddbc04054ee242629b383b3b6a7ced32f9":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-07-31 00:35:39.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/743426/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/743426/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/743426/8"}}},"commit":{"parents":[{"commit":"3418c235b332d032229c481ccf1a1e186ef3ecc4","subject":"Add image_delete_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3418c235b332d032229c481ccf1a1e186ef3ecc4"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-31 00:35:33.000000000","tz":-420},"subject":"Heartbeat the actual work of the task","message":"Heartbeat the actual work of the task\n\nThis introduces a CallbackIterator object that can sit in the data\npipeline when setting image data and provide periodic callbacks to\nallow recording status, checking for cancellation, etc. If the\ncallback raises an exception, that will bubble up to the read\noperation and cause an abort.\n\nThe heartbeat is timer-based, but cooperative, which means if the\ndata pipeline stalls, we will stop heartbeating. The goal is to\nprovide a callback about once per minute, although blocking on\nthe pipeline could make that less regular of course.\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7def80ddbc04054ee242629b383b3b6a7ced32f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7def80ddbc04054ee242629b383b3b6a7ced32f9"}]},"branch":"refs/heads/master"},"065a4603cad1926768dbade9ba6e65024f2c0156":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-08-06 17:33:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/743426/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/743426/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/743426/9"}}},"commit":{"parents":[{"commit":"02f905aecfd6adef47d5b63c7860c2aecd6d478c","subject":"Add image_delete_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02f905aecfd6adef47d5b63c7860c2aecd6d478c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-06 17:33:25.000000000","tz":-420},"subject":"Heartbeat the actual work of the task","message":"Heartbeat the actual work of the task\n\nThis introduces a CallbackIterator object that can sit in the data\npipeline when setting image data and provide periodic callbacks to\nallow recording status, checking for cancellation, etc. If the\ncallback raises an exception, that will bubble up to the read\noperation and cause an abort.\n\nThe heartbeat is timer-based, but cooperative, which means if the\ndata pipeline stalls, we will stop heartbeating. The goal is to\nprovide a callback about once per minute, although blocking on\nthe pipeline could make that less regular of course.\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/065a4603cad1926768dbade9ba6e65024f2c0156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/065a4603cad1926768dbade9ba6e65024f2c0156"}]},"branch":"refs/heads/master"},"5c29ba45ddfd9e947aea9242e90062b6f2b967fb":{"kind":"REWORK","_number":10,"created":"2020-08-07 16:31:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/743426/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/743426/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/743426/10"}}},"commit":{"parents":[{"commit":"02f905aecfd6adef47d5b63c7860c2aecd6d478c","subject":"Add image_delete_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02f905aecfd6adef47d5b63c7860c2aecd6d478c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 16:23:36.000000000","tz":-420},"subject":"Heartbeat the actual work of the task","message":"Heartbeat the actual work of the task\n\nThis introduces a CallbackIterator object that can sit in the data\npipeline when setting image data and provide periodic callbacks to\nallow recording status, checking for cancellation, etc. If the\ncallback raises an exception, that will bubble up to the read\noperation and cause an abort.\n\nThe heartbeat is timer-based, but cooperative, which means if the\ndata pipeline stalls, we will stop heartbeating. The goal is to\nprovide a callback about once per minute, although blocking on\nthe pipeline could make that less regular of course.\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c29ba45ddfd9e947aea9242e90062b6f2b967fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c29ba45ddfd9e947aea9242e90062b6f2b967fb"}]},"branch":"refs/heads/master"},"e49f23c04d784b5312d7df4556e2e5064555a84f":{"kind":"REWORK","_number":11,"created":"2020-08-07 19:31:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/743426/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/743426/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/743426/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/743426/11"}}},"commit":{"parents":[{"commit":"b093ae3514dea32a8829239e0297e3e170efff58","subject":"Add image_delete_property_atomic() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b093ae3514dea32a8829239e0297e3e170efff58"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 17:12:40.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 19:22:15.000000000","tz":-420},"subject":"Heartbeat the actual work of the task","message":"Heartbeat the actual work of the task\n\nThis introduces a CallbackIterator object that can sit in the data\npipeline when setting image data and provide periodic callbacks to\nallow recording status, checking for cancellation, etc. If the\ncallback raises an exception, that will bubble up to the read\noperation and cause an abort.\n\nThe heartbeat is timer-based, but cooperative, which means if the\ndata pipeline stalls, we will stop heartbeating. The goal is to\nprovide a callback about once per minute, although blocking on\nthe pipeline could make that less regular of course.\n\nChange-Id: I136ef792d601326a67266e9ea9fcadd79ddba69e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e49f23c04d784b5312d7df4556e2e5064555a84f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e49f23c04d784b5312d7df4556e2e5064555a84f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
