)]}'
{"id":"openstack%2Fglance~743427","triplet_id":"openstack%2Fglance~master~I8667c17813f6e701db98595b0b30df9e7b275294","project":"openstack/glance","branch":"master","topic":"task-heartbeat","hashtags":[],"change_id":"I8667c17813f6e701db98595b0b30df9e7b275294","subject":"Update task message during import","status":"MERGED","created":"2020-07-27 22:07:03.000000000","updated":"2020-08-11 16:38:42.000000000","submitted":"2020-08-11 16:34:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743427-1597163666711-76a6aaaf","meta_rev_id":"cbfcd9e08f648fcd596de0ef112e1626665715d6","_number":743427,"virtual_id_number":743427,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-11 13:30:48.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-08-11 13:30:24.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-11 16:34:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-10 19:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-08-10 17:26:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-10 19:20:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-08-10 17:26:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 22:08:05.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-31 00:26:37.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-08-10 17:26:09.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-10 19:20:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-11 13:30:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-11 16:34:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d2b34fbf5f579ca2b0b6f58b513ecfe62263b1b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-27 22:07:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d9bbfebdf54064539e4da1852f47565d1f59c81","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-27 22:07:44.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"9fd2703a4289ef88bfd4659fd090c7a868cc064e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-27 22:10:12.000000000","message":"Patch Set 1: Workflow-1\n\nThis gives me the following on the task:\n\n | message    | Copied 500 MiB\n\nI dunno if it\u0027s legit to update that, but I don\u0027t really see it being\nused anywhere else. This not only gives the operator some details that it\u0027s really running, but should be updating the updated_at on the task (too frequently, but this is good for testing).\n\nSo, if this seems reasonable, then we can use this to at least facilitate liveness checking of the task for lock-busting purposes. Perhaps we could even write a liveness check into the task_info structure so that we\u0027re not fooled by anything else updating the task itself.","accounts_in_message":[],"_revision_number":1},{"id":"b3289f76b1d0b30b4041369129921c963fa0bbfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 23:26:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/870ce4b6129b4244968a1c9975a567ec : SUCCESS in 1h 10m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c1cb7954634d473ab5e5ab014177c967 : SUCCESS in 1h 17m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6dec7c667a5442ca37588a372b0ff09 : FAILURE in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e85f4a6179d04152a82b6f207acbef94 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55c893fdd91e43f5884f90413a06299c : FAILURE in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/761a706fed174f199caf25dd890d9042 : FAILURE in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee78171c066429fa63ab04eff7544ef : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff2c864d9cc64ee1a0ff524ad8e42f8b : SUCCESS in 9m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/61f8c3af4e9f44bfb9176b8255c3b60d : SUCCESS in 6m 44s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2a3d7f527b28471187533c46f624408e : SUCCESS in 5m 29s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bf61b62a119149088ed8e87a243eb7a8 : SUCCESS in 52m 24s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/999c47a430e64c1091bbf233de6e2a78 : SUCCESS in 1h 15m 08s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/05a1adb32f144098b263a46f1ac7fa42 : SUCCESS in 1h 04m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5090d2c2c9124068b878ff0b3679d339 : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d782a5f43a4eb72ceb2e2552c73079415d6da2d4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-28 17:08:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6c95ed9bddb37d0fd30a6824c2932041ad552c92","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-28 17:09:28.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"c86cb87f7b6c640a62dd293705d31ab0814aa7ba","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-28 20:37:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"07e27da9b766e39ab181499ca2afaf62cec5ead0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-28 20:38:50.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"10258df9564a0a1030a5b1484d5acebcb7851faf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-28 22:54:32.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"82d4fb9c28f244f46b191cc075f42e106d2c9fef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-28 22:55:48.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"1f6c81f3a1c9e5a5a7d87789943a83d17e2882ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 00:26:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/77c3f2d8dc934bd6b4dad6d153f4570b : SUCCESS in 1h 04m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60b2cb0784154364803494b8e6f1f60f : SUCCESS in 1h 23m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ad0870f87d94b21817e10bf5995f7f7 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da38a027d23f4490971ecf8d025e48e0 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7bc46f56b1545fd89c2a687a7b16f03 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e726f8633504fb285579141fa0f3784 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c3c2fdff544b33aba2b9f252e9ef13 : SUCCESS in 7m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b7796eb91bf74da58da5e35fee9818cd : SUCCESS in 7m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fb8cfcecf8064c85a7f3774f51b3199d : SUCCESS in 7m 03s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/293e317ce6314963abb5fe1bd7a99f23 : SUCCESS in 4m 21s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b3c8402c31e74b7495802eacf14fdf45 : SUCCESS in 58m 36s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/3939ba5ade9847de94d2f18f2a137a9b : SUCCESS in 1h 01m 10s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/875bd489c112403cbf9cbe8caa53735e : SUCCESS in 1h 12m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da7588f2254f418eaa93749cb34bacea : SUCCESS in 36m 17s","accounts_in_message":[],"_revision_number":4},{"id":"99ee443dc388e69bd24fe63fc6d98ebad8327ffb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-29 20:09:44.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"4d6de178fa7fcffb609c35832524fef0ded818bc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-29 20:11:41.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"5446e32b2405f88180286b497e94e982b3d6b980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 01:54:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e2d2f25739044cfb533a08cffcc392c : SUCCESS in 2h 08m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9660c140c28644868bb6be9e5ce86c04 : SUCCESS in 59m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b27d47d2d2b45eeb4f2f6aa7a7215f2 : FAILURE in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/750e93835a4e4bb9b2cb16f1f0185532 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b66c25a72a443678003a3dd13161a10 : FAILURE in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9c392da157645f9bb58479b450101f0 : FAILURE in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a07e7c3caea407092cf9e4d90e6080f : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e86d7cda754040888ccd19c98b3366cb : TIMED_OUT in 30m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9fb43a42e89040b1bb52c3c5ce811326 : SUCCESS in 7m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fc995412dacd4bf0953c4f0ada90e9b3 : FAILURE in 4m 33s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/028b48b844584ce2beef751cdf88ae19 : SUCCESS in 1h 06m 18s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/b73e557e543c4aec9ca4ec7aadd3e066 : SUCCESS in 43m 36s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/ceb1d9995bf54c7eabe269166fc3c7fc : SUCCESS in 1h 35m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb4653b9d7b549cfa0dd76e8765c7f80 : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":5},{"id":"24a261e81b97bb76d51fac8d70eb453305fa6a0a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-30 13:41:40.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"3c000d0a12caa09a3f31a942ffdf2d3ea940b822","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-30 13:43:58.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"383a54322a369874501f4952e2f006e0a52d127d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-30 13:44:31.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"5fd855e1f269eb945e480266bcc9b2ce93aa4e8b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-30 13:47:28.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"2b6873a36d206bfce8ec6f6e04fa16667b1a308d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 20:00:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/452b9744219b4348871d266f3fd4b352 : SUCCESS in 1h 08m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e0059aea6ee400997041dc36084e12a : SUCCESS in 55m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3b95a505ca34846ba2b03f54f7b2040 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f539863feac432aa06f5e9503074ee2 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/932e8086297f46efb188608ee3f335e9 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0d839b76a104a8986487f8b6781c18f : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b9c5b1dac514f638f97523418c1a8c9 : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29523e9eba144e2588c4678067cd08e7 : TIMED_OUT in 31m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2e07061ec3c44eefbb5e6e4a672d0cc8 : SUCCESS in 9m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c9ccd09bbef24c7ab6ba7e2ff5837a50 : FAILURE in 6m 50s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e511037e688e47c7bc8d4f9a152ace5a : SUCCESS in 1h 13m 48s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/5aa20a1818334ee8b387ff934d8bce9f : SUCCESS in 1h 05m 17s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/8f7163cebdca472c97feceeb29b85d0f : SUCCESS in 1h 04m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49de220952f34b71a969fcd50ea6c6bd : SUCCESS in 1h 13m 32s","accounts_in_message":[],"_revision_number":7},{"id":"790baa57e59a2a671bbb87019696196fc3aa66ef","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-31 00:26:37.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"c0a357fdd0fe6b487ee77b520fdfd27f278abd73","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-31 00:28:17.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"b3da294fb2bd7c1edad3a24152cadb3bc3b651b9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-31 00:35:39.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"c3ed6428a775b77313635fd66e50370c7224326b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-31 00:38:48.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":8},{"id":"721d7ee7534429a669b0f05602e4567cc4765705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 03:01:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b5872ef51eef4b5d9aca616889f50fcd : SUCCESS in 1h 00m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a09b20e5a21e4d9685eeab5b9ffeb421 : SUCCESS in 51m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83098712b64c4c5da906cb2ad7dac79f : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded21b7b4bcb45a48b516a1b90f67d2e : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ca2c3bcadc64553b0855cf8ee319cc4 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb2a9d84854640a694b290ba11ad762e : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1523280d1041ac8bd8382da9559191 : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a317837eeaa4f22b37d4c8b6cf7580a : SUCCESS in 6m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0347c53026f044aca6a047c3ef109ac2 : SUCCESS in 8m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/68e3ad54f0b34e6c8a70a73837863236 : SUCCESS in 5m 01s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e2d66cc83c05477280dd3cff9216418e : SUCCESS in 46m 50s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/b416c2a4eeb84c4da56476cce129f17d : SUCCESS in 1h 14m 24s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/c0e64784a8e94263a34b3a17b8d993db : SUCCESS in 1h 15m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d39b4ce2d42c458ba07ab7b194b082b3 : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":8},{"id":"fe93d85951cf60c46eb355c84fedc7d555742502","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-06 17:33:38.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"d8052b3ee96557a5e05f5ecdef0ceefe15328f6e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-06 17:35:57.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":9},{"id":"27503faddc4cce7a468e51b28e32d0d4ec651c1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 18:56:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/64bab7b97a7d455094fd13ccaf7f033d : SUCCESS in 1h 02m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/74e1dc3fe07144acbfccda642ad7e1d4 : SUCCESS in 1h 08m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02091faefed74015bbbddd8da1cc1a6c : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6054ee08b3bb46cbbcf1c9e6a194f048 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/312ddb29130442f492e286861c952d77 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a2d17e8930d4813872e87a97cb006a1 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdd1580eb00b4267ac6b278a4336ed83 : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/82eeabe5769046d38a4fa07163d8e404 : SUCCESS in 7m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5280ab1a7164426cb5f5beb050ba2555 : SUCCESS in 7m 43s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ec8d580c56d94ae4b78dea5dee3d7ccb : SUCCESS in 4m 16s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/692955f5d1c448f4a438c3b7aeb252dc : SUCCESS in 1h 05m 02s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f1e3bde2d4544355822daf74981a17df : SUCCESS in 50m 12s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/6ea82032a3c24ebf8494a9675e3d49fb : SUCCESS in 1h 16m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61a58a0a54514a969099aa2327ab3ee5 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":9},{"id":"04ac89f3e87a5ff2aa06f27c4dd270beae8616ff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-07 06:46:27.000000000","message":"Patch Set 9: Review-Priority+1","accounts_in_message":[],"_revision_number":9},{"id":"61db0e8834c0f6067c4e839040af7557da78a85f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 16:31:03.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7db1b1c537115e1db6134114b479d39e46a86b26","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 16:32:21.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":10},{"id":"63299d241854deb59004d8d1441433086936822f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 18:06:21.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1b5753c8cec243d9980f3ccc01da5a89 : SUCCESS in 1h 09m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/51cfa575450d46d3a5d3e0604f66a497 : SUCCESS in 1h 18m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c739cc9654ff40b18f3edb5bcc89a315 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dec59f70ebe74bfc833c4d6610468272 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b24b8ccd7a045c1bea9c36b69f3d152 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8563454ac4145528749f6f213ab240a : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb8ed6c229ad4fcfb182c6e02de77199 : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/96274cd3d7334d4aa57c7dd13a74e0c8 : TIMED_OUT in 31m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/587e52bfd25e48c48714d4c3564a8ac1 : SUCCESS in 11m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/44f34c6b5ebc48b4a4aae4a999aee822 : SUCCESS in 8m 44s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5e97ecd06bd14d8bab09d8dea842ccc8 : SUCCESS in 1h 21m 43s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/20648d830b19481f958082b6aa082ffa : SUCCESS in 1h 02m 53s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/af2220186295421a82600566252b891f : SUCCESS in 1h 13m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a149055375c43c196e017b122d205b6 : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":10},{"id":"58b37c408d71de652622cafd070380dc9fe379fd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 19:31:31.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5c277aa6f66348631428fcdadcf83a9f3451d422","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 19:36:46.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"ad97f2c2f3aa99c629879b5e8fcc2d1285fd559d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 21:23:08.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/30de400ce950464fa436c3aa9e75c64e : SUCCESS in 54m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f071d2c7570f477a8035ead05d46518a : SUCCESS in 1h 12m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/381b6f7f066b415baa6db7b53dae4cd9 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/960c86b9957e445181131921cf03942a : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed606d90a79c4259bed82723cedc0a0d : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3a0847cf8904726a654c41eb223c157 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c12d48c6b1c4151927fcbdac1cf699a : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5fdc5f55e88345468161123bcf6e2953 : SUCCESS in 7m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d2fb36745609408894506dcade05bc4a : SUCCESS in 7m 26s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ef89a39043b9459faff6d1ae6e56632c : SUCCESS in 4m 22s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2536820611b1421f95fb1bf74232016d : SUCCESS in 1h 00m 04s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/67ef9a9c11bb4f84b07e5f94bf7117f6 : SUCCESS in 1h 17m 08s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/f171a63018d5402b95d74ea152ed8d2d : SUCCESS in 1h 14m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3f43174a42049e0b8977af30817104c : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":11},{"id":"d3ad4436301b57892f29a951407ca633e665c5b5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-10 17:26:09.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":11},{"id":"ecd12cab90d9a7ee56358687a8e72d74225697c3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-10 19:20:55.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"29d189d570790e5ed0fdfbb93ccf7b5741530047","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 19:21:09.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"3aba43e58b6c4e54a05cefcc98f476ad77738936","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 21:08:07.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"2f704eee72bf08ce54f0961e2d837c28de3dcc89","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-10 22:01:00.000000000","message":"Patch Set 11:\n\nrecheck dep dep failed on cinder","accounts_in_message":[],"_revision_number":11},{"id":"7453df1d3de16ac8d8a2ab20f9ff4b5bf702dcd2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-10 22:01:29.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"80601d62df1060ea8d63d505f009a76370d5576f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 23:20:47.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a4f6b23cd8a04485a65c7f6467352b5c : SUCCESS in 1h 12m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d037673f0af04e75b6b7883c02918498 : SUCCESS in 1h 11m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e7f714b6bf445e8a4a10eea00eb3e6e : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b70cee8cecde466ab425172781a4c822 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3c7211b734e42fb9e5f19ed1439efd2 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/153894bbd2d14758a14f9a282afb3e9d : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc972de78d4e4481af17d1bb276721ad : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6080bc7bb1404f1a8506c506cad58b4c : SUCCESS in 7m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aa28dad6135b4a56ae28ac7f2a0b717a : SUCCESS in 7m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ef322da6e278405b88e5410434f90f6d : SUCCESS in 4m 21s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/cd1c6baa1912422bb89729214ad5f8da : SUCCESS in 1h 12m 17s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/adf9e5d83e0a4781bad7ed83226e1016 : SUCCESS in 1h 08m 19s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/e961150d87e140b7af0963622406c711 : SUCCESS in 1h 08m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93164e0c29ef4f538170013080354934 : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":11},{"id":"e59b19593df7a8200ec831d124d91db8eac5ceaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:03:01.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"a1d05ccc9d7959c8b61bd34ad95d29f81c06a2b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 01:27:26.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"3878327b27c74169a22ec45f154d685456d707e6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-11 06:02:53.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"e4f585cec93a74555ca5fa8c1f9db41ec66fcffa","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-11 06:03:22.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"190c77ea56c80f8b5f41921625926a860ee55d6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 08:24:02.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/20c9b9d28dfe469e9edd3b94fc81f56e : SUCCESS in 1h 08m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/edcbd5f13e6c4a3883657a4dc197d930 : SUCCESS in 47m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9c528e4bfd04bc197cd4d1cdb5683bb : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d928239034ce4bee91ab1ce0289e018f : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4afe7ff0579c473e93db3d90bad21aa3 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6654a9e7430c4236ba52c1a173e56769 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18da26b17d7a49b6a07b5bcbe38a8cc2 : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/95239955d8324881be3bfe4990b34e43 : SUCCESS in 7m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1176a75ce0024c36aee8eccb84f02f6c : SUCCESS in 8m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a2169c7fd9ae428b8ca3b1075b492224 : SUCCESS in 5m 27s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d1ae29231ec54ab593c80512ebafd1e3 : SUCCESS in 1h 01m 47s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d1c93c975874424b93096ed6f95c9187 : SUCCESS in 1h 11m 40s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/a8c4a1a87fea4c159c66193a7dc9456e : SUCCESS in 1h 12m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/878f844653cc42d697f109ba2512017b : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":11},{"id":"58badae337d6c285aba3e6a8419de5571c148954","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 11:56:59.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"5eeccedf8435ba2cd8b755ae13d9944bc36a31f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 13:26:01.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"3fece530e3f3950872fb904b37a885f4521e1181","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-11 13:30:24.000000000","message":"Patch Set 11:\n\nrecheck dep failed on functional timeout","accounts_in_message":[],"_revision_number":11},{"id":"284ba5bcdc4f65314296875c36ad2f5655893ecb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-11 13:30:48.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"55b49b6b17a344c1af1cf6b13017652c45031b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 15:02:50.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2fca1f37062942edb9e939a7cba409bc : SUCCESS in 54m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5db936345a424424a4b65d75c74c0e30 : SUCCESS in 1h 19m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40ad8d3d3d32406dabc81813310a6eb2 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c40f492619ab46f7844534982727ae6d : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93c7f70af9aa4cb8a6083afb979d4b72 : SUCCESS in 8m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12df788c536646458bc27fc4f4834c9e : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91790dc40aea46f1b345c1db4cac5f0c : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/578d8769a4094595bbdf7d45dff997c8 : SUCCESS in 10m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/16131a48d8cd48188957d534a0ff0fbc : SUCCESS in 12m 37s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ef190e903bd9453ea7af407e1764f9e4 : SUCCESS in 4m 33s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d15ee00680254a6a801ffd7d79e48780 : SUCCESS in 1h 20m 23s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/618d2da1a9b948f399a6866501bf2f24 : SUCCESS in 1h 23m 59s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/607aca2f54034abda332cc5fec9c5cd4 : SUCCESS in 1h 11m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52ac1e9e816846f097e51b30324399ca : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":11},{"id":"6945c6cd6eb1c617db54883f25e7385b2389061d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 15:08:23.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"31dfedc21e526fe734582ff118b350f5df0fef05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 16:34:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"f8ad11c931875c7610660f5369957b77db65d61a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 16:34:26.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8c060a10bc4c41d6ac833cf08c57170e : SUCCESS in 59m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/746aa92b5bb34d6a9af107af9214b1ea : SUCCESS in 57m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/850171d3ed1b4d6fab62210a95a7459b : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f403ec7bcf594ca6af009b2bb7af9230 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1127525aaba140d8affe0db2cda7e5d5 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb9f603583b146a7a2a5d711e0308803 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5445019a3bfe4b72b3e64efaf9410d76 : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ccec92e337f466883a14894756a0af6 : SUCCESS in 8m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/55350d3a21b64e919a84bba3440e5188 : SUCCESS in 9m 56s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/8c2bb5fb6a57491fb711acbfeffea704 : SUCCESS in 1h 17m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e795a52767964f5cb4b5506f65df39c6 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":11},{"id":"62a28d29696d4d3a1edf63e6a2390438df057633","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 16:38:42.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0008a37f5ee24360847e35e8af0c65fd : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"77d9cfa66e3c2e70a1cc515e7504fd5dd946549a","revisions":{"205367f1cd5370cf0b81d74d49589eaee68853dd":{"kind":"REWORK","_number":1,"created":"2020-07-27 22:07:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/743427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/743427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/743427/1"}}},"commit":{"parents":[{"commit":"122d2866d037d1dfb5f5ef0456e9b17bed787929","subject":"WIP: Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/122d2866d037d1dfb5f5ef0456e9b17bed787929"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"subject":"WIP: Update task message during import","message":"WIP: Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/205367f1cd5370cf0b81d74d49589eaee68853dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/205367f1cd5370cf0b81d74d49589eaee68853dd"}]},"branch":"refs/heads/master"},"fd820f66043836f8906e085d95fa50a19cef6326":{"kind":"REWORK","_number":2,"created":"2020-07-28 17:08:12.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/743427/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/743427/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/743427/2"}}},"commit":{"parents":[{"commit":"5d9d726585e024c39e90d6bcc778ac1f42c5854a","subject":"WIP: Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d9d726585e024c39e90d6bcc778ac1f42c5854a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 17:01:38.000000000","tz":-420},"subject":"WIP: Update task message during import","message":"WIP: Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fd820f66043836f8906e085d95fa50a19cef6326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fd820f66043836f8906e085d95fa50a19cef6326"}]},"branch":"refs/heads/master"},"e365613dbed53669e02260a4f1cd76f75628c3d5":{"kind":"REWORK","_number":3,"created":"2020-07-28 20:37:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/743427/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/743427/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/743427/3"}}},"commit":{"parents":[{"commit":"df951d07ab8fdf54c1e79a4a39d5157618402e15","subject":"Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df951d07ab8fdf54c1e79a4a39d5157618402e15"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 20:11:35.000000000","tz":-420},"subject":"Update task message during import","message":"Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread. Note that it also checks the state of our task, and will\nabort if it has been pushed out of \u0027processing\u0027 state externally.\nThis will be used in the following patch to faciliate import lock\nbusting behavior.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e365613dbed53669e02260a4f1cd76f75628c3d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e365613dbed53669e02260a4f1cd76f75628c3d5"}]},"branch":"refs/heads/master"},"e618cca1911a9e4bb562a1b9950f5e79d6bf7fb5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-28 22:54:32.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/743427/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/743427/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/743427/4"}}},"commit":{"parents":[{"commit":"f1a938f930a9cffe9809f5236189e7d861354fb4","subject":"Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1a938f930a9cffe9809f5236189e7d861354fb4"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 22:54:27.000000000","tz":-420},"subject":"Update task message during import","message":"Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread. Note that it also checks the state of our task, and will\nabort if it has been pushed out of \u0027processing\u0027 state externally.\nThis will be used in the following patch to faciliate import lock\nbusting behavior.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e618cca1911a9e4bb562a1b9950f5e79d6bf7fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e618cca1911a9e4bb562a1b9950f5e79d6bf7fb5"}]},"branch":"refs/heads/master"},"2052a995a6df63bce0b7001c5157ed0c3ab1ca16":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-07-29 20:09:44.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/743427/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/743427/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/743427/5"}}},"commit":{"parents":[{"commit":"331ee31ce405145c563ee064eeaa467a50c8a608","subject":"Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/331ee31ce405145c563ee064eeaa467a50c8a608"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-29 15:16:46.000000000","tz":-420},"subject":"Update task message during import","message":"Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread. Note that it also checks the state of our task, and will\nabort if it has been pushed out of \u0027processing\u0027 state externally.\nThis will be used in the following patch to faciliate import lock\nbusting behavior.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2052a995a6df63bce0b7001c5157ed0c3ab1ca16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2052a995a6df63bce0b7001c5157ed0c3ab1ca16"}]},"branch":"refs/heads/master"},"a2712e73212c4254dc25836790dd3f6e2718464f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-07-30 13:41:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/743427/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/743427/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/743427/6"}}},"commit":{"parents":[{"commit":"a5996cdaadf2fb80b23027360c5aa2b10ed49689","subject":"Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a5996cdaadf2fb80b23027360c5aa2b10ed49689"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-30 13:40:55.000000000","tz":-420},"subject":"Update task message during import","message":"Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread. Note that it also checks the state of our task, and will\nabort if it has been pushed out of \u0027processing\u0027 state externally.\nThis will be used in the following patch to faciliate import lock\nbusting behavior.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2712e73212c4254dc25836790dd3f6e2718464f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2712e73212c4254dc25836790dd3f6e2718464f"}]},"branch":"refs/heads/master"},"6474e14b2850262cafb424245a0ff25c9e353fe2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-07-30 13:44:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/743427/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/743427/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/743427/7"}}},"commit":{"parents":[{"commit":"ddc1ab753f111602ae4768204948c9ec87a70f12","subject":"Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ddc1ab753f111602ae4768204948c9ec87a70f12"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-30 13:44:24.000000000","tz":-420},"subject":"Update task message during import","message":"Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread. Note that it also checks the state of our task, and will\nabort if it has been pushed out of \u0027processing\u0027 state externally.\nThis will be used in the following patch to faciliate import lock\nbusting behavior.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6474e14b2850262cafb424245a0ff25c9e353fe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6474e14b2850262cafb424245a0ff25c9e353fe2"}]},"branch":"refs/heads/master"},"72aa41a2b51ca7aaeeb86c49903603e4ba268bcd":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-07-31 00:35:39.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/743427/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/743427/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/743427/8"}}},"commit":{"parents":[{"commit":"7def80ddbc04054ee242629b383b3b6a7ced32f9","subject":"Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7def80ddbc04054ee242629b383b3b6a7ced32f9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-31 00:35:33.000000000","tz":-420},"subject":"Update task message during import","message":"Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread. Note that it also checks the state of our task, and will\nabort if it has been pushed out of \u0027processing\u0027 state externally.\nThis will be used in the following patch to faciliate import lock\nbusting behavior.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/72aa41a2b51ca7aaeeb86c49903603e4ba268bcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/72aa41a2b51ca7aaeeb86c49903603e4ba268bcd"}]},"branch":"refs/heads/master"},"9075fac9b79011733b0d25a81886df5d291f7169":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-08-06 17:33:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/743427/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/743427/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/743427/9"}}},"commit":{"parents":[{"commit":"065a4603cad1926768dbade9ba6e65024f2c0156","subject":"Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/065a4603cad1926768dbade9ba6e65024f2c0156"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-06 17:33:25.000000000","tz":-420},"subject":"Update task message during import","message":"Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread. Note that it also checks the state of our task, and will\nabort if it has been pushed out of \u0027processing\u0027 state externally.\nThis will be used in the following patch to faciliate import lock\nbusting behavior.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9075fac9b79011733b0d25a81886df5d291f7169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9075fac9b79011733b0d25a81886df5d291f7169"}]},"branch":"refs/heads/master"},"70fd3ee8a10f4d62c5727aeb15bc4f77ccc21500":{"kind":"REWORK","_number":10,"created":"2020-08-07 16:31:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/743427/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/743427/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/743427/10"}}},"commit":{"parents":[{"commit":"5c29ba45ddfd9e947aea9242e90062b6f2b967fb","subject":"Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c29ba45ddfd9e947aea9242e90062b6f2b967fb"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 16:28:03.000000000","tz":-420},"subject":"Update task message during import","message":"Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread. Note that it also checks the state of our task, and will\nabort if it has been pushed out of \u0027processing\u0027 state externally.\nThis will be used in the following patch to faciliate import lock\nbusting behavior.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/70fd3ee8a10f4d62c5727aeb15bc4f77ccc21500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/70fd3ee8a10f4d62c5727aeb15bc4f77ccc21500"}]},"branch":"refs/heads/master"},"77d9cfa66e3c2e70a1cc515e7504fd5dd946549a":{"kind":"REWORK","_number":11,"created":"2020-08-07 19:31:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/743427/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/743427/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/743427/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/743427/11"}}},"commit":{"parents":[{"commit":"e49f23c04d784b5312d7df4556e2e5064555a84f","subject":"Heartbeat the actual work of the task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e49f23c04d784b5312d7df4556e2e5064555a84f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-27 22:05:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 19:22:31.000000000","tz":-420},"subject":"Update task message during import","message":"Update task message during import\n\nThis updates the task.message field with details about the copy,\nwhich will also update the updated_at field on the task. This will\nfacilitate some rudimentary liveness checking from outside the task\nthread. Note that it also checks the state of our task, and will\nabort if it has been pushed out of \u0027processing\u0027 state externally.\nThis will be used in the following patch to faciliate import lock\nbusting behavior.\n\nChange-Id: I8667c17813f6e701db98595b0b30df9e7b275294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77d9cfa66e3c2e70a1cc515e7504fd5dd946549a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77d9cfa66e3c2e70a1cc515e7504fd5dd946549a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
