)]}'
{"id":"openstack%2Fglance~743596","triplet_id":"openstack%2Fglance~master~I0b71a7df04cd330749f35b07f96a120b49b412c7","project":"openstack/glance","branch":"master","topic":"task-heartbeat","hashtags":[],"change_id":"I0b71a7df04cd330749f35b07f96a120b49b412c7","subject":"Add image_delete_property_atomic() helper","status":"MERGED","created":"2020-07-28 17:08:12.000000000","updated":"2020-08-11 13:13:10.000000000","submitted":"2020-08-11 13:10:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743596-1597151436901-b7eb80e6","meta_rev_id":"7e67c8f9f06d22e3b70f117f6c9e39399917428b","_number":743596,"virtual_id_number":743596,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-11 09:39:45.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-08-10 22:00:20.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-11 13:10:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-10 19:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-08-07 19:31:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-10 19:08:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 00:27:14.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-31 06:16:19.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-10 19:08:58.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-11 09:39:45.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-11 13:10:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a8e95645b8ca7d5c5f803cc2da9087076709f9b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-28 17:08:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f55d382011f3c4c55b639088c91f98e1dece636c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-28 17:10:25.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"5c7bbaf2da8cdb725da70e7d96d62d50a4fdd973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:19:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7942c5b59d084e95a1a769fdc1e47352 : SUCCESS in 1h 03m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32bd2bbfce504d9fb82db3dbed05e2b9 : SUCCESS in 1h 30m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da0fe062854547ed8f6f960d667999d1 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9e49fa797e94ad991bfdfa19b98e817 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51c1c6adc6254852a06fe07de57d5a46 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7c37efde40d4f3183e6b6774d9a80e2 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663c41feaeb04f31a8e31ea46d17ed4e : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8164d0b60d364930b0a7cd85c2498f2f : SUCCESS in 8m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cec24076984848259a73be3f0fae1742 : SUCCESS in 9m 15s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bf2e81c6595d4677ba5a421e07cd3ee5 : SUCCESS in 5m 09s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/091e1236c8194b8da70e2094fb3444cc : SUCCESS in 1h 04m 00s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/30418dd8e8464d0286dff1020e2c5a93 : SUCCESS in 1h 08m 25s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/0fcd42eaf6324303a7e6119503ec1c7b : SUCCESS in 1h 06m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/344b84af24da4a42a93839a923033ffa : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9b93e7e948256e08db54ed2bacb8495f23a23f25","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-29 20:09:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"53ebf8006e2f592afb160fd856ffbe1991bb3138","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-29 20:10:29.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"dab9e5c184bd098e15e9d41efe93fc4ed78dc4b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 01:03:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/19efd359b33d4621bb33eed2150a667b : SUCCESS in 1h 09m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b243fbd1eb2c43078cbd63cab7359930 : SUCCESS in 1h 10m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64848eb302d04170a5489dfd2d80e7a7 : FAILURE in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5baaad364bff4de88fccc6f2e051fe70 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59e6f9dcd6d24a33ab5282a55a2cca72 : FAILURE in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b4a394fa4c745ce83b5149b87596b46 : FAILURE in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee07f9d0c534e1e9dae4824dfcfadbb : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c021b6bc481e406b951bd6eb617fdb97 : TIMED_OUT in 31m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0794420068784b34a3ddd70ba0bbe7da : SUCCESS in 7m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/427f57da9cd448e0a200528e7837866c : FAILURE in 4m 12s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2d982e09ca264837983f75132cd518b9 : SUCCESS in 56m 26s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/ff4ec129723b42c1af0cff6c6750e293 : SUCCESS in 1h 26m 03s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/9d312cc0eb97409dbd099a1e87b5353d : SUCCESS in 1h 11m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de4666bd5d40435990f37d2f52fb1833 : SUCCESS in 1h 18m 28s","accounts_in_message":[],"_revision_number":2},{"id":"c08bd4a196cb6f6fbbd88d129dd344903e956fe8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-30 13:41:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"035e547b172481a5dd26787b68d222dfb0d962ea","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-30 13:43:01.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"a431b700652e0a50bd34f2f4294e524ee7b8e3d8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-30 13:44:31.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"5a3ea14cab68766211e76e8774a027a8d17c2983","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-30 13:46:15.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"1a660fb0cedcdc721cabda2c6d9a7d124eff0f1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 19:57:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f648a3792fc64395a94ba25131de7471 : SUCCESS in 1h 10m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88940b6585314be599e0511836f90a68 : SUCCESS in 1h 17m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f560b777f3543988652ddb74308cc8e : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05a5004b518c4b65b208d2ef9b5c2f09 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a11ccf12d64424280b988ca101dd5f5 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0555fb3ca6f4cb0bded1c571d7c96ce : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/339bed9aee0b4b5782bfd08539b52244 : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b09aa19b79044f1eaf7c294f95341815 : TIMED_OUT in 31m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/40357bbf41ec468297bc8deee90c8e4b : SUCCESS in 16m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a919f29320b94cfab700fca531f7d37f : FAILURE in 4m 22s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/dabdfa97163f456487924dab0a56a39e : SUCCESS in 1h 10m 56s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/6b69c794f0e84a74bda7e0d67ff4b748 : SUCCESS in 1h 07m 01s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/b6e2609ac6694c1b920642d9771655ef : SUCCESS in 1h 17m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a05b8e38080649239de27f3e428a4e0c : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":4},{"id":"c7907d9347148ec2713d4698e13e0421ad5bc301","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-31 00:27:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9b46bbb72d725f6e52c3e869d95c172bc2aa7946","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-31 00:29:25.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"cb4efd2fa82017b7fb3409d53f764f4a95518730","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-31 00:35:39.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"6960c78aca3f5bc95f2597dfaeb03c5f95e1d280","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-31 00:37:27.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"0d90ccec71781e655bffd4dfa022a79293ee7c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 03:06:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2eaca2daec024962aee1c01323a8b37f : SUCCESS in 1h 02m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5fa35e88c91476f81b742df6c2cd209 : SUCCESS in 1h 03m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0634426984c84eda8d287a91ad481f20 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b332db2ded4c779423130718d48e4d : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9d2a1fb5a8043299bd0c60f703faa93 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f506f08d07c46899c0d7248c64d199b : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a78a677ce0144f19757595f11d57de5 : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f60f4ceeae742cfb6978bee76cca5ad : SUCCESS in 10m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/182cfce599ee4912a97764d0db79c36a : SUCCESS in 10m 28s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/244265773ba440208796c57cc1f68768 : SUCCESS in 5m 31s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/49ab6d18bbc14a788b8766b9a5978325 : SUCCESS in 1h 12m 17s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/9a2ed1f014744569995322724c00acf4 : SUCCESS in 1h 23m 45s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/415894e33c9046feb9233265ac25ed89 : SUCCESS in 1h 20m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b6902e60eda4f988efa533e18bbd719 : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":5},{"id":"161c694ff89a9e58a475536d3568c4defca5fdf3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-31 06:16:19.000000000","message":"Patch Set 5: Code-Review+2\n\nShould be tagged with related-bug as we need to backport?","accounts_in_message":[],"_revision_number":5},{"id":"2f8fb9954d0605be7e467822ae9b28ca57d8df2b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-06 17:33:38.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"c947f9cc716768b4e4cc4e1b79fd93f275ab93b8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-06 17:34:55.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"1e62a1a7b511a1cab338687bc96c32b5c896471d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 18:57:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1bdc040851394ddcac0aef0ce5025fd6 : SUCCESS in 1h 13m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4eb08f51d584c809859c5fa016dbb1d : SUCCESS in 1h 16m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/779bbd86779744518cb7d1b021b9246a : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ef1e21241e420f923638b928175726 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f6bcf569c8444808ff0138c1c04a65e : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bc5e62fc33342d5a37f3e55acfe3a28 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/036a87e1256c4174b696b04a1436c219 : SUCCESS in 8m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9a9e06c2db924b8282bcaa6b747ed7b8 : SUCCESS in 7m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e65152edb30845898a401950e18ababb : SUCCESS in 14m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/38af87a5b8234e09905b4c5737626492 : SUCCESS in 4m 55s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a507076f368e40e1bfa39d0e135c02b2 : SUCCESS in 1h 19m 30s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/53a6ef02e83e4e2ab29943c912f9dd4a : SUCCESS in 1h 12m 06s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/7b083125fcaf4de3836e6ce47b37c755 : SUCCESS in 1h 21m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbf880d7f5b6466d8bca4a5e23b28d16 : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":6},{"id":"4d17c84ed91888f88783e5222aa7f3e8996bbae6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 19:31:31.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"df0d386e278a1a6d73986ced94d03815e22af3bb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 19:35:38.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"51774897bf0ae561df82930d08ada34e0777317e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 21:15:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/88118ec47a604ccab351871f423b665c : SUCCESS in 1h 04m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85e8529b00d24c3c809ee1ccf99c3f33 : SUCCESS in 1h 24m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da88c9794b1743e394be183e726e4b42 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a058ceff2676415fa1598310fc5e9ed7 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/647f9c84b569431f83eb922e2ffc77c8 : RETRY_LIMIT in 0s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56d4036715ac4915beac5ef6589cefd4 : RETRY_LIMIT in 0s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f7e0f73d68144a193fb7884a4d4e491 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa0363b38c494dc1baafda890829c25b : SUCCESS in 6m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3caf7dccb20f4d4697feb3481039f3fe : SUCCESS in 11m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6d2ef931674c4be8b6aaf1060d4ac475 : SUCCESS in 5m 29s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0f26efc335524d6087bdf4d6ccd3bee2 : SUCCESS in 1h 02m 34s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/2e2bde5083ed46fd9613488e6aa37649 : FAILURE in 1h 36m 28s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/177c2366e80240fcb57e112da0b59419 : SUCCESS in 1h 08m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72a9bf9f1c7e4bdf9c60eae5d5fc0f61 : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":7},{"id":"302ffb9e01c475a07f94d07b1ee57aae82aff3d9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-10 14:14:34.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"21bafe0c9a1888b5efce266ac651479d6495a7a6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-10 14:49:25.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"4431fed3fd02c7535ad6353e5c9fb5950d93bac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 17:37:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9846432792a44e5cb963774a479573f2 : SUCCESS in 1h 07m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3495de3815d2466587ed178dc42dc9e0 : SUCCESS in 1h 13m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8979ab78e0e148a7ac055ef908a0fcbe : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc7c17130fd34216a1593bb4c8e036a7 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01063d2b536c4190a5a909e6b6e2a81d : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/115eeeb8bcbc481584ace8a2457693f2 : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7650edfa680f4992b09b8b8ea601c1a3 : SUCCESS in 8m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e3ddade097e5449ca5aecf28c9e02bbe : SUCCESS in 8m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bca784e17ecf405b959a2cc359f5f2fe : SUCCESS in 9m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ecb12f6c2c164099b721a635d94f3fff : SUCCESS in 5m 29s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7697163773d24a4491d2e4133ab3f046 : SUCCESS in 52m 16s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/2371163db25847808132e73336f55847 : SUCCESS in 49m 10s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/1999a6264050435b9461c89b4f83f857 : SUCCESS in 50m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e8e795138e94efc91de1ba14f235bae : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":7},{"id":"df3c2d1b967c040513ac7c79c3c3c7f4611b0ec1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-10 19:08:58.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"fc441ec2df4fb1eb3d67ae36acd5b1e563787f93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 19:09:13.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"da6da2b222176285a10a21181313325d6a5c6c83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 21:08:05.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a1c9163321c742758767de733562f053 : SUCCESS in 1h 08m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/08e6e6307a9a44b1a9e73fccd1d8fe3e : FAILURE in 1h 50m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d7ff826a8364476be74b57df5b3676a : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a3b79bab274e7aa5d817954dfb2375 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8399252481834aceb618451f65c1124d : SUCCESS in 10m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/955356bcd8534004b98a6b64d5ba3550 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1763693f544c62b6465fa6da93955b : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c37cc7656eea41f9b4a52b030c1a9e86 : SUCCESS in 17m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/55d992b880ca44048723b2cb7531abb1 : SUCCESS in 18m 25s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/221e7717c7ff4b639c23127c465d5315 : FAILURE in 1h 57m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e30b41db37564a918143729f8d8be06a : SUCCESS in 1h 23m 54s","accounts_in_message":[],"_revision_number":7},{"id":"9fbfba611b3566b7470f97218c811a26a3ee5544","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-10 22:00:20.000000000","message":"Patch Set 7:\n\nrecheck volume backup failure","accounts_in_message":[],"_revision_number":7},{"id":"4ca31e7f6b0fdaaf724deffd323798cd7e113e7f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-10 22:00:43.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"d1826c94077c3618fab22ad36b8d9a667345e05e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:02:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/93bb1de412004ed5ad90ef30be45a26e : SUCCESS in 1h 03m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/40108d501b454365893e020842437929 : SUCCESS in 1h 15m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/715cae16415341f6ade01aa2d4c26901 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/135821e2c1bd421293b45aeb34ad56f1 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5ff65e9917d4476b170d8c09989caa8 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deacc183a5bf4f908823f9a1fa190899 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b0dfd74c24a4845961109c8a7ae60ba : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fea7e369eb0c4511a7fc37e1344832d0 : SUCCESS in 6m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e75d9a7730314e17a9f80979dc1679f0 : SUCCESS in 7m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/51d54677a9d14d5a89e12fd98232af66 : SUCCESS in 4m 37s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e2ce336dfa5448a886058df283d20fc4 : SUCCESS in 1h 44m 26s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/305e832f37ac41c0828c4a3f79c2f3de : SUCCESS in 1h 12m 40s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/65984479f9504b02963c4c885d1993ed : SUCCESS in 1h 52m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f90d99829034aa081f60a1d35a75177 : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":7},{"id":"8710ca3982d66925c01a37b1e2da854179acc89b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 00:03:00.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"dc723dd58c53950583b47d86147d4b6fca4e3786","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 01:27:25.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a5b2cfa0bb884a1da4fe328de60bd29c : SUCCESS in 1h 11m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5dd1b9a6a4fc41a390e6a273f4e88624 : FAILURE in 1h 19m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1210fce4b9924c93831b4463537d0210 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9fbcffb3b5f44e3ac547c6d5f0d7b9a : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e595416360174cf8bd9b315ee97cedd4 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b31622d6a9e435ca685b7042d429e14 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2676e8aec54d54986389ff0de08c1f : SUCCESS in 9m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9649ba232952459f87116073a611636f : SUCCESS in 11m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/002914cca2e54a599b60c6023c35e3bb : SUCCESS in 12m 14s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/bfb117a6870946ffbc9cc60f40e898a7 : FAILURE in 1h 22m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32aee065cc21416490a9735b1ae3093d : SUCCESS in 1h 06m 45s","accounts_in_message":[],"_revision_number":7},{"id":"d9b559f6b7cc301b94f0ca7c2e373457fc8c61dc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-11 06:02:29.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"af0b734a3138cc8fefe068977f73b17adb52b8b5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-11 06:02:51.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"c63e0a7618fc8293d928e0ae40fb3bb1fb38da7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 08:25:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4ea1582981314a18a32e84a91f961105 : SUCCESS in 1h 04m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4d142c9fb9141d0aecbfc0c800225b6 : SUCCESS in 1h 10m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e5814eeccbd4ba6a1118a0fb31874e0 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc46906a73364c3cb690ce369b2c589d : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59593dc69e91410b993effce166fd872 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bde9b832de954f05b3d70a6c0b413d41 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ccbbdb44044099a3dfd37ed1e32b8e : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5597779e0ffb4e2ca93e5babd5f3e815 : SUCCESS in 7m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca461410bcf74d4dac4b0843fcd4644b : SUCCESS in 8m 28s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/18d3cac581b84425b8d83266e935b439 : SUCCESS in 6m 45s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6c3550efa16c423b9d5ed2adc7a72748 : SUCCESS in 1h 01m 45s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/c0baff00964247ada5d9a8e3b209f649 : SUCCESS in 1h 05m 50s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/6d36b85962bd439abaff55893226308e : SUCCESS in 1h 07m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e01ef066a7794230b450b0d079115aa3 : SUCCESS in 53m 13s","accounts_in_message":[],"_revision_number":7},{"id":"d29104a8272d69c0195aafd68e19ab94f3bf1b9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 08:25:27.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e4d0f95f116860e461abdae843a2399d77ca3b01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 09:32:40.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/72c4343445fd462995b1aa33fe34697e : SUCCESS in 56m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/175c31bdaf114e408c8d3afc4899c450 : SUCCESS in 48m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6840fb7b36f041cebd9713007981f709 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/606b5c82dc994dac9428967290c1a17d : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73424f42c3ae4c5eb56492795188d420 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47cbd3f037c041fbbab1df7f5b7d73c7 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c49c709f1a47c0a3f9bdfcb9e9302a : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/42a7d9600ca74523b046b68f67a8fec4 : SUCCESS in 6m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/63c6485305c840db8666731e7388c06c : SUCCESS in 6m 46s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/bbf9779f7baa4ffd856e6c237fef6990 : FAILURE in 1h 02m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d8c2de9bd0440578c5f5d80ad8953d1 : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":7},{"id":"f020da1aa177a78dcbad8c2bbea8f18dc1bc1cca","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-11 09:39:23.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"2ab19ec61912dda5f4a46018697ae46fddb9bb1f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-11 09:39:45.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"045281726264b0dae62aba239f3fc3d488d554ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 11:56:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2567c9731752494f96ba1649f897831a : SUCCESS in 53m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aa37c8fb69bd4765a575260337c155fc : SUCCESS in 1h 15m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf1d078fa9284a06b9ac21847be333b1 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18016467d8bf4c72b57004f87a7b01f4 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52f4496b8dae48499264da8fe34c21aa : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6734f02d03c7453493d876dfba82b5fa : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a652cae3f2e94341a923efde43f0355b : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9b0f2bdcec14411cb3077509c1652222 : SUCCESS in 9m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc51fe551fab4c4b9cf29392fefedde3 : SUCCESS in 7m 52s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4173896c35f24580903f98a267727287 : SUCCESS in 4m 41s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8de1fed567394f80a7e296ce42775406 : SUCCESS in 1h 00m 12s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/68ccd33df2cd420295a7822f0f2b3d40 : SUCCESS in 1h 00m 05s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/119f67ca5e1d46fda435b51167bed34c : SUCCESS in 1h 07m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09d6cdc5c58248d49fac67031c75eb32 : SUCCESS in 2h 03m 44s","accounts_in_message":[],"_revision_number":7},{"id":"c7063426b14c09651a8f2e378a5247ba7686852d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 11:56:49.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e30da02fdfbc932ddcb1b6213f87bafb7ad992d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 13:10:36.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a3061744a3a54f84b25e8601efc97852 : SUCCESS in 55m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2e73646fbc9946b7ba39f96e9b6589f2 : SUCCESS in 1h 04m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3993b3c14e94eefb8fd9268a2b1a4da : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae52e59192647c29cedd1ae6765eade : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fd74b706b524dadba3e99ba3129a728 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dd8337eb12446e4b4bfb61d9c3211c4 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8720e409b0e41e7a24e8ef7b0f6e1cd : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c4eb4d6b4e548be94a13118fba376ec : SUCCESS in 6m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c9996bcfe4934073b5b1ad4c7eb26b02 : SUCCESS in 6m 55s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/7401b4b997ae443bbebd244c5793f63a : SUCCESS in 1h 08m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81197ec67cd04814b57cf101ff416846 : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":7},{"id":"6eef0a8e68d9729c6be2db37c2789d159a1c2396","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 13:10:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"c2d8a48f319b7f3cd1f84043b68a429bba70400a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 13:13:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57970d8b098d41aea61f736ba4b3a402 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b093ae3514dea32a8829239e0297e3e170efff58","revisions":{"f2cece8fe2a786018bc6dbbb1b824b4f1c9cb9e5":{"kind":"REWORK","_number":1,"created":"2020-07-28 17:08:12.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/96/743596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/743596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/743596/1"}}},"commit":{"parents":[{"commit":"2b983e9631a3269724b943664b4a061a32b1262d","subject":"Add tests for _ImportToStore.execute()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b983e9631a3269724b943664b4a061a32b1262d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 13:49:50.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 16:34:53.000000000","tz":-420},"subject":"Add image_delete_property_atomic() helper","message":"Add image_delete_property_atomic() helper\n\nThis adds a new DB API method to atomically delete a property on an image\nin a way that we can be sure that it is deleted one and only once, and\nwithout affecting the rest of the image. This can be used in conjunction\nwith image_set_property_atomic() to create and delete properties for locking\nwithout the risk of clobbering other image data in the process.\n\nChange-Id: I0b71a7df04cd330749f35b07f96a120b49b412c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2cece8fe2a786018bc6dbbb1b824b4f1c9cb9e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2cece8fe2a786018bc6dbbb1b824b4f1c9cb9e5"}]},"branch":"refs/heads/master"},"2dd5a3d5af6666e940d8e828963b485bb7131fc2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-29 20:09:44.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/96/743596/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/743596/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/743596/2"}}},"commit":{"parents":[{"commit":"1d22b1f69bddc871156c27d9aabe4802014928ce","subject":"Flesh out FakeImage for extra_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d22b1f69bddc871156c27d9aabe4802014928ce"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 13:49:50.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-29 15:16:46.000000000","tz":-420},"subject":"Add image_delete_property_atomic() helper","message":"Add image_delete_property_atomic() helper\n\nThis adds a new DB API method to atomically delete a property on an image\nin a way that we can be sure that it is deleted one and only once, and\nwithout affecting the rest of the image. This can be used in conjunction\nwith image_set_property_atomic() to create and delete properties for locking\nwithout the risk of clobbering other image data in the process.\n\nChange-Id: I0b71a7df04cd330749f35b07f96a120b49b412c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2dd5a3d5af6666e940d8e828963b485bb7131fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2dd5a3d5af6666e940d8e828963b485bb7131fc2"}]},"branch":"refs/heads/master"},"a0d2b1bacd69540d4635323625bd1a0915172bf1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-30 13:41:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/96/743596/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/743596/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/743596/3"}}},"commit":{"parents":[{"commit":"5dc69e61038038d4d159f1204771a7129089d0eb","subject":"Flesh out FakeImage for extra_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5dc69e61038038d4d159f1204771a7129089d0eb"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 13:49:50.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-30 13:40:55.000000000","tz":-420},"subject":"Add image_delete_property_atomic() helper","message":"Add image_delete_property_atomic() helper\n\nThis adds a new DB API method to atomically delete a property on an image\nin a way that we can be sure that it is deleted one and only once, and\nwithout affecting the rest of the image. This can be used in conjunction\nwith image_set_property_atomic() to create and delete properties for locking\nwithout the risk of clobbering other image data in the process.\n\nChange-Id: I0b71a7df04cd330749f35b07f96a120b49b412c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a0d2b1bacd69540d4635323625bd1a0915172bf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a0d2b1bacd69540d4635323625bd1a0915172bf1"}]},"branch":"refs/heads/master"},"c0c2d5f2e8a3720e25c91a93adc41a8c7d51ae78":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-30 13:44:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/96/743596/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/743596/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/743596/4"}}},"commit":{"parents":[{"commit":"d9289566958f21404f65eec62250fa9d90bc5143","subject":"Flesh out FakeImage for extra_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9289566958f21404f65eec62250fa9d90bc5143"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 13:49:50.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-30 13:44:24.000000000","tz":-420},"subject":"Add image_delete_property_atomic() helper","message":"Add image_delete_property_atomic() helper\n\nThis adds a new DB API method to atomically delete a property on an image\nin a way that we can be sure that it is deleted one and only once, and\nwithout affecting the rest of the image. This can be used in conjunction\nwith image_set_property_atomic() to create and delete properties for locking\nwithout the risk of clobbering other image data in the process.\n\nChange-Id: I0b71a7df04cd330749f35b07f96a120b49b412c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c0c2d5f2e8a3720e25c91a93adc41a8c7d51ae78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c0c2d5f2e8a3720e25c91a93adc41a8c7d51ae78"}]},"branch":"refs/heads/master"},"3418c235b332d032229c481ccf1a1e186ef3ecc4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-07-31 00:35:39.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/96/743596/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/743596/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/743596/5"}}},"commit":{"parents":[{"commit":"c8b70f45b88127e4f8ab24443933a8e5072e8721","subject":"Flesh out FakeImage for extra_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c8b70f45b88127e4f8ab24443933a8e5072e8721"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 13:49:50.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-31 00:35:33.000000000","tz":-420},"subject":"Add image_delete_property_atomic() helper","message":"Add image_delete_property_atomic() helper\n\nThis adds a new DB API method to atomically delete a property on an image\nin a way that we can be sure that it is deleted one and only once, and\nwithout affecting the rest of the image. This can be used in conjunction\nwith image_set_property_atomic() to create and delete properties for locking\nwithout the risk of clobbering other image data in the process.\n\nChange-Id: I0b71a7df04cd330749f35b07f96a120b49b412c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3418c235b332d032229c481ccf1a1e186ef3ecc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3418c235b332d032229c481ccf1a1e186ef3ecc4"}]},"branch":"refs/heads/master"},"02f905aecfd6adef47d5b63c7860c2aecd6d478c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-08-06 17:33:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/96/743596/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/743596/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/743596/6"}}},"commit":{"parents":[{"commit":"f7f471af7ab62852c0ee5e11321733877fd7e3b6","subject":"Flesh out FakeImage for extra_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f7f471af7ab62852c0ee5e11321733877fd7e3b6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 13:49:50.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-06 17:33:25.000000000","tz":-420},"subject":"Add image_delete_property_atomic() helper","message":"Add image_delete_property_atomic() helper\n\nThis adds a new DB API method to atomically delete a property on an image\nin a way that we can be sure that it is deleted one and only once, and\nwithout affecting the rest of the image. This can be used in conjunction\nwith image_set_property_atomic() to create and delete properties for locking\nwithout the risk of clobbering other image data in the process.\n\nChange-Id: I0b71a7df04cd330749f35b07f96a120b49b412c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02f905aecfd6adef47d5b63c7860c2aecd6d478c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02f905aecfd6adef47d5b63c7860c2aecd6d478c"}]},"branch":"refs/heads/master"},"b093ae3514dea32a8829239e0297e3e170efff58":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-08-07 19:31:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/96/743596/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/743596/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/743596/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/743596/7"}}},"commit":{"parents":[{"commit":"7c91a177edeb3782e99993d0568cddf31bff96a8","subject":"Flesh out FakeImage for extra_properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7c91a177edeb3782e99993d0568cddf31bff96a8"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-28 13:49:50.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 19:13:22.000000000","tz":-420},"subject":"Add image_delete_property_atomic() helper","message":"Add image_delete_property_atomic() helper\n\nThis adds a new DB API method to atomically delete a property on an image\nin a way that we can be sure that it is deleted one and only once, and\nwithout affecting the rest of the image. This can be used in conjunction\nwith image_set_property_atomic() to create and delete properties for locking\nwithout the risk of clobbering other image data in the process.\n\nChange-Id: I0b71a7df04cd330749f35b07f96a120b49b412c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b093ae3514dea32a8829239e0297e3e170efff58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b093ae3514dea32a8829239e0297e3e170efff58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
