)]}'
{"id":"openstack%2Fglance~743914","triplet_id":"openstack%2Fglance~master~I709a5d5b164031ffaabbf4164f4da4b080200543","project":"openstack/glance","branch":"master","topic":"fix-migration-version","hashtags":[],"change_id":"I709a5d5b164031ffaabbf4164f4da4b080200543","subject":"Change database migration version to victoria","status":"MERGED","created":"2020-07-30 06:20:19.000000000","updated":"2020-07-30 23:39:09.000000000","submitted":"2020-07-30 23:37:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743914-1596152239265-60d190a1","meta_rev_id":"e55ad35c32784ad842e15cb6ad32725ced71461c","_number":743914,"virtual_id_number":743914,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-30 15:59:16.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-30 23:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-30 18:42:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-30 18:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-30 18:55:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-07-30 15:58:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-30 13:53:35.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-07-30 15:59:16.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-30 18:42:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-30 18:55:33.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-07-30 23:37:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb8abd44dbd66b7a0e06c6f6f31ab51d367ba5b6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-30 06:20:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c22446b7efd3a0e6e927b86808e15985ec2f921c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-30 06:20:41.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"3b6119af90e4edf88cc6808df945f6946a0afe0a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-30 06:30:20.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"28725ea6d896db5570c27f69faf3e6cb5dd8a1c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 09:34:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a0993fed12964e8c90aa88a7b2fd0e55 : SUCCESS in 1h 06m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/93c94486bf824ab993a9c876c2b65a61 : SUCCESS in 1h 26m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/193bd78fc838482ba1da307cdc9bf9ad : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4ce0ef34c8346058c4ccf7a566d4815 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1b7d6258acd44cc97e0c49a1b6740ff : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07dac5f9ef314519b73cd27c4752038a : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1062f352e449f5b966d6606f7de769 : SUCCESS in 6m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e4d100b18ed4563b013120a03ed135e : SUCCESS in 10m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f2ce9029c4f4622a41ad83498314f5e : SUCCESS in 6m 46s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d6099a45f4ad411ca22b28bd29ab92ca : SUCCESS in 5m 36s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0dfd6f3bb00741b78958685ed7b1e662 : SUCCESS in 1h 04m 28s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/efa9ae250e3d4c848fa64c45cbc530ce : SUCCESS in 1h 19m 06s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/25817f91b8f0482baaf75c5bbb208b76 : SUCCESS in 1h 16m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b710f54df58422a959d70b32f583fe3 : SUCCESS in 37m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5c10a777693cf6bfe3689e70e9648b83429d517d","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-30 11:55:56.000000000","message":"Patch Set 1: Code-Review-1\n\nThese shouldn\u0027t be necessary unless we actually do have something to change for db.","accounts_in_message":[],"_revision_number":1},{"id":"e46905b008b453696e1a5ce49606431902e1c83e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-30 12:31:01.000000000","message":"Patch Set 1:\n\nCould you please ask Brian, he is the one who has written this test. We also did the same in Ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"2ed404568eaa1b17d69bb0554a8ba2cd69ffa8e7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-30 13:11:46.000000000","message":"Patch Set 1:\n\nIt was my impression from past releases that we had to have these in place, or else we would start to get test failures. Would be good to get clarification on that though. And maybe once we fully understand it, we should get that documented somewhere.","accounts_in_message":[],"_revision_number":1},{"id":"4e3691a0ae21edd183c4271895dfc17f176a5aff","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-07-30 13:53:35.000000000","message":"Patch Set 1:\n\nI think maybe this shows that we don\u0027t need the empty migrations? https://review.opendev.org/#/c/701041/2","accounts_in_message":[],"_revision_number":1},{"id":"d793784f0061477f132f48e92c9217f4f77e9ef7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-30 15:58:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a2fa4fd20a4f6f0aff0712bed90d91fe453f1fef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-30 15:59:16.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"e304c27594cf25b80ab75048bb87f991418c67fe","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-30 18:42:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6a69eed34bec713bc58aa28c1d0d7ecb72e23359","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-30 18:55:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b26b8048e2d5087b0b8c7812e03577b8bfcabff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 21:37:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0f92a03b542f463dbea53246f1db1c81 : SUCCESS in 1h 09m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b0f09e74100442e815b9c96024f4f3a : SUCCESS in 1h 04m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/321f690bb2334b9d976deebd45c5ae6e : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db1d228f9a1d42e7ae7395f3fd4dc312 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cb8bcfb34b34e4299f7e35ce56befa8 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cddd8f2f41f24c2197fc022a7d8a7f65 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6cdea715d8141a9a6c0e350265bf91c : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c13c9736f47e451e9f04afbd195413f4 : SUCCESS in 6m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e340ef510e741c88443a20a60ab278a : SUCCESS in 9m 08s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e1c547c4228142e1803347f629f8c92f : SUCCESS in 4m 35s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/01d62d5d86e54b77966cc62d5d1428e5 : FAILURE in 1h 24m 03s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/7371737bc7104093b82d2a13bfc5656c : SUCCESS in 1h 09m 37s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/84cf598bf15441aea6bfa1858bd95ffb : SUCCESS in 1h 08m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f474a78a360f47a08371f7d3a8c3cf95 : SUCCESS in 35m 51s","accounts_in_message":[],"_revision_number":2},{"id":"95bc0f915f5a80a097c3381a3de759803fa7a14d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 21:37:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2dc65f8b939731a96143f1c77856cea54805f6c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 23:37:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e0068d6f67c5456fa5178f2564005737 : SUCCESS in 57m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d45825d36a9d4191a6b53d6906e6a6e4 : SUCCESS in 1h 04m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2478c45c70144897af34d76e4539d8bb : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0559eccc806d4ec38875b42b31721d24 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9de600032efc44a3974204160ca5f4ed : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b8607e5714b49cd83002eee76d9a6bc : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32922a384f1b4876ae03376deff87f36 : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d619b8a86f7b42e280b6cbd1c8792b38 : SUCCESS in 7m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/10349d3abb314154bbdc8799acfec79f : SUCCESS in 6m 36s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/7563a600f1794253944f3f5a3c223035 : SUCCESS in 1h 47m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54fdbd29d8c54249875edfc089ea89a7 : SUCCESS in 1h 06m 33s","accounts_in_message":[],"_revision_number":2},{"id":"1d5129526b92de2914160aaf08d7e415ddb635d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 23:37:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"22981fccf12fdffdc10b325174af99dc3a3c0249","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 23:39:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a258f23bfc39497bbe90412949011df2 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2276f9a73f067937563b82992a20cbc25c608e7","revisions":{"691cf810b89f2cbaf576d97a1a86c121d9291467":{"kind":"REWORK","_number":1,"created":"2020-07-30 06:20:19.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/14/743914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/743914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/743914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/743914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/743914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/743914/1"}}},"commit":{"parents":[{"commit":"cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82","subject":"Merge \"Refresh Glance example configs for Victoria milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-30 06:19:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-30 06:19:36.000000000","tz":0},"subject":"Change database migration version to victoria","message":"Change database migration version to victoria\n\nChange-Id: I709a5d5b164031ffaabbf4164f4da4b080200543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/691cf810b89f2cbaf576d97a1a86c121d9291467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/691cf810b89f2cbaf576d97a1a86c121d9291467"}]},"branch":"refs/heads/master"},"f2276f9a73f067937563b82992a20cbc25c608e7":{"kind":"REWORK","_number":2,"created":"2020-07-30 15:58:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/14/743914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/743914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/743914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/743914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/743914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/743914/2"}}},"commit":{"parents":[{"commit":"cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82","subject":"Merge \"Refresh Glance example configs for Victoria milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cfbe5f7630c8b71c480f9a4bbbbf923c3e166a82"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-30 06:19:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-07-30 15:58:42.000000000","tz":0},"subject":"Change database migration version to victoria","message":"Change database migration version to victoria\n\nChange-Id: I709a5d5b164031ffaabbf4164f4da4b080200543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2276f9a73f067937563b82992a20cbc25c608e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2276f9a73f067937563b82992a20cbc25c608e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
