)]}'
{"id":"openstack%2Fglance~744011","triplet_id":"openstack%2Fglance~master~I6256a25d1907ec15fbdde560fc8eae37f3067213","project":"openstack/glance","branch":"master","topic":"bug/1889640","hashtags":[],"change_id":"I6256a25d1907ec15fbdde560fc8eae37f3067213","subject":"Fix active image without data","status":"MERGED","created":"2020-07-30 16:23:03.000000000","updated":"2020-08-06 00:35:56.000000000","submitted":"2020-08-06 00:34:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744011-1596674044711-1ab9881c","meta_rev_id":"f2c0f584b0068adadcebcd82b19c212cbead2169","_number":744011,"virtual_id_number":744011,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-08-05 21:38:12.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-05 21:37:48.000000000","_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"date":"2020-08-05 19:54:57.000000000","_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 00:34:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-05 06:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2020-08-05 06:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-05 12:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-08-06 00:35:56.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 21:14:43.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-08-05 12:19:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-05 19:54:57.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-05 21:38:12.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-06 00:34:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4abee65491c88a915dcc5eef25687260f93aa209","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-30 16:23:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a9c991f07b6511c7a402214048b4873235c4326","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-30 16:23:35.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"a57604ec8ff7976a21b321719404c39756092f93","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-30 16:29:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9fe1681898625847d1aba49749301ed1e29d69ce","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-30 16:29:49.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"8e41428fbf587d2f17b5bfe4c5598e2c988cc444","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-30 16:52:23.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"75a059680e60d285d1dfe808e8250121266a8dcd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-30 16:53:25.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"e31c357281cb27d360e4f087a1a8c2b5fe4f7ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 22:46:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/610b80b8146f480bb5b3d4ac8c144eb3 : SUCCESS in 42m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88df36f681834b3eb6c5fa7d077ea10e : SUCCESS in 1h 08m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98ea1eb6c0de440a9bfe3184d3176415 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47fc27357ab144bc8f4a10d79d7cd34c : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b88700a219042b4b7eeee297284a56b : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/012b0389bf374debaa3da5bb08497487 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65137d36eb44d24930eae59f846695b : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/091016e6fda14f4eba5bd41a781d1a1e : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e86b573f54e94a5eb318042fcf183dfe : SUCCESS in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9f126db65f784c0a901f6b6a5309c5de : SUCCESS in 11m 47s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e0eb955086834741a72beb733d3bf31f : FAILURE in 5m 20s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/01376af96b6f45aab75cd997c341ab19 : SUCCESS in 1h 10m 49s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/6b5024b2811744dd9e1b641946ec4948 : SUCCESS in 1h 07m 57s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/24d1d50cf2f54d7b846c9d533901a483 : SUCCESS in 50m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89b308fe089e481b94bb0fa92cda093c : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":3},{"id":"8421cdc569a852a05491ec431c26f0f01299e0cc","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-07-31 00:25:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ce2b6adbc11ba9d7ad2b7d05e4f2e7fa1a59c6e2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-31 00:26:03.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"28ae65e8cb81bc2c45b5c83f21833f6e8c2a11d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 03:09:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8dd9c3b5dde1472fb5f6ca5ff79fdcab : SUCCESS in 1h 15m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d273092d1ac468580fe6f6ad96d8057 : SUCCESS in 1h 05m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0eaa95897cf844aa8aafe174cc586914 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d201fb0552438f97653fa509f8f7e0 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e638eaae9ea402a9d460da7471e4a40 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecf82a00d2104fbfa71e600f8572d7de : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2f291d96194d7cb091380a257355f4 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0494516184ce48739166e64957b1bc56 : SUCCESS in 5m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8f08dce5702c43c8ad4d5ad7d769cefb : SUCCESS in 9m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0b0c0f7c107c4a239397d92bb9dd232d : SUCCESS in 7m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ce05a89c2b6d4d679e98f7c91458a087 : SUCCESS in 4m 38s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/dce276fc4932420bbd4352ddc7c63945 : SUCCESS in 1h 39m 12s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/bcea0bb342a84c3294278fc8ad67d645 : SUCCESS in 1h 10m 53s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/41da7ec09bf74a7ab782343c963baaf5 : SUCCESS in 1h 39m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d375135800342a1bc5db6fc0ead1ac1 : SUCCESS in 1h 21m 12s","accounts_in_message":[],"_revision_number":3},{"id":"7b8bddf0a981078d5a76b4b72310c57d0d86ec6c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-07-31 04:57:34.000000000","message":"Patch Set 3: Code-Review+1 Review-Priority+1\n\nIMO its difficult to provide functional/unit test for entire scenario but we can write one test which will test execute method of _DeleteFromFS task","accounts_in_message":[],"_revision_number":3},{"id":"baed58d4eaedbc3b3c50eadb7e848f0c243e7f2d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-04 13:03:09.000000000","message":"Patch Set 3: Code-Review+2\n\nTried to add unit tests but it really difficult and needs lots of mocking.","accounts_in_message":[],"_revision_number":3},{"id":"cc7d5c5e8600dd9ba8e29af8694c4a0eddf980fd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-04 19:46:54.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b06cb66ab6ca82d867ed1af41b6b8c85d0e7a990","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-04 20:37:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0046dfb944af26a804fd0609a0468e4043cb7750","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-04 20:38:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b6c67975b00d9b474d6fea09560369223fea5046","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-04 20:38:37.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"19b61d1462bec3dc87642523066714ff929885f0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-04 21:14:43.000000000","message":"Patch Set 4: Code-Review+2\n\nAdded a test and fixed the bug that it turned up, but otherwise this is unchanged from previous reviews.","accounts_in_message":[],"_revision_number":4},{"id":"e4b6364742283c58590eb789bc43a0bce1d48881","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-04 21:44:22.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"85368ef89734e17925ebfe3f382f2f12d05b68eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 22:12:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9eede3b5e72c4ae69a0ee6699e1d463d : FAILURE in 1h 12m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a77020f8322f49cfa13d885d7e03e33a : SUCCESS in 1h 32m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28e16060448f410a922a0b34759abd1a : FAILURE in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a671c76d0a5c4bb48f29803b2a01c855 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce404984f5874d09aa5167cb6a04168c : FAILURE in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d75b4a017b194c6ead569a0c53530e5d : FAILURE in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293d1ff389c54c9cb5a9c8a263d36135 : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6571193e9d8404f9dd7eff28717a3bb : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/62e46c5dd30742f2a21867c7b2173a0e : SUCCESS in 6m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fbc976f86f5a4fa39e877bf618f683a5 : SUCCESS in 7m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4b500fe44a6644c69dd4c7043766b2ae : SUCCESS in 6m 55s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/048f8711d8aa45d980b786c9db81eaf7 : SUCCESS in 1h 08m 39s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/9acd29be09de4c2d8217a688c4eac30e : SUCCESS in 47m 13s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/3bef687abbc84de68fe3d3ac147d0c93 : SUCCESS in 1h 08m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf9e6f96890f47e081649af269f94eaf : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":4},{"id":"037d3012626ffb6f82a8d7499dbadf31b15dafcc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-05 06:02:43.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"45a8b81ef0129d0a4007ab52bb39e5958ae35350","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-05 06:03:16.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"30c2863165777873fdfa78c519411a7532278a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 08:43:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f6e115d699654823977f28068fb8b6e7 : SUCCESS in 53m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a4fa4ce5cea42179112711a93c2c657 : SUCCESS in 1h 27m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33552a5c85234dbebd41bb2452bfe483 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b8d03b5ecf54c54ab1ce7c6231c2be8 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b49dd286458749afb0d8f077ca356484 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b6a4e9699ea402dbe6b5eb06538e8cd : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ddb3526b7c843adb44baac63ef40e61 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/733cac62803046f1b912f820cdbbebe9 : SUCCESS in 5m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d771d14e94042e9bfe26fec182de838 : SUCCESS in 7m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bd5a7dc43f6044098df4b46c0db44b49 : SUCCESS in 7m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d4b927fc54724fad83b5cd7f3d70f3e0 : SUCCESS in 5m 48s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/67014837074a4b3bb2811c349608d125 : SUCCESS in 1h 03m 13s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d981b89ae1d543cdb3acadba4ef1101a : SUCCESS in 1h 09m 48s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/82e021195c804b5abe8648d335eb805e : SUCCESS in 1h 14m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77ff3ef2e148470b821d75c0e5ae7890 : SUCCESS in 55m 08s","accounts_in_message":[],"_revision_number":5},{"id":"e4b23d50d6b593b25ee75ef309e37b1349fdf241","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-05 11:57:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"21f2f095b802b79ec62aedd0cd5ed74ee63ac10e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-05 12:19:12.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"594cdb85acddfe95ecde1941f81c259a27f371e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 17:23:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"39c48987527adf3486a91d00e304cc93434ecd93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 18:58:32.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/79d718db825e4d2fac9685b3bd2084ec : SUCCESS in 42m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47d72fcebb154c1a9393d9f27bb52594 : SUCCESS in 1h 16m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dee72c3b2b7e4b85a58710f947a15b0b : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2fd748b218f478d9570ec1b745074c8 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23a15a4062964a2888b769b8b6b8b0ed : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82251047e7104a909de864fb7d5d35b4 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949f1f04007d48a08c0f8cce8d46f977 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c248b2579854395bca1485cad2967e7 : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/232efc54813a43268b31e14e29809fc3 : TIMED_OUT in 31m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eac15c9b813640abb61394ea4c0ffed0 : SUCCESS in 11m 29s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/b33cee0112de4bbbb0f50f58e4cb6241 : SUCCESS in 1h 30m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/602538cf0f5c46d7838fb2ffc4a82e49 : SUCCESS in 1h 14m 06s","accounts_in_message":[],"_revision_number":5},{"id":"ba5ab2bc8ce1fc74fbda67474ce59a1057628dc7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-05 19:54:57.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"fdc25038a6812fde1da01f4086d036cb3b076ef1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-05 19:55:17.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"ecd6f4314b4295b7a2d6d2f3896b25bdbb1acfb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 21:25:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/48a3b6970b2a48489c8f816084691d06 : SUCCESS in 42m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/509c7da4f99142df909e2d91189f9940 : FAILURE in 1h 23m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/971143a4237243539ef982394e1ec875 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b31a47454c49e3b53698189089e484 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c54781e8e56641cd9369a2d5f56bc1a3 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c5287012aab4e519ed1c1f98e0422a0 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/204f8617e3c842a7bd341188a3274d4d : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/356d1fbc970e47ec92ef9c64ae9cc7e5 : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5a55402a65b7424696b05d325e832d30 : SUCCESS in 6m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f18ce96a2fb0403b9a0ea6156bfe0cd3 : SUCCESS in 6m 26s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/539b01c589a14a0b806cb5c3b34b6f10 : SUCCESS in 8m 19s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/71d7d0cebc8145888ed2847fb00318bb : SUCCESS in 45m 49s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d1713d6d5e9342a7b1f8c52d19843718 : SUCCESS in 1h 05m 25s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/1bd03c4ff43244ebb4c5882cbac84e87 : SUCCESS in 1h 28m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0c8afdc30494952a4818045d9e27de9 : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":5},{"id":"bb4712c096f4c0e7bfb01eed57f321720ba5e4f4","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-05 21:37:48.000000000","message":"Patch Set 5:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":5},{"id":"8f171636ffa3377779679920ed9070ca8801a9f9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-05 21:38:12.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"70cf441cff719fdac3badfdec658a5d8021b1b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 23:03:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/461793b6378b44e0bda041119fe29c8b : SUCCESS in 1h 03m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd32bc7dbc9548de8e43934fe7e72710 : SUCCESS in 1h 13m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b4d5702fbcc45f2a90b59e7120cc994 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db1ab1069a234dcab444a30c3e02a535 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f51b23f7992433cb6d4662b1160aaff : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/270fa13e0bf74af0a6dca73446cbc591 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86957c778c6c4e45a5ec0852cc41473f : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d7c86b9eda2431ca76eeac40d26f357 : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/44ae3f05c49841088135ecfe64f601f5 : SUCCESS in 7m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1c8f8f493dd94469a2ca7f381b83e0f1 : SUCCESS in 7m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/83b72b81f37e4c9d95f0121212cf6c54 : SUCCESS in 5m 46s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/438f5b41dafa4386a3027daf2b25fbd3 : SUCCESS in 1h 19m 08s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/b9d6a498f76f442d8677dcfbb81defb7 : SUCCESS in 1h 16m 53s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/df7393b595074092ba9109b18ed86f82 : SUCCESS in 1h 10m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db1b5fc87d5b428fbc4e4fbbe33fb888 : SUCCESS in 47m 43s","accounts_in_message":[],"_revision_number":5},{"id":"9db5c296b924ac6f3435e0c4b2082a0bac026eff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 23:03:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"28dcc640d0a83f95bff4b3a89774d83596ded0e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 00:34:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9685b0142b1f40388d7c68d7eef5036a : SUCCESS in 1h 13m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d8fbfa7497b42c2861851c49cbccbec : SUCCESS in 1h 16m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/882c11dba69c4d00a4e02cde1ee00933 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dafd26fc4b794d46928b9aef3a9cb340 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edb99240dc684a97ba7672993136c546 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5404215dd2244ed935576de5f3edbb0 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57710f8df7e7444c8a42fec5990c4fee : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5054f818f5994ccea7f7c3a6bc27bc79 : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d650dadba034d93abc7bd228858c989 : SUCCESS in 7m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d278a8eecc5443fd99302dcb48f46002 : SUCCESS in 9m 18s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/a6fa2d30612d4cdfb3eda2d7390fd80a : SUCCESS in 1h 03m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4edaaf182841487885a870c87ba61c79 : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":5},{"id":"c1e1649ff7d130b784b012299e6df1c0312de599","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 00:34:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3d15a1aab2f569e9963336b62382c4e92f5b7094","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 00:35:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3ca3c7fe4d44d9a5b5e8d14eae3bc3 : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f4dfc40fe10463ab2ac613f69be7da4 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"b985e44f795340edfceed3b46abf9b4b28907cbb","revisions":{"d0cf474281aff9cd05510a2020e1aa94e7bd5c17":{"kind":"REWORK","_number":1,"created":"2020-07-30 16:23:03.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/11/744011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/744011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/744011/1"}}},"commit":{"parents":[{"commit":"6741174ea71ae997f444b8fcba5422aab9d15acc","subject":"Fix active image when all uploads fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6741174ea71ae997f444b8fcba5422aab9d15acc"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-30 16:20:36.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-30 16:22:44.000000000","tz":60},"subject":"Fix active image without data","message":"Fix active image without data\n\nFixes image import resulting active image without any data\nwhen the image happens to be deleted from staging before\n_DeleteFromFS task is ran which is supposed to cleanup.\n\nChange-Id: I6256a25d1907ec15fbdde560fc8eae37f3067213\nCloses-Bug: #1889640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0cf474281aff9cd05510a2020e1aa94e7bd5c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0cf474281aff9cd05510a2020e1aa94e7bd5c17"}]},"branch":"refs/heads/master"},"be46df37ead9475ecec8b6adfc9940fc15db8e8a":{"kind":"REWORK","_number":2,"created":"2020-07-30 16:29:24.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/11/744011/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/744011/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/744011/2"}}},"commit":{"parents":[{"commit":"6741174ea71ae997f444b8fcba5422aab9d15acc","subject":"Fix active image when all uploads fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6741174ea71ae997f444b8fcba5422aab9d15acc"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-30 16:20:36.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-30 16:29:07.000000000","tz":60},"subject":"Fix active image without data","message":"Fix active image without data\n\nFixes image import resulting active image without any data\nwhen the image happens to be deleted from staging before\n_DeleteFromFS task is ran which is supposed to cleanup.\n\nChange-Id: I6256a25d1907ec15fbdde560fc8eae37f3067213\nCloses-Bug: #1889640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be46df37ead9475ecec8b6adfc9940fc15db8e8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be46df37ead9475ecec8b6adfc9940fc15db8e8a"}]},"branch":"refs/heads/master"},"d9d73b322bacd05362263cf35d6a20fa9e37f8f1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-30 16:52:23.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/11/744011/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/744011/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/744011/3"}}},"commit":{"parents":[{"commit":"bd243b62d8546637a1218094a9d3f5b53a76957e","subject":"Fix active image when all uploads fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd243b62d8546637a1218094a9d3f5b53a76957e"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-30 16:20:36.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-30 16:52:10.000000000","tz":60},"subject":"Fix active image without data","message":"Fix active image without data\n\nFixes image import resulting active image without any data\nwhen the image happens to be deleted from staging before\n_DeleteFromFS task is ran which is supposed to cleanup.\n\nChange-Id: I6256a25d1907ec15fbdde560fc8eae37f3067213\nCloses-Bug: #1889640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9d73b322bacd05362263cf35d6a20fa9e37f8f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9d73b322bacd05362263cf35d6a20fa9e37f8f1"}]},"branch":"refs/heads/master"},"79f52a36b3dd654283a388cc2c89f92b1c108369":{"kind":"REWORK","_number":4,"created":"2020-08-04 20:38:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/11/744011/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/744011/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/744011/4"}}},"commit":{"parents":[{"commit":"d94f86ea22f841501d72a565662eabf4908b02ab","subject":"Fix active image when all uploads fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d94f86ea22f841501d72a565662eabf4908b02ab"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-30 16:20:36.000000000","tz":60},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-04 20:37:47.000000000","tz":-420},"subject":"Fix active image without data","message":"Fix active image without data\n\nFixes image import resulting active image without any data\nwhen the image happens to be deleted from staging before\n_DeleteFromFS task is ran which is supposed to cleanup.\n\nChange-Id: I6256a25d1907ec15fbdde560fc8eae37f3067213\nCloses-Bug: #1889640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/79f52a36b3dd654283a388cc2c89f92b1c108369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/79f52a36b3dd654283a388cc2c89f92b1c108369"}]},"branch":"refs/heads/master"},"b985e44f795340edfceed3b46abf9b4b28907cbb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-08-05 06:02:43.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/11/744011/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/744011/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/744011/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/744011/5"}}},"commit":{"parents":[{"commit":"e8b13ce1b44313a9604aa9cd8cdbf8bd4d1bda52","subject":"Fix active image when all uploads fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8b13ce1b44313a9604aa9cd8cdbf8bd4d1bda52"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-30 16:20:36.000000000","tz":60},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-08-05 06:02:21.000000000","tz":0},"subject":"Fix active image without data","message":"Fix active image without data\n\nFixes image import resulting active image without any data\nwhen the image happens to be deleted from staging before\n_DeleteFromFS task is ran which is supposed to cleanup.\n\nChange-Id: I6256a25d1907ec15fbdde560fc8eae37f3067213\nCloses-Bug: #1889640\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b985e44f795340edfceed3b46abf9b4b28907cbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b985e44f795340edfceed3b46abf9b4b28907cbb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
