)]}'
{"id":"openstack%2Fglance~744234","triplet_id":"openstack%2Fglance~master~I25cd3cde94fefaa5d8ac72f10a075fe34a5df7bf","project":"openstack/glance","branch":"master","topic":"bp/calculate-virtual-size","hashtags":[],"change_id":"I25cd3cde94fefaa5d8ac72f10a075fe34a5df7bf","subject":"Inspect upload/import stream and set virtual_size","status":"MERGED","created":"2020-07-31 18:02:42.000000000","updated":"2020-08-08 05:57:30.000000000","submitted":"2020-08-08 05:56:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744234-1596866162371-480d4010","meta_rev_id":"b364f5e0ca9fded89da0c9c65b680e367e13372e","_number":744234,"virtual_id_number":744234,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-08 03:46:32.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-08-07 23:35:57.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-08 05:56:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-05 21:37:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-08-07 07:35:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-08-07 07:35:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 21:37:00.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-08-07 07:35:56.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-08 03:46:32.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-08 05:56:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87a503844f92ee934fd5e8d077c6c615cdcd618a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-07-31 18:02:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a43fe06922fdab63d2812ff95f87acc56ae6613","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-31 18:03:10.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"0c37372e8f48cacfaa9a0d192246017ab059ade7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 00:27:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/29365a63b01345d5bbcaeddacbb8e022 : FAILURE in 1h 21m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5dc9764d988c4d45be104c2e082428e8 : SUCCESS in 1h 38m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdaba81071414ab98968addb3575c5c2 : FAILURE in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a761a0b8c6346cdac48fcf2d5e36b75 : FAILURE in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1a809567c4648c696fc114589bc4495 : FAILURE in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6be26bbc9faf49a9b084313b060139ac : FAILURE in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde48d41a1fd40e9a39063bad48c04eb : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6733c6174eb4437b64cb9478888f5e9 : SUCCESS in 7m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d3b08ec7331d4104981d55e08578dfd0 : FAILURE in 14m 47s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f7ce84a1c1e04133ba6b86891e17f8fe : SUCCESS in 4m 36s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/079a967ec7234f6682b856c95e1c7b33 : SUCCESS in 1h 11m 28s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/21d6c43601694dcd847f1b34697c13bc : SUCCESS in 49m 56s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/e572d79e0cb24759be6a71766320852a : SUCCESS in 1h 11m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0387116edbb54eca8e959b4b06c12dfe : SUCCESS in 53m 19s\n\nWarning:\n  Comments left for invalid file glance/common/format_inspector.py","accounts_in_message":[],"_revision_number":1},{"id":"2e08a82ce8480233c9f7377383869a042bd19cc7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-04 17:30:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"568a09c4392e372a5d654312456714301526f87d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-04 17:30:35.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"20cf54efd471dd9c617891edf14c75dd201aefc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 18:39:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab41b1662e8e402da60a5b2db5103900 : FAILURE in 49m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88eaa55aa91946aa8d107d81eba1be88 : SUCCESS in 51m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10da8d3e3ed64627b2bc3d5fb22e9be4 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53d3ba3b0274df1acc8d10920b1acdc : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eba01140e29542bfa391c670bc6170bf : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12050e7e28d14bedb54b7fb0f65be063 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58133fcd47045e6b33c66d57aa083df : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1449fa454bf74ad1ac49630a3d32c47c : SUCCESS in 9m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5959e6028479475b8650517caf0738ca : SUCCESS in 7m 52s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bd2ec2c4774b46929729eb630ec1be77 : SUCCESS in 5m 06s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d4e4d6dc270949bfac1b9da80e30ad75 : SUCCESS in 46m 55s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/9d8671e91e984b71a622dd195bd8b08d : SUCCESS in 1h 07m 44s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/8042d84314474bfd959d8b44ce9971d5 : SUCCESS in 47m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cb9dfb11fab4540840bcbb4c3c35260 : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":2},{"id":"d1242415e3a810646cd540fc27ba2b529bef574c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-04 19:38:26.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"35077c1804fb6fde482a1faff801954c1ae703a9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-04 19:38:59.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"98a1c135123491c7c0155929679d0229ab393778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 20:56:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3f22b4a9d4ea43588e7009d3f940da94 : FAILURE in 47m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6e2447a092e4249be0b5a9d2a4ce304 : SUCCESS in 1h 03m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c4c1272402b4ea2b7fa94afbdb97a3e : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce80083c4a54daeb0cd542b7defcdd8 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d57d79203984af49ac408b590c0632b : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a77934146be426ab61904dcf505852c : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d177131115f0408ea9edfe155fd019dc : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2a24e57fbe994e659cbbcd4842000738 : TIMED_OUT in 31m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8bb013e197484aad8535d02b05e3ec7e : SUCCESS in 7m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/84335a5a3e6b45588e637c4d599ea5b1 : SUCCESS in 4m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/65b361e90e99449397003afd84c62e93 : SUCCESS in 1h 00m 24s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d005be39d7ad4a27907ef240a6a549f7 : SUCCESS in 1h 12m 20s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/f60c897bfbfd461f96ad01002b5af7be : SUCCESS in 1h 06m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b9104ceb78b4e9e9df4c500d4017e62 : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":3},{"id":"4a41081d39bdbc7e12a0828964a68da13309335c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-05 14:43:53.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"a8b511d6edeaefb2229a4bff2dfeddb12dfcb593","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-05 14:44:26.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"a212f177284323e9d985c918fc0bb8912a8ce792","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-05 14:55:03.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"ffd33d1e2717862654d8563db47d877ae77faa67","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-05 14:55:36.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"cd1416fe4928e7a533370e82e4ed39169597f683","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-05 15:23:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b35fe33d5965131089546ed824f71bafdbc05ec3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-05 15:23:25.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"167821001bf602ee08286ff31b0d81e7f27e095f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-05 15:42:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b11f45e5f31b84fc8b4b253e91470c101599398d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-05 15:42:53.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"04694f9f83b31ae1522db48e123bf64523251b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 19:53:32.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/63298efca91c4b0ea5125e7aac553e29 : SUCCESS in 1h 13m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bbf614221db54a1b8cadaa373e79606d : SUCCESS in 53m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b5fb28fd7a649f9ad587e5a7eca2e2d : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47944dd584824c88b31acca3ffd19676 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/716af247972442a094f67d6cc3ce828a : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2e48841c16e431480c74a386e2ea868 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5cf005971a5439d911c2bb38768bb9f : SUCCESS in 7m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa3098aa98ae4e51a24f999b05be5fad : SUCCESS in 6m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/248ab83955ec40a986699026badb5c09 : SUCCESS in 21m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d5e3c2505c824ab89f91aab602e01144 : SUCCESS in 6m 18s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7610daebf4f24e918a275bb8af2924e9 : SUCCESS in 1h 22m 12s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/2da4d20396e84d6bbe95fd4719f298f8 : FAILURE in 1h 09m 38s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/2b9fa6f11bd54592bb67604cc333ec3a : SUCCESS in 1h 30m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/322e4d25f1ae4f1ca27017cf408968ba : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":7},{"id":"2c030bda43a11370ace58c5d4a60c4363e9804d4","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-05 21:37:00.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"84c02d4779cb055ad4871a068a66b769a85ee630","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-05 21:37:39.000000000","message":"Patch Set 7:\n\nrecheck same cinder fail","accounts_in_message":[],"_revision_number":7},{"id":"617fe44841fd297cc16c4ac5a6ae7aba5fed0c66","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-05 21:38:05.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"d73638cae1550919fc557ba0495cbdaf20319822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 23:05:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2af45c6c6f31430cb952f620236df9cf : SUCCESS in 1h 04m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/234fb355f6564bfea17aaa1780eacb03 : SUCCESS in 1h 16m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4c4ea3962d848a58c9ed403b1603bfe : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1274035f769240fdbcc189afe70c79c0 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef48e04d54184ac8a2503ac47798600f : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20b8a4702e1e4463a0b3e9d09318043c : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82ad6b6a30334ee1a90628ff00e3c51f : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/63709ebea1f140419b4a4b95bd65f466 : SUCCESS in 7m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/70e00e8926f942aca65f133097a5fbb8 : SUCCESS in 7m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a0a96e09f30b42d787786a05a9aaf431 : SUCCESS in 4m 52s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d06d828b835143889348b20695d72f20 : SUCCESS in 1h 05m 15s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/fb540d81f5e74499ab152177e0ff78f7 : SUCCESS in 1h 15m 23s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/3ea24a4a84994fadb3c43cc8723ab4fa : SUCCESS in 1h 21m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb412180635b41d9ae725bb4d6df49c0 : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":7},{"id":"1ffa499e138f1b79f68450831ca3678222283d78","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-07 07:35:56.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nVerified manually, working as expected and charm!\nThank you Dan!","accounts_in_message":[],"_revision_number":7},{"id":"94185c6ada4f7e4231452de1dbaa87c3fa7cc21e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 07:36:10.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"fdbc41ad57a1380115feb7581abeead7952f6a3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 09:09:00.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e7471065ccd743e79b732240286ab97f : SUCCESS in 1h 30m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ddd90a4ed3a4aefb5c758fd4bcbd1aa : FAILURE in 1h 15m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08ecfdc376e547a39b29aff1f2262278 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f62bfe5346084157a274d0c7d0758702 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3428583585134859aeec960495bdcb85 : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba07368c936c48c2bfcc124df4ed67e6 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c94dba5dbe9463c9aac519c06fdfdaf : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4ac9ea40243b4056a68af928e4cf9936 : SUCCESS in 9m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca64b4875fae43ad9b2fa8e90cd18a6a : SUCCESS in 9m 33s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/233b3ade189f4bc08dfe8526f602841f : SUCCESS in 1h 07m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a34dc528c16e49bc9ba61b65610918ff : SUCCESS in 1h 01m 54s","accounts_in_message":[],"_revision_number":7},{"id":"62115f0b7fd85a0ad62e5d7662f50a6a63240411","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-07 10:02:15.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"8744016dc4958cd40b771959119553378c0d7122","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 10:02:37.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"78bf90cec20d77393e4aac79d0cf5c799efe40ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 11:45:03.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/09b0c1d7a830445e95ce58b45895280e : SUCCESS in 1h 03m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3fc9b616db3343b38ae433737c6aaadc : SUCCESS in 1h 08m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/562417210a474a70808e89ab751ed584 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f7f4e94f4f24cbfa8684208d0172d08 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfba143e45cd42288dcd939dd40e1786 : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1706f344322d4af2bac98dd693ce7501 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa70a82a094485789bae1ad8e14c40f : SUCCESS in 8m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/75a986fcec974ad4948a57f8b637fc57 : TIMED_OUT in 31m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e998c082b2b14fe3ac9c2b1ee6f3e63a : SUCCESS in 10m 28s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d3dca6d67af94b9fa8d7fd86a8a8b25d : SUCCESS in 8m 30s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/38b7adbe5cd9449d810d4c2471f5ed74 : SUCCESS in 1h 04m 33s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/63427f15a19d400fb87cabd8c009f1db : SUCCESS in 1h 07m 12s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/bc612fd362654290b4aa2474d9c2ae4a : SUCCESS in 1h 13m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ef88af6c1b940d892243c03e0d351cc : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":7},{"id":"6139fb40ceaaf2b565e768258be81469a10a7eb3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-07 12:12:55.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"f29771a8c8b98fbae2ffcced706d1a40f4275740","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 12:15:01.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"7db2e28f7a2212f73373faa9b43e626e9ceddbe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 13:52:37.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/139e6c8ad5fd49fb9d1ffc475b5a6b14 : FAILURE in 44m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f374c8a8a8e4a7da55708bd64f70042 : FAILURE in 1h 21m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e645e5a708a140c2a0841338a4325d9f : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a5ce82d610489892a9288ce612f815 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c0d662cc7da4991910fc47c2b30e669 : SUCCESS in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1169c0db75094619ba7ce0c542701942 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45576338c0274cc0b05f9d260dea2c83 : SUCCESS in 9m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e57f5758bd94238abb8697b8feab51c : SUCCESS in 13m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7d4c2441dc8c41a8b766dbc005606e75 : SUCCESS in 12m 23s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/75d1bee30b784378a7af9a3211912016 : SUCCESS in 6m 00s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/07d16d26119e45c08c66995e8e6778d1 : SUCCESS in 1h 35m 33s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/477b52f5df324299a7348aa749b5ea9e : SUCCESS in 1h 37m 02s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/c1d978ae2a3d4b0f8b008bab3bc0e30b : SUCCESS in 1h 37m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3938e365a71445cace159b6350e757c : SUCCESS in 35m 33s","accounts_in_message":[],"_revision_number":7},{"id":"d2c6f956ebf1651c57deab5e53a30f5f380a09b4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-07 14:33:24.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"aa5579be753cdc6a606ed86247ae8f337ca70e4f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 14:33:45.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"9198a91ad5ffdb2cd76f0554cb0c6d11fb17e994","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 16:46:11.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/fadea7a764a84913b0340d83dfb83165 : SUCCESS in 1h 07m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/64ede63e8f5c4ed88e710103ddc9988d : SUCCESS in 1h 30m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fcd3d51934b43ffb4cdec5bb0b4b880 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2386cb54cbaa47778b6f72c8e9d29d47 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df0e2023bfff47cb9f115f9f8a01995b : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0bd131daa3c473895fdd7d21caa4393 : FAILURE in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5522e3f4bff4bcbb22d9d6138624f48 : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b9267b2fd4f400b91c146c822883d86 : FAILURE in 3m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1b0c8cb256c246439c011b367cd84924 : SUCCESS in 11m 26s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d15621f45b184791b685a685f62acdd2 : SUCCESS in 5m 18s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f924e9cb922947fdbc85b3200bda35a8 : SUCCESS in 1h 31m 42s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/42926fed15e648bebbfa7321e05959a8 : SUCCESS in 1h 18m 15s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/ddfe6606a7aa4f19b947c1647cc49c03 : SUCCESS in 1h 20m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aedeaa1d73e14116986e2e892f8c211c : SUCCESS in 1h 05m 53s","accounts_in_message":[],"_revision_number":7},{"id":"d900dc1bdd129354bbd4e56f65f52b369848fd9b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 19:08:17.000000000","message":"Patch Set 7:\n\nrecheck pypi mirror problem","accounts_in_message":[],"_revision_number":7},{"id":"7129f3f8c2bfe6046ea7be170eb7fb9f35597f05","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 19:08:37.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"0893b919c3af0742715055b12eff3159b593056f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 20:42:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/67e083a175c04910b92ed4338f1429e1 : SUCCESS in 1h 12m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fbee4dcdc5ba43469ac8813f7b2373ad : SUCCESS in 1h 23m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21a965ff718f4c87bc2bda4685ee6ba9 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc95b1e74144a5a821fa557df6de839 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/311bc71bb43e4c2e9378fcf3cda5d9b6 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5cf72effc0b4d32bcc3893eb5d7ead6 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6197264420ae4783890c6d28714a02d7 : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0498f21fd8e445409f64dfc82d6fb648 : SUCCESS in 14m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/51049217bec7402c814c39dcb424a123 : SUCCESS in 15m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ef5bd9fc875744a6a8cd5ff7d0d529f9 : SUCCESS in 7m 12s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8c98c641a144452eadf5ef9143d649a4 : SUCCESS in 1h 32m 30s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/6e97b2a0a3a4413caf6d8163a15e9d24 : SUCCESS in 1h 27m 35s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/c1e0dcf9cf4243969a0b299f24599983 : SUCCESS in 1h 26m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8919d9b2d5d3480ab8fb7f197dfa7f08 : SUCCESS in 1h 10m 54s","accounts_in_message":[],"_revision_number":7},{"id":"8f14bd98846c19e74d16ee654ff2225321a7814a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 20:44:29.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ff8ffcd71b6902b17c390ede9c6a9a167230d11e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 22:49:29.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0c091b3d839f4693ac8edf1bb3c61eba : SUCCESS in 1h 01m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c892f25ec73400b8698755fd8c45ace : FAILURE in 2h 00m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4498a56ffa94c07a0b77aeb3c36baac : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f83f876a6d4515bb892af5e156cc0b : SUCCESS in 11m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c49cee7f7aa945a0bd3211e6656115d6 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc75c791c4d4489cb377a4d0471adb84 : SUCCESS in 12m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5305bcb64b40dc81fc1c171aa8817e : SUCCESS in 12m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/96d1088eb45d4205ba5c54864579894d : SUCCESS in 22m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5bddccc514e446e19b8afedce11c0ece : SUCCESS in 22m 19s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/ca8bb5f067de4cb7af7c52f367ae85b7 : FAILURE in 1h 58m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c923a0fa17ba4d63822c3e69010cc89c : SUCCESS in 1h 49m 15s","accounts_in_message":[],"_revision_number":7},{"id":"d596cf91d4797647cd18eab9206e5107cfe6cf06","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 23:35:57.000000000","message":"Patch Set 7:\n\nrecheck functional fix merged","accounts_in_message":[],"_revision_number":7},{"id":"14e460252adae252d188f877b8dde5181aa9184a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 23:36:19.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"371fc1f6599f61949f108720701edfc5116ae7aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 01:02:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/41dc915455b6487381afc6dc5cbf6917 : SUCCESS in 56m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99ebc434312f40388fdeea85d544a3f5 : SUCCESS in 1h 06m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b56ed161cb646a78ab662ef3dd95e26 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5daaf9d3e5748869bcaa9e9685718f2 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/005849e159db4ecd92deb85b68728be6 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7919d9a6425432884dc31e1406b173f : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a28b3395fad4dce983e2b3ca09c366a : SUCCESS in 10m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be987bfce3e24b84a1e3721f28435502 : SUCCESS in 7m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d27f4f97021b4d54ae2fd137134ec0be : SUCCESS in 8m 28s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a7995567f3184155a4a08c93ba8e7c2b : SUCCESS in 4m 18s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bc132adb77964f3f8a48b34ad3c220c9 : FAILURE in 1h 14m 16s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/e3e275e6ca584976a4a3b6c273fd0c47 : FAILURE in 1h 24m 19s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/c6680fbad85948919ca1fcc8280a7d41 : SUCCESS in 1h 21m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b70ecae2fcd4c34a0ace26f244b21d8 : SUCCESS in 50m 31s","accounts_in_message":[],"_revision_number":7},{"id":"bf4d93353c14573bfef5ed808d60912db4991f85","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-08 03:46:13.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"1a7226d52962fe50aca59f589f83d9d6785f2f63","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-08 03:46:32.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"54f6b103bd3053934684ac153d8f467ba3673deb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 04:52:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/854d44a39945466b80ec83489476f9a5 : SUCCESS in 1h 00m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dcba84197a4c410ebb418daa5ed2cb78 : SUCCESS in 1h 05m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e3424c41ee04989aa0973b808c6bc5c : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c2d2ce41d81483b81c11867a9062d51 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a06d5b2829b44ba793b0a272c37344f8 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/487307256db64d5aab892e48d4317d14 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9658f923c3843b0a8bfd4675bb43e84 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e0d14bf3f09408ea1485e8847a23d4f : SUCCESS in 8m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f0894f9f00914dd7904654a14b961095 : SUCCESS in 8m 52s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3f1d7506bbce44afbdecc7678cc4da68 : SUCCESS in 4m 03s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5ad0189388a94133bb6d0696bac02a5a : SUCCESS in 54m 30s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/c1f3f61f6c9447f684a34ad673311815 : SUCCESS in 1h 03m 04s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/8e75d8539dee4fb9b8bdb3e7cacf6a38 : SUCCESS in 1h 02m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f0428415af2415d8aa50d0636fa72c9 : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":7},{"id":"a04bf68683a98f8470114fa24b906863977dbdcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 04:52:42.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7a39a2823b06b92b22439454d8cd7c11d97bb4af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 05:56:02.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7a48acc1b7844b5b9fe7cfaa90205841 : SUCCESS in 59m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f1933f8996214c7e9ac4070a1d498699 : SUCCESS in 1h 02m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c27c411037c4bde87525db8b3be12fa : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d684f370dcb14ff182432034eae2cfc3 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e431da96da9b424f9eadf3678811ae97 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1c936435c9e41cf988831ed2f98c53e : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f8ea0b4c1b945bb9ca6e1c93a8332ae : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b54348c35b749e99c35b7d5758310f7 : SUCCESS in 7m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a45cd9ae73da4477ae5c259e5302a90c : SUCCESS in 9m 35s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/93c8b2cecb8e4ae1a403a9b52aacb9a7 : SUCCESS in 1h 00m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0799871fd63945ccb6ecb9184dc5eb7a : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":7},{"id":"260328ee93b9ea969f3eb42b74ca8076d7d82709","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 05:56:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"5bf9d0535b6e0b799068550a35151d7ffe13b7cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 05:57:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05026a1dd5194ca28f2ecc7126edc3ed : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b0197656684e3fa8445b859434fe29ba44499fa1","revisions":{"04b8714fb5521ef55dd87e5df32d4ddb02b3f4f3":{"kind":"REWORK","_number":1,"created":"2020-07-31 18:02:42.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/34/744234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/744234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/744234/1"}}},"commit":{"parents":[{"commit":"2461aa35e8efc64057cf754168628437722ebfd6","subject":"WIP: Stream-friendly disk format inspection module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2461aa35e8efc64057cf754168628437722ebfd6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-31 18:02:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-31 18:02:15.000000000","tz":-420},"subject":"WIP: Inspect upload stream and set virtual_size on upload","message":"WIP: Inspect upload stream and set virtual_size on upload\n\nThis is just hacked into place and tested by hand.\n\nChange-Id: I25cd3cde94fefaa5d8ac72f10a075fe34a5df7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04b8714fb5521ef55dd87e5df32d4ddb02b3f4f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04b8714fb5521ef55dd87e5df32d4ddb02b3f4f3"}]},"branch":"refs/heads/master"},"fbadbff38843791e7ef09ea0637994b59b4a22e2":{"kind":"REWORK","_number":2,"created":"2020-08-04 17:30:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/34/744234/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/744234/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/744234/2"}}},"commit":{"parents":[{"commit":"5d21bdd3c8962a0f76c7f25e376ec3d488fcbe9e","subject":"Stream-friendly disk format inspection module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d21bdd3c8962a0f76c7f25e376ec3d488fcbe9e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-31 18:02:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-04 17:27:11.000000000","tz":-420},"subject":"Inspect upload stream and set virtual_size on upload","message":"Inspect upload stream and set virtual_size on upload\n\nIf we are processing a disk_format that we know about, wrap the\ndata pipeline with the format inspector and set virtual_size\nafter upload is complete.\n\nRelated to blueprint image-introspect-plugin\n\nChange-Id: I25cd3cde94fefaa5d8ac72f10a075fe34a5df7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fbadbff38843791e7ef09ea0637994b59b4a22e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fbadbff38843791e7ef09ea0637994b59b4a22e2"}]},"branch":"refs/heads/master"},"44e9ff939d5acfa4d0ec7f3cc817b5324ef59947":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-04 19:38:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/34/744234/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/744234/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/744234/3"}}},"commit":{"parents":[{"commit":"aeba2a4222ff29dca4a777b614bfe72486b7aedd","subject":"Stream-friendly disk format inspection module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aeba2a4222ff29dca4a777b614bfe72486b7aedd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-31 18:02:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-04 19:38:19.000000000","tz":-420},"subject":"Inspect upload stream and set virtual_size on upload","message":"Inspect upload stream and set virtual_size on upload\n\nIf we are processing a disk_format that we know about, wrap the\ndata pipeline with the format inspector and set virtual_size\nafter upload is complete.\n\nRelated to blueprint image-introspect-plugin\n\nChange-Id: I25cd3cde94fefaa5d8ac72f10a075fe34a5df7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44e9ff939d5acfa4d0ec7f3cc817b5324ef59947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44e9ff939d5acfa4d0ec7f3cc817b5324ef59947"}]},"branch":"refs/heads/master"},"c66e09596031a2dc139bc7f130d2b10f4c966b7f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-05 14:43:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/34/744234/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/744234/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/744234/4"}}},"commit":{"parents":[{"commit":"4e5825736dc900290fbc13f32b6f34c3ffd7caac","subject":"Stream-friendly disk format inspection module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e5825736dc900290fbc13f32b6f34c3ffd7caac"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-31 18:02:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-05 14:43:46.000000000","tz":-420},"subject":"Inspect upload stream and set virtual_size on upload","message":"Inspect upload stream and set virtual_size on upload\n\nIf we are processing a disk_format that we know about, wrap the\ndata pipeline with the format inspector and set virtual_size\nafter upload is complete.\n\nRelated to blueprint image-introspect-plugin\n\nChange-Id: I25cd3cde94fefaa5d8ac72f10a075fe34a5df7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c66e09596031a2dc139bc7f130d2b10f4c966b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c66e09596031a2dc139bc7f130d2b10f4c966b7f"}]},"branch":"refs/heads/master"},"d38c54bd3a8a66a1d9a267d198b25f5dfe81c25e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-08-05 14:55:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/34/744234/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/744234/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/744234/5"}}},"commit":{"parents":[{"commit":"48d408eeb9abaee9d0fc409f7f0c101244396f4e","subject":"Stream-friendly disk format inspection module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/48d408eeb9abaee9d0fc409f7f0c101244396f4e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-31 18:02:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-05 14:54:57.000000000","tz":-420},"subject":"Inspect upload stream and set virtual_size on upload","message":"Inspect upload stream and set virtual_size on upload\n\nIf we are processing a disk_format that we know about, wrap the\ndata pipeline with the format inspector and set virtual_size\nafter upload is complete.\n\nRelated to blueprint image-introspect-plugin\n\nChange-Id: I25cd3cde94fefaa5d8ac72f10a075fe34a5df7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d38c54bd3a8a66a1d9a267d198b25f5dfe81c25e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d38c54bd3a8a66a1d9a267d198b25f5dfe81c25e"}]},"branch":"refs/heads/master"},"e1b1a6b3e3cb5377802fea1ff6ede25a569245e1":{"kind":"REWORK","_number":6,"created":"2020-08-05 15:23:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/34/744234/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/744234/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/744234/6"}}},"commit":{"parents":[{"commit":"48d408eeb9abaee9d0fc409f7f0c101244396f4e","subject":"Stream-friendly disk format inspection module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/48d408eeb9abaee9d0fc409f7f0c101244396f4e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-31 18:02:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-05 15:22:56.000000000","tz":-420},"subject":"Inspect upload stream and set virtual_size on upload","message":"Inspect upload stream and set virtual_size on upload\n\nIf we are processing a disk_format that we know about, wrap the\ndata pipeline with the format inspector and set virtual_size\nafter upload is complete.\n\nRelated to blueprint image-introspect-plugin\n\nChange-Id: I25cd3cde94fefaa5d8ac72f10a075fe34a5df7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1b1a6b3e3cb5377802fea1ff6ede25a569245e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1b1a6b3e3cb5377802fea1ff6ede25a569245e1"}]},"branch":"refs/heads/master"},"b0197656684e3fa8445b859434fe29ba44499fa1":{"kind":"REWORK","_number":7,"created":"2020-08-05 15:42:20.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/34/744234/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/34/744234/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/34/744234/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/34/744234/7"}}},"commit":{"parents":[{"commit":"6a2a6283ae01c5195beb949a8a7b607199f24396","subject":"Stream-friendly disk format inspection module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a2a6283ae01c5195beb949a8a7b607199f24396"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-07-31 18:02:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-05 15:42:04.000000000","tz":-420},"subject":"Inspect upload/import stream and set virtual_size","message":"Inspect upload/import stream and set virtual_size\n\nIf we are processing a disk_format that we know about, wrap the\ndata pipeline with the format inspector and set virtual_size\nafter upload is complete.\n\nRelated to blueprint calculate-virtual-size\n\nChange-Id: I25cd3cde94fefaa5d8ac72f10a075fe34a5df7bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b0197656684e3fa8445b859434fe29ba44499fa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b0197656684e3fa8445b859434fe29ba44499fa1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
