)]}'
{"id":"openstack%2Fglance~745100","triplet_id":"openstack%2Fglance~stable%2Fussuri~I6256a25d1907ec15fbdde560fc8eae37f3067213","project":"openstack/glance","branch":"stable/ussuri","topic":"bug_1889640_ussuri_2","hashtags":[],"change_id":"I6256a25d1907ec15fbdde560fc8eae37f3067213","subject":"Fix active image without data","status":"MERGED","created":"2020-08-06 09:18:50.000000000","updated":"2020-08-06 19:18:28.000000000","submitted":"2020-08-06 19:17:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745100-1596741430867-ff0595dd","meta_rev_id":"8fedbb07b3cf67caa9edbaac2df781d3177b809e","_number":745100,"virtual_id_number":745100,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-06 14:46:00.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 19:17:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-08-06 17:36:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-08-06 13:38:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-08-06 17:36:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-06 19:18:28.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-08-06 13:38:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 13:38:07.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-06 14:46:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-06 17:36:53.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-08-06 19:17:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09c07c7d48b2c1e7ab9592b1e3686d786c0c8594","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-06 09:18:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac1786992f95473e71881332cdc35f3a19d9d01a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-06 09:21:43.000000000","message":"Patch Set 1: Code-Review+2\n\nClean and required backport!!","accounts_in_message":[],"_revision_number":1},{"id":"10e7b51cf5a12dac1e7741882cc2cb028502df37","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-06 09:48:33.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"6f3f142e7c3cf895329d577560402b52a306e9c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 11:09:46.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a148ea82f67a4efc92d5627b8e5fbffd : SUCCESS in 1h 14m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99b36d97f5d941c39a33efa5d63ba80b : SUCCESS in 1h 16m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6765262556445c787b44d096fb25893 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61b78b9576c24d0598935c879325e558 : FAILURE in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3805d601f7ec4044a0743e5ec35c1324 : FAILURE in 7m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a7712035ea3411b8da61d4b9df01c33 : FAILURE in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/805d022b9a844bf1884a92d613340838 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30feb680afbf466ca7f18a070fd0b613 : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c05ad51f8edc405c8c3aeaddaa54702a : SUCCESS in 10m 38s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6ba7d64577b24de68e2a869000f251d1 : SUCCESS in 8m 19s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0d62a06601da4133ae9e094f20e45772 : SUCCESS in 5m 11s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/87f276065584493e8deeef93a53ed1f9 : SUCCESS in 58m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed75da13947c4a1e847ee90d87ec7530 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8e6c99c3990bfec9111f6c092250d1c5dbc28363","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-06 13:10:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5363cc5bfb6e162519de129eabf9d652c6b9c6b7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-06 13:38:07.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"966c9b840444d681ea0df29fe9af3c2cd499e045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 14:44:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c471b29482994efeaa0d876d258efa09 : FAILURE in 1h 31m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3063b7b6c2214a72982d7559c778299d : SUCCESS in 1h 10m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c456201e06814b8ab25f0cbf9a44faf6 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580471492b9a444381c2c01168d90ea9 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/276c02cdce9c46a8a39918751fd7fd18 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98dc06021e374977afb6ed03e478981a : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2c7d343b0e49c98e0793e4a423e99f : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4229e9c15ba94ed59e2637a28e145bd3 : SUCCESS in 4m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e7853fc06c9426ba14f089928a448f9 : SUCCESS in 6m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/0a8815cb9b9e4fd69acadb4e40c926bb : SUCCESS in 8m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/88d04e86c6334173bbbe891fb3eaa858 : SUCCESS in 4m 14s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fe44daf4896941839f76915c2c375d67 : SUCCESS in 1h 07m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33ef2f6426b54e50ae4a079159675637 : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":2},{"id":"5b701381fd1f5a9277061aebc2d338df7496d082","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-06 14:46:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"131601c4568334166c5453cb1bd697717f2b45c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:57:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/16e089bdec534cafa8cee555f9477f06 : SUCCESS in 1h 17m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b426d4d917f343559552bff30afcfef5 : SUCCESS in 1h 32m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ea94cf8473c4ecaa55dd15faa0b3169 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d09b958d414b909a1be27df884d9d8 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd8eb4a96a5f44b48a520ffeba81c923 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f2a5fece5044515b8cfbe7e6466a4df : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe49b16b511f4c9d83bcd31e12cd2a2d : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/254cffaef47747cb97fd3550faa2bb09 : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a10ba3d915c427cb27baf88ad134709 : SUCCESS in 8m 34s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e053e61d25dd466786b5789f5caa0f54 : SUCCESS in 8m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/310055556d1042a09c7143e952e5608c : SUCCESS in 5m 05s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bddec00f924b4b818fa037dcfb4eb175 : SUCCESS in 1h 00m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f85e7b454fd344e48035364b08fd08e1 : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":2},{"id":"1cd8c6a6c6c4cc04d5eb32c711d99b0baf10a9f7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-08-06 17:36:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nGood to backport.","accounts_in_message":[],"_revision_number":2},{"id":"4778ef3dffb0185d42531e518dfa491be0a30a3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 17:37:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0f0d0456cb2a24a0d368bfc1b062324b5b926683","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 19:17:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0599124998ce305abce5201fbed63e85501c6f83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 19:17:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/dbaf5a458fee487f9a3c70df9c39eaca : SUCCESS in 47m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/56ed1ab380834016ae21416bff877680 : SUCCESS in 53m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2992331eec64b1ea088154fc2255e53 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6292c4e835ab4a8c8c1b0c1b38b8abf9 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3099e15851f34cb08ba0214ffbd24276 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00e481b774dd4e099c80883381e0a16f : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115bcadfe05147b7b7ebb1b89bb1c26e : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/567395221b4543869cb44825ec6cb3cc : SUCCESS in 4m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/362044779a4b4236b3ec7435dee39aa8 : SUCCESS in 7m 07s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e8b5f6d590754549a4199f2ec5d8c80a : SUCCESS in 7m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f249265ab0c94ba59f69d4fd746b21b3 : SUCCESS in 39m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ab47a178ae3fe079b7d6d1708d517ecb5b6452f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 19:18:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2c1d546fb624c86a22e0a5adc60468d : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de3fa82f708b49a6928eb674df752d7d : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2b3ce07c32624f9987df87520c5ff8d538c8e5fd","revisions":{"bd9524f1003499208871426167e89b3e0f15dc82":{"kind":"REWORK","_number":1,"created":"2020-08-06 09:18:50.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/00/745100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/00/745100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/745100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/745100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/745100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/00/745100/1"}}},"commit":{"parents":[{"commit":"69848bfc582c827e5da374724a1055673eef6afa","subject":"Add \"stores\" to disallowed properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69848bfc582c827e5da374724a1055673eef6afa"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-30 16:20:36.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-06 09:17:56.000000000","tz":60},"subject":"Fix active image without data","message":"Fix active image without data\n\nFixes image import resulting active image without any data\nwhen the image happens to be deleted from staging before\n_DeleteFromFS task is ran which is supposed to cleanup.\n\nChange-Id: I6256a25d1907ec15fbdde560fc8eae37f3067213\nCloses-Bug: #1889640\n(cherry picked from commit b985e44f795340edfceed3b46abf9b4b28907cbb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd9524f1003499208871426167e89b3e0f15dc82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd9524f1003499208871426167e89b3e0f15dc82"}]},"branch":"refs/heads/stable/ussuri"},"2b3ce07c32624f9987df87520c5ff8d538c8e5fd":{"kind":"REWORK","_number":2,"created":"2020-08-06 13:10:22.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/00/745100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/00/745100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/745100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/745100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/745100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/00/745100/2"}}},"commit":{"parents":[{"commit":"69848bfc582c827e5da374724a1055673eef6afa","subject":"Add \"stores\" to disallowed properties","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69848bfc582c827e5da374724a1055673eef6afa"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-07-30 16:20:36.000000000","tz":60},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-08-06 13:09:50.000000000","tz":60},"subject":"Fix active image without data","message":"Fix active image without data\n\nFixes image import resulting active image without any data\nwhen the image happens to be deleted from staging before\n_DeleteFromFS task is ran which is supposed to cleanup.\n\nConflict: Correct import order used in test.\nChange-Id: I6256a25d1907ec15fbdde560fc8eae37f3067213\nCloses-Bug: #1889640\n(cherry picked from commit b985e44f795340edfceed3b46abf9b4b28907cbb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b3ce07c32624f9987df87520c5ff8d538c8e5fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b3ce07c32624f9987df87520c5ff8d538c8e5fd"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
