)]}'
{"id":"openstack%2Fglance~745392","triplet_id":"openstack%2Fglance~master~Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c","project":"openstack/glance","branch":"master","topic":"task-heartbeat","hashtags":[],"change_id":"Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c","subject":"Add functional test for task status updating","status":"MERGED","created":"2020-08-07 18:58:20.000000000","updated":"2020-08-26 19:28:28.000000000","submitted":"2020-08-26 19:22:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745392-1598469741572-f4e8b138","meta_rev_id":"a5792e545dbbcf1d3b34bb51cb833d8178bea795","_number":745392,"virtual_id_number":745392,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-26 07:01:47.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 19:22:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-24 13:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-08-24 13:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-08-25 15:23:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-08-24 13:41:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-18 14:24:04.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-08-18 15:16:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-25 15:23:38.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-26 07:01:47.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-26 19:22:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a68bfaa26fa18a7b4e254e9a9e0ba0a49b88da99","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 18:58:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5e1b598d2a7782b1a9066339a67d74fe2f0a058","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 18:59:20.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"0eff9a5f2d17d0233278cbba6077da6b179c200a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 19:11:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c48602a725004c24b4e39831c097dae0 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53953823c3e2400ca875dc84e93c478b : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aac771e912724be981569d369d7f114c : FAILURE in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/def4e21fdc1f44ea914f1dc2239d458b : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85daeedbaf2741e0a880ee133be5d911 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a260bc794f1e45a7a6fdd385fac91200 : FAILURE in 11m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e743623f9bc437c931fbaf2b70cc521 : SUCCESS in 7m 44s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/40db3f8c5f49481a9b3c2a0ca6eefb5c : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"05db4483e3542143c653a9344ae0e26f4f8d00ae","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 19:31:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a2fce6c287e21811263c4364abada7e97cd6b233","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 19:38:25.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"76a63aee43316542686467ff4fda7112b65abdeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 20:10:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2be3b0fd7f3744559c014af1bc91e322 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0346dfba339f42c898363a9ed7e29fd3 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1128830208df42fd91b4582a7773f798 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d43f8d68d274ce49e5b7943ac114648 : SUCCESS in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b4467c17ee54f70938acbdaeb119bb1 : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9f6329a5e3274ebeae2cc7d05770a59e : TIMED_OUT in 30m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0c2747de9ae41de8b96f76d217604a3 : FAILURE in 8m 01s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/48ac1e1731794525ad60b075b4d304c2 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":2},{"id":"15c13a2acc7fc40fc92d7189f2c60cdd13b568a1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-10 13:45:11.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9c54bf95263661a95a1509f6107b87fa1a532830","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-10 14:34:35.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"9bbc781f0b883e449e867a33af777bc029b74d6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 14:45:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe8d5bd851c84974b7249d5db9046cd7 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aedb9c65b5be490995768146e42e3bf5 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de5a9d80da8148afa67627a164d91934 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d09c7b95a914f5b8bd0083358b39d8d : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0024589d5f044ed8cfd88df8e6d232f : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e4f302330f324f65bfb50e2d60480cb7 : SUCCESS in 9m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca7c3216fa8942279b22482250061175 : SUCCESS in 9m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bef9513fdba54fc3819f4bed830e999d : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":3},{"id":"9f8039f1d8aaafbc65f7aa9d286813eafd621118","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-10 18:01:05.000000000","message":"Patch Set 3: Code-Review+2\n\nNice test to have!!","accounts_in_message":[],"_revision_number":3},{"id":"0bd66782109109b4c32cd30dcda7d4b1e31cd601","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-10 19:27:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5cbb18873203d55f2bc026c388e2b5d9b0346520","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-12 17:27:56.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"4e7e3cdaabe9764ad20a7b0ecfa90810405326d4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-12 17:28:31.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"8fb445a01b7dfdddd0c9ff416624e19ae5524cb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 18:08:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e14668151b92415985f9d6588dae4078 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11726c6871f342848ac74b9e81e9b2d1 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c1be4a6e443400c95e600050b418596 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ab7a623bcc6481a846dd15547b1fcbe : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031af2cfe27d4f85a98f4efc76734367 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3020e42188dd444ebb236b00c9620643 : SUCCESS in 7m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/230333789551466ebc4e54b2f7e859cf : SUCCESS in 7m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a072e13abe294fe0aa0e1646d8c3fa7a : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":4},{"id":"4af466655b4b437f1efcb7f4107b7d973c3a0972","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-18 14:24:04.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5847e34d07488d5916dc022eee0f6ebf5df61afb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-18 15:04:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"35d8e4f0c315dfa1530c261ae9f8f0b0180d5bf0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-18 15:05:28.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"f19a243b00c69b70a16d8bcbf77b40cefddff02b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-18 15:13:24.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"15b2ade3af4d56944016d3033ce06717bad30d27","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-18 15:16:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"72a756334eba9f6efdf5cacb6b1a6c6a33e59fbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 15:57:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea50e0978a89426292da043e0f50c98e : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75dfdbada4c6488984208560f3cc84d6 : FAILURE in 9m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e27a8c5f1c34214a3301bed1d99d5d7 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1537697cd0ff4d3db2e6d9b6a4fd718d : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e8ac8a72db4eb6bda3c1e1a60c123b : SUCCESS in 6m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb26c0861e2f44308d5fc009ab1a8315 : FAILURE in 13m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0fc8f9a700db4485800e51be2d9e4c0c : SUCCESS in 13m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3acc77d45a9f487f850e5f9bad9387de : SUCCESS in 7m 15s\n\nWarning:\n  Comments left for invalid file glance/tests/utils.py","accounts_in_message":[],"_revision_number":5},{"id":"a92d7c92a2e7c3bc4b1d156bbaac05eefcf839ee","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-18 16:34:38.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"33648e43549df096df7d878f3a6e867201132260","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-18 16:35:23.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"1f39e2b3538a698b8139c693fe7c7c5995324617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 16:50:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6953d26dd39c4072821124332cf7d8d8 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf4be725a3747c9bc9978073cddf65d : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbcd822a2c12499c86a5033b66adea05 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62d1961c6a5c49bdaa9e6f6ef6b3c738 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51092d6ae25b4bc48ddb4f9e9d2a5afa : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/70520432947148c7933bda12b6a70ed7 : SUCCESS in 8m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/06c5a7c7b595444394422b65a7533bdc : SUCCESS in 10m 58s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6fad0f796e624035a302416e3506122b : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":6},{"id":"e49560ce0b5d84bb85243eef3e245a496a8f1068","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-20 22:30:43.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"925359baf93e98ac689bc000b3eff08acf9fde13","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-20 22:31:38.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"7154bb58599f29835f6aae8bdb835830c8645d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 23:24:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c05dae6ed57344c385305f35a70d079a : FAILURE in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f7779bc26a443590a0dc1d1400387e : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d22750a64bd4d41b2c084b0cea88e8e : FAILURE in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f81149bd9c6483cb7d1f2a81c425231 : FAILURE in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f32cc68a4047f0b10ae842587d49e2 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c25cf9177d18421fb3057728d44e3997 : SUCCESS in 9m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/efb6b42164a64a2083d0bc7f108b8fb9 : SUCCESS in 9m 16s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3d2f7d3e7520441e966f5740856d37aa : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":7},{"id":"3124e6a7e2fc8d1f7734fb891157b025309eaf28","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-21 14:30:36.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"9c001e1ae4e439b2a774e9aa6e1d10ef5a6d62ce","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-21 14:31:32.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":8},{"id":"0d8633ea261b2f1be0904a6c109167dcc9ece687","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-21 15:18:01.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"06b625d8d4573fbae48570308885f5eee57ae5d1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-21 15:18:50.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":9},{"id":"6b4aab7b15735a96aaf28638115cfb0f7412fba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 18:23:01.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e7df6cefd3e4716acff78f18b4d1712 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ef596b30df4457ac933490aae73d9e : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db97b975b08d4b858440d9ede6282ba1 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccba3b51cf894af09fae1e5311ccdee1 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9989bfe337e451aa07a2ec95d016345 : SUCCESS in 7m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d7b6f0d3a9174111bebcd19d8611eae9 : FAILURE in 12m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6253952a4a2b4a3887b7af67469f3af9 : SUCCESS in 9m 15s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/99ed896de6714e219b7f160014963550 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":9},{"id":"916713c7ceb02084f8d5afad32166e92eb2240c9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-21 19:53:47.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"fdef98a4c1f2c4846e945cd0afc093951cc2412b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-21 19:55:03.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":10},{"id":"7c7e227768ec1baa8fc8566bd5f60911af9091b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 21:44:20.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d747e055211d40ee842e0a96b5b99a77 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e569b76bcfae45559c6b910edd2de40e : FAILURE in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/108a49b3d4c54078941e7f4c673d1fc8 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d750105ac285422db95cfe712785fce9 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a54568c8d8484636a744077a144d2fd9 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b272b757edf748e09fefdcce6edf6196 : SUCCESS in 9m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/79a3b212f62645918b51903bb9b7fef1 : FAILURE in 13m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/59d42c98397f4365954d490529ae4aaf : SUCCESS in 7m 16s\n\nWarning:\n  Comments left for invalid file glance/tests/functional/v2/test_images.py","accounts_in_message":[],"_revision_number":10},{"id":"91be4207759cf79eb6edd39d41243ccff7114623","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-24 13:41:26.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"4a0943717c3a1687d15c3ed3e1bf0357274c5767","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-24 13:42:44.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"1cbbb4237d27025c07a258a502e1ef1f0084fb7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 16:01:12.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aed4d82feead43c0b796b8638d867792 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6158d6d90ff24fbb992a34d355199c69 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9efd8446a12b411d9e36d2224d1d5a29 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1636ffd0f82b442b8dfce90147710c73 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5fb6d1bb21f479e8f9d3713b90c6b88 : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d0f3717c7f0e450b9fd6772a20a39bb8 : SUCCESS in 10m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4cec85b5bb1543058b71387a2414b372 : SUCCESS in 9m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c427772a328b44a2ba0cef4b853ce96d : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":11},{"id":"af4c58ddd163f6084122915daeeace9f909e9230","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-25 15:23:38.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"2f5526ee376457f33792e7baaa284b567a10e821","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:24:10.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"e97e7f90822bf740c38ce06f03299369152eb532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 16:36:52.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"f31104ad7dbe4e858a66f6e52a32b16357c3524f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-26 07:01:23.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"336313b8b80865c22c0dc7841a25d88809148475","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-26 07:01:47.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":11},{"id":"daa47821ff306b19a5845137a7a6f7fce903d39a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 09:21:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f28b97f03d74a03a91c1ab55326ab76 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6da5843afb51439d90afd14e821eafd5 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37fb53f90ba34b69a7bca5fcbf453ccf : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91a607bac696407ab7fc11aa56fc3678 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34c8a24356a349cf9ecb979270aa5ffa : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b7451953629d44faa1f45393b9f413b2 : SUCCESS in 11m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d3b131c716db46bbb6613b615d0e5617 : SUCCESS in 12m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ce8527a35f5e43f3a5502e240c283095 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":11},{"id":"2ff58c83e490eb1c766703beba814355781af4af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:56:42.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"343c4a86112e5bfcf6d994618a8f8e776d6fb890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:22:20.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/235ecc4745774a699bf1df56e9ef562c : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bcb5a1bfddb4a2b8c0fae71edd0001d : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22a123a2b21141b0b8d25328974a3c19 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ad40f9f14c949febeecddc05f079c96 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b869601e0d0498aad43f78d3ae3f526 : SUCCESS in 5m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/272e3ce8c0e746a5b19f80c6a1553d12 : SUCCESS in 9m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/778ba6aba70347eea37953ef57aa2e84 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":11},{"id":"11f71994001da0bf4e9e9c68f135cd43a4a781f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:22:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"09294ff49f79aa95243bba59a7b88474c786fd2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:28:28.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80a041d4c6246a6b2fbff5952c2e587 : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"1b006c4f44e56d08d5ca1fb9090dc83990395d87","revisions":{"63982a10e96271f42680d4e80f2b7b323ff9abb3":{"kind":"REWORK","_number":1,"created":"2020-08-07 18:58:20.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/745392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/745392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/745392/1"}}},"commit":{"parents":[{"commit":"51d812ab97f18ec03cf035ef1c5d6fbbd7a4c68e","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51d812ab97f18ec03cf035ef1c5d6fbbd7a4c68e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"subject":"Add functional test for task status updating","message":"Add functional test for task status updating\n\nThis was really hard to test before the introduction of the\nSynchronousAPIBase test class, so this is added after that patch for\nconvenience.\n\nThis tests uses two mocks, but it\u0027s otherwise quite \"functional\" in its\nrealness. It mocks out the timer so it always fires, and it mocks out\nthe effective call to glance_store and uses that as a hook to grab the\ntask state in lockstep (and avoid making another copy of the fake data\non disk, since we need to generate a couple MiB of data to test this).\n\nChange-Id: Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63982a10e96271f42680d4e80f2b7b323ff9abb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63982a10e96271f42680d4e80f2b7b323ff9abb3"}]},"branch":"refs/heads/master"},"83d3517c6792fca94e1eef65d3677555c458b3ff":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-07 19:31:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/745392/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/745392/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/745392/2"}}},"commit":{"parents":[{"commit":"df7c8e023e63751bccedb54433433e223ab6f28c","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df7c8e023e63751bccedb54433433e223ab6f28c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 19:22:33.000000000","tz":-420},"subject":"Add functional test for task status updating","message":"Add functional test for task status updating\n\nThis was really hard to test before the introduction of the\nSynchronousAPIBase test class, so this is added after that patch for\nconvenience.\n\nThis tests uses two mocks, but it\u0027s otherwise quite \"functional\" in its\nrealness. It mocks out the timer so it always fires, and it mocks out\nthe effective call to glance_store and uses that as a hook to grab the\ntask state in lockstep (and avoid making another copy of the fake data\non disk, since we need to generate a couple MiB of data to test this).\n\nChange-Id: Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83d3517c6792fca94e1eef65d3677555c458b3ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83d3517c6792fca94e1eef65d3677555c458b3ff"}]},"branch":"refs/heads/master"},"3f819cb456e1de4f047284a5774e6b09d12abb3e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-10 13:45:11.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/745392/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/745392/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/745392/3"}}},"commit":{"parents":[{"commit":"f62b959efbea79c35776eacbbd13933d113fc39e","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f62b959efbea79c35776eacbbd13933d113fc39e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-10 13:44:36.000000000","tz":-420},"subject":"Add functional test for task status updating","message":"Add functional test for task status updating\n\nThis was really hard to test before the introduction of the\nSynchronousAPIBase test class, so this is added after that patch for\nconvenience.\n\nThis tests uses two mocks, but it\u0027s otherwise quite \"functional\" in its\nrealness. It mocks out the timer so it always fires, and it mocks out\nthe effective call to glance_store and uses that as a hook to grab the\ntask state in lockstep (and avoid making another copy of the fake data\non disk, since we need to generate a couple MiB of data to test this).\n\nChange-Id: Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f819cb456e1de4f047284a5774e6b09d12abb3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f819cb456e1de4f047284a5774e6b09d12abb3e"}]},"branch":"refs/heads/master"},"43ae46e141cd6ecf05761aa9914e6682d096169d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-12 17:27:56.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/745392/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/745392/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/745392/4"}}},"commit":{"parents":[{"commit":"30474daa3ab0f79cae658274e36e2130ededaf21","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30474daa3ab0f79cae658274e36e2130ededaf21"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-12 17:26:18.000000000","tz":-420},"subject":"Add functional test for task status updating","message":"Add functional test for task status updating\n\nThis was really hard to test before the introduction of the\nSynchronousAPIBase test class, so this is added after that patch for\nconvenience.\n\nThis tests uses two mocks, but it\u0027s otherwise quite \"functional\" in its\nrealness. It mocks out the timer so it always fires, and it mocks out\nthe effective call to glance_store and uses that as a hook to grab the\ntask state in lockstep (and avoid making another copy of the fake data\non disk, since we need to generate a couple MiB of data to test this).\n\nChange-Id: Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43ae46e141cd6ecf05761aa9914e6682d096169d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43ae46e141cd6ecf05761aa9914e6682d096169d"}]},"branch":"refs/heads/master"},"f78b669f8d625aac2c2ec8c1f16e58581e5ad311":{"kind":"REWORK","_number":5,"created":"2020-08-18 15:04:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/745392/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/745392/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/745392/5"}}},"commit":{"parents":[{"commit":"15ed9eb9a5fba0bc73f83a4db916534d5777ef9b","subject":"Add FakeData generator test utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/15ed9eb9a5fba0bc73f83a4db916534d5777ef9b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 15:03:45.000000000","tz":-420},"subject":"Add functional test for task status updating","message":"Add functional test for task status updating\n\nThis was really hard to test before the introduction of the\nSynchronousAPIBase test class, so this is added after that patch for\nconvenience.\n\nThis tests uses two mocks, but it\u0027s otherwise quite \"functional\" in its\nrealness. It mocks out the timer so it always fires, and it mocks out\nthe effective call to glance_store and uses that as a hook to grab the\ntask state in lockstep (and avoid making another copy of the fake data\non disk, since we need to generate a couple MiB of data to test this).\n\nChange-Id: Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f78b669f8d625aac2c2ec8c1f16e58581e5ad311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f78b669f8d625aac2c2ec8c1f16e58581e5ad311"}]},"branch":"refs/heads/master"},"3ed8fd47eba10d35066b1501e27ba03374843233":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-08-18 16:34:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/745392/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/745392/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/745392/6"}}},"commit":{"parents":[{"commit":"c4ded9f657453870dc9e392dbe175d4ab7a0c71a","subject":"Add FakeData generator test utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4ded9f657453870dc9e392dbe175d4ab7a0c71a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 16:32:06.000000000","tz":-420},"subject":"Add functional test for task status updating","message":"Add functional test for task status updating\n\nThis was really hard to test before the introduction of the\nSynchronousAPIBase test class, so this is added after that patch for\nconvenience.\n\nThis tests uses two mocks, but it\u0027s otherwise quite \"functional\" in its\nrealness. It mocks out the timer so it always fires, and it mocks out\nthe effective call to glance_store and uses that as a hook to grab the\ntask state in lockstep (and avoid making another copy of the fake data\non disk, since we need to generate a couple MiB of data to test this).\n\nChange-Id: Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ed8fd47eba10d35066b1501e27ba03374843233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ed8fd47eba10d35066b1501e27ba03374843233"}]},"branch":"refs/heads/master"},"25cb2a8a28dae26ff7a3617f5bf2d21da16d09c7":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-08-20 22:30:43.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/745392/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/745392/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/745392/7"}}},"commit":{"parents":[{"commit":"5658a1e3859201d3c3735b1ab39f077806f3329d","subject":"Add FakeData generator test utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5658a1e3859201d3c3735b1ab39f077806f3329d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-20 22:30:23.000000000","tz":-420},"subject":"Add functional test for task status updating","message":"Add functional test for task status updating\n\nThis was really hard to test before the introduction of the\nSynchronousAPIBase test class, so this is added after that patch for\nconvenience.\n\nThis tests uses two mocks, but it\u0027s otherwise quite \"functional\" in its\nrealness. It mocks out the timer so it always fires, and it mocks out\nthe effective call to glance_store and uses that as a hook to grab the\ntask state in lockstep (and avoid making another copy of the fake data\non disk, since we need to generate a couple MiB of data to test this).\n\nChange-Id: Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25cb2a8a28dae26ff7a3617f5bf2d21da16d09c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25cb2a8a28dae26ff7a3617f5bf2d21da16d09c7"}]},"branch":"refs/heads/master"},"364bc8a16fe2e1749f0380080a71d79970c26c4d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-08-21 14:30:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/745392/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/745392/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/745392/8"}}},"commit":{"parents":[{"commit":"3e8df230a5ea8970c73b390d93824d504cbb788c","subject":"Add FakeData generator test utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e8df230a5ea8970c73b390d93824d504cbb788c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 14:27:34.000000000","tz":-420},"subject":"Add functional test for task status updating","message":"Add functional test for task status updating\n\nThis was really hard to test before the introduction of the\nSynchronousAPIBase test class, so this is added after that patch for\nconvenience.\n\nThis tests uses two mocks, but it\u0027s otherwise quite \"functional\" in its\nrealness. It mocks out the timer so it always fires, and it mocks out\nthe effective call to glance_store and uses that as a hook to grab the\ntask state in lockstep (and avoid making another copy of the fake data\non disk, since we need to generate a couple MiB of data to test this).\n\nChange-Id: Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/364bc8a16fe2e1749f0380080a71d79970c26c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/364bc8a16fe2e1749f0380080a71d79970c26c4d"}]},"branch":"refs/heads/master"},"d76a550765d1d2a22088c6c22e13bd35676710c5":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-08-21 15:18:01.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/745392/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/745392/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/745392/9"}}},"commit":{"parents":[{"commit":"0593907da56ff36fca00feffce7579fc2f786fa3","subject":"Add FakeData generator test utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0593907da56ff36fca00feffce7579fc2f786fa3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 15:17:55.000000000","tz":-420},"subject":"Add functional test for task status updating","message":"Add functional test for task status updating\n\nThis was really hard to test before the introduction of the\nSynchronousAPIBase test class, so this is added after that patch for\nconvenience.\n\nThis tests uses two mocks, but it\u0027s otherwise quite \"functional\" in its\nrealness. It mocks out the timer so it always fires, and it mocks out\nthe effective call to glance_store and uses that as a hook to grab the\ntask state in lockstep (and avoid making another copy of the fake data\non disk, since we need to generate a couple MiB of data to test this).\n\nChange-Id: Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d76a550765d1d2a22088c6c22e13bd35676710c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d76a550765d1d2a22088c6c22e13bd35676710c5"}]},"branch":"refs/heads/master"},"9c6bcd4380afa4ece66cf15cab29bd09de9f4371":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-08-21 19:53:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/745392/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/745392/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/745392/10"}}},"commit":{"parents":[{"commit":"3d6fe925a767de904cf975e0ff5bdf25a65aa2d4","subject":"Add FakeData generator test utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d6fe925a767de904cf975e0ff5bdf25a65aa2d4"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 19:52:05.000000000","tz":-420},"subject":"Add functional test for task status updating","message":"Add functional test for task status updating\n\nThis was really hard to test before the introduction of the\nSynchronousAPIBase test class, so this is added after that patch for\nconvenience.\n\nThis tests uses two mocks, but it\u0027s otherwise quite \"functional\" in its\nrealness. It mocks out the timer so it always fires, and it mocks out\nthe effective call to glance_store and uses that as a hook to grab the\ntask state in lockstep (and avoid making another copy of the fake data\non disk, since we need to generate a couple MiB of data to test this).\n\nChange-Id: Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c6bcd4380afa4ece66cf15cab29bd09de9f4371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c6bcd4380afa4ece66cf15cab29bd09de9f4371"}]},"branch":"refs/heads/master"},"1b006c4f44e56d08d5ca1fb9090dc83990395d87":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-08-24 13:41:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/745392/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/92/745392/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/92/745392/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/92/745392/11"}}},"commit":{"parents":[{"commit":"dc08127f059a227928400bf2b629ef503dbf28e7","subject":"Add FakeData generator test utility","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc08127f059a227928400bf2b629ef503dbf28e7"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 18:56:28.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-24 13:41:13.000000000","tz":-420},"subject":"Add functional test for task status updating","message":"Add functional test for task status updating\n\nThis was really hard to test before the introduction of the\nSynchronousAPIBase test class, so this is added after that patch for\nconvenience.\n\nThis tests uses two mocks, but it\u0027s otherwise quite \"functional\" in its\nrealness. It mocks out the timer so it always fires, and it mocks out\nthe effective call to glance_store and uses that as a hook to grab the\ntask state in lockstep (and avoid making another copy of the fake data\non disk, since we need to generate a couple MiB of data to test this).\n\nChange-Id: Ibd0c802efa2723e11ece1097c9bf1ad68b1a820c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b006c4f44e56d08d5ca1fb9090dc83990395d87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b006c4f44e56d08d5ca1fb9090dc83990395d87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
