)]}'
{"id":"openstack%2Fglance~745398","triplet_id":"openstack%2Fglance~master~I66cdd29341089ef89123fa1f5420258433a660fe","project":"openstack/glance","branch":"master","topic":"task-heartbeat","hashtags":[],"change_id":"I66cdd29341089ef89123fa1f5420258433a660fe","subject":"DNM: Try to debug spurious DB file failure","status":"ABANDONED","created":"2020-08-07 19:31:31.000000000","updated":"2020-08-10 17:11:11.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6eb8ee6b78ba2b46621ecbbe9edb7d89d040fdb7","_number":745398,"virtual_id_number":745398,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-08 02:21:17.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2020-08-08 02:21:09.000000000","_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"date":"2020-08-10 13:44:27.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-08 03:38:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-08 02:21:09.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2020-08-08 02:21:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-08 03:38:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"668ba0ee8fcb7f8a6c742c2f30479e198e0b9bef","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 19:31:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30e7c35205a33ad14e75b56ffe0fd08a05cdb0b2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-07 19:39:24.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"394fb56ee370a6ebd6c1880b8d1db579b06d70f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 19:57:02.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82747a816723449da4cc5f9ae493316f : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a6aa633b216474b8b7f47acc044c177 : FAILURE in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14ff3384f9514de5a3cb54889942e2da : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f1ddb8235604bd2bc1d8c1c01c33f2c : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6254400265d44be1bf71db5b3e40a21b : SUCCESS in 6m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/856072736ea14899a151950e03089819 : FAILURE in 12m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/80236da1ffd34d8e866ee1d1dd85a020 : SUCCESS in 17m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a847f41749034d399f58c4f99803a417 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"45d39e315ab2c9475b6c1cc93b2500fa5c1a897c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-07 23:27:11.000000000","message":"Patch Set 1:\n\nMike, I\u0027m trying to add some testing against a DB here in glance, which I think is mostly stubbed out for all their others tests, other than some where they start a full API server and hit that.\n\nIf you grab this patch and run:\n\n tox -efunction-pyXX -- test_images_import_locking --serial\n\nI think you\u0027ll see the error I\u0027m hitting every time. Basically, the db_sync() in the setup_database() of the test class always works the first time, but fails on any subsequent calls. The DB file does exist, the directory is writable, but it fails. I think maybe there\u0027s some internal state I\u0027m not resetting and it\u0027s still trying to use the DB file from the first test, which has been reaped.\n\nCan you help?","accounts_in_message":[],"_revision_number":1},{"id":"72a9855b3ed78b6bd53b15e74144c8957bb1082c","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-08-08 02:20:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ced7d9daa869ea337d8db54c345754139eed6f7d","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-08-08 02:21:09.000000000","message":"Patch Set 2:\n\nalembic env.py is getting the DB information from a different source than that which the test code is changing on each test.  im not sure where these tempfile names come from but they change each time.   tests run with the change i made here.","accounts_in_message":[],"_revision_number":2},{"id":"4b0edd25e59b885abc48876d2690084d2c31ecb7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-08 02:21:17.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"8dcc2b1e3234f488793ea673d90548af29511aa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 03:38:21.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7eaefafc96b84b1ea0be87cf34138317 : SUCCESS in 1h 04m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7a13c47eeaf54ff9bf7dd4d32f569392 : SUCCESS in 1h 04m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e271dd69fce40b590ed4a60636c35c6 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/701e22e82ce448b48c66320a08ab093d : FAILURE in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa0087cc93074d4981c3c9b4996b19ac : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f10a6850cf64af7a843f50b37b4212f : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea72e947c0b046ac910bf9092359fecf : SUCCESS in 4m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f79d3eefd12e4f598f06836a11934751 : FAILURE in 6m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/967b8ed006e44c41a88cb8b199d7884f : FAILURE in 7m 01s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7467be95fffc4a1a8d18311ba37d2b5f : SUCCESS in 4m 44s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/aafd6ed689e94f1697ce6bb24681853f : FAILURE in 1h 16m 45s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/09bff75519884b9bbf3eba81217a1a58 : SUCCESS in 1h 06m 57s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/4b8ba92472f4489d82bfbfde31e9650e : SUCCESS in 1h 15m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0384881401d4caebda646d33b870e82 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":2},{"id":"ca6e65a68413976289a10d1f46d2b3994420df6b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-10 13:44:27.000000000","message":"Patch Set 2:\n\nMike, thanks so much for figuring this out. This gives me exactly what I need to proceed!","accounts_in_message":[],"_revision_number":2},{"id":"05c7bfa8cfa7b349b75ab205122c0d22da509cc9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-10 17:11:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"47f994b28810e13ca3c427938d70cf585863fc89","revisions":{"724ccc29a9eb1d3e975e5c91852bac92710655c1":{"kind":"REWORK","_number":1,"created":"2020-08-07 19:31:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/98/745398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/745398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/745398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/745398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/745398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/745398/1"}}},"commit":{"parents":[{"commit":"83d3517c6792fca94e1eef65d3677555c458b3ff","subject":"Add functional test for task status updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83d3517c6792fca94e1eef65d3677555c458b3ff"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 19:31:14.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 19:31:14.000000000","tz":-420},"subject":"DNM: Try to debug spurious DB file failure","message":"DNM: Try to debug spurious DB file failure\n\nChange-Id: I66cdd29341089ef89123fa1f5420258433a660fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/724ccc29a9eb1d3e975e5c91852bac92710655c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/724ccc29a9eb1d3e975e5c91852bac92710655c1"}]},"branch":"refs/heads/master"},"47f994b28810e13ca3c427938d70cf585863fc89":{"kind":"REWORK","_number":2,"created":"2020-08-08 02:20:09.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/98/745398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/745398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/745398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/745398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/745398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/745398/2"}}},"commit":{"parents":[{"commit":"83d3517c6792fca94e1eef65d3677555c458b3ff","subject":"Add functional test for task status updating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83d3517c6792fca94e1eef65d3677555c458b3ff"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-07 19:31:14.000000000","tz":-420},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2020-08-08 02:18:37.000000000","tz":-240},"subject":"DNM: Try to debug spurious DB file failure","message":"DNM: Try to debug spurious DB file failure\n\nAlembic env.py is using oslo_db.sqlalchemy.enginefacade\nfor connectivity but the test suite seems to be creating\na new CONF.database.connection tempfile directory + file\neach time and assigning it to _FACADE, which is a legacy facade.\n\nnot sure where glance wants to go here but switching\nenv.py to use _FACADE allows the tests in question to pass.\n\nChange-Id: I66cdd29341089ef89123fa1f5420258433a660fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/47f994b28810e13ca3c427938d70cf585863fc89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/47f994b28810e13ca3c427938d70cf585863fc89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
