)]}'
{"id":"openstack%2Fglance~746518","triplet_id":"openstack%2Fglance~master~Ib910274472346ce0c336cd1ead8370d5799d0b96","project":"openstack/glance","branch":"master","topic":"task-heartbeat","hashtags":[],"change_id":"Ib910274472346ce0c336cd1ead8370d5799d0b96","subject":"Handle atomic image properties separately","status":"MERGED","created":"2020-08-17 14:35:50.000000000","updated":"2020-08-29 19:14:49.000000000","submitted":"2020-08-29 19:13:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746518-1598728413450-bc9ee2cc","meta_rev_id":"20cf0202d939c71a42ca70a8f6934f0d598a7a2c","_number":746518,"virtual_id_number":746518,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-29 09:20:54.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-29 19:13:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-28 15:26:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-24 13:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-28 15:26:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-24 13:41:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 14:36:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-28 15:26:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-29 09:20:54.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-29 19:13:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce80ffdb8827ec3b566170a09f75cbaa2270d2c0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-17 14:35:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c5368d1df10ea7b16ed3f76211db4dbbda39e10","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-17 14:36:23.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"37e8caf0b2fc467101f90a03d01c9e96d4ff0fd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:30:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/23d9539d18244cd2a6d93e415567e2e5 : SUCCESS in 1h 08m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4465f0da5ac144a89a17380da83fc057 : SUCCESS in 1h 13m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdb452246d9f473ba693bcd318bac21e : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5679d59379e74a98b42f199b59f87bca : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9874821adc0a42c4bfafde02c8620083 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c860dda528e74641a1f79c795db2a7dc : SUCCESS in 10m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c9d5be1dd94424b91b678b9941b6ab : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/62fbd7f94aa94be4a45f5f2339bca45b : SUCCESS in 7m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/10d3342a722b411da7fb0b4d8e52486b : SUCCESS in 9m 29s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2dbac5cee05f4057b53221252b858ecc : SUCCESS in 11m 45s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6b032b852f1449c3aeb2573a82ac5d19 : SUCCESS in 1h 19m 26s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d7c442cbd8f5495887881042be71b267 : SUCCESS in 1h 10m 10s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/fadda46fd43647708421dbe48bcb1c1b : SUCCESS in 1h 08m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1af6e0ee3d0e4f9a8e01781f6ecce7c0 : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":1},{"id":"090bc3899bf316ab1426969a75fa4eb33e95c5d7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-18 15:04:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ebb1d4215c5d36163461cb999dadfcb137e55c77","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-18 15:06:17.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"1456fdb9f2af8860de66fb0506a346d30db0969c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-18 16:34:38.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"62cbce1d81bab7c1b3e05ac127082276589e751d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-18 16:36:13.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"29c95152368be6754c5b971a9dd02eb9ed7948d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 18:01:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/579dd85af8ef4801b868f5043ba796b3 : SUCCESS in 59m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c23b1e979e3e41a5bd4b576f3b366fc9 : FAILURE in 1h 10m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1cfb176153a4e93ad43e752d6f3be33 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbe4edc8329e4a4d8147db35cf882cd4 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a252f79e0f24773be1525504d3ce45d : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69b3f77a539b4ca38b7045852a95a19d : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8bc5e646574ca5accb8811a2796f3e : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/43cd9913adc4419fabea005abf4164b9 : SUCCESS in 7m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c5664a6ec6444f619b766531e26959ae : SUCCESS in 8m 37s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/56584b4dc80742d6865172979cecca1e : SUCCESS in 4m 00s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/03eadd2c3bd4407ea12a6fafdad25b47 : SUCCESS in 1h 01m 30s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/7bf08823d27e475caaef3662ccb64e2b : SUCCESS in 1h 02m 44s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/f702401c6d354bada5a9a2528bd9deea : SUCCESS in 1h 14m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7300a34a34a6463d8dd31de11825835a : SUCCESS in 55m 53s","accounts_in_message":[],"_revision_number":3},{"id":"fa2932bb3b79c81bd0c912e1f11968ac55d11d6d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-20 16:43:37.000000000","message":"Patch Set 3:\n\nrecheck nova tests only failed","accounts_in_message":[],"_revision_number":3},{"id":"530cc5640057ad870461808289c6b784a5dc3f61","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-20 16:44:22.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"4a64f8bf6ddd5a771e59f4a3ba501fe6a59c186f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:53:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8adefedc03a540c8b753c8973726bead : SUCCESS in 1h 07m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83ed8ac318ef4d1cad4390fea2a0d849 : SUCCESS in 1h 14m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db282b6a650b4eb68455d57468b80af9 : FAILURE in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b2547d132d949cdbcc2199f699ca015 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e8670f9d0e14b48b932f1d1ae87c796 : FAILURE in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3123dd598df44801907b5640456fc691 : FAILURE in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c09685575cf345f0adcb42bc1a851b23 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/73ef8dc06c1b4012b9744adcc0b1c802 : SUCCESS in 9m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c51cf850ff03418eabc239df1957f17c : SUCCESS in 9m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/85cef8e576f442f2b93f2abf39794652 : SUCCESS in 5m 28s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/98e7117afb6f4aebbd4f0779b2fd0e5b : SUCCESS in 1h 02m 08s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/c745c3d4654048729d9ea8de7dc7de70 : SUCCESS in 1h 06m 27s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/b3f2a9d5590447ee907bef5170967a9c : SUCCESS in 1h 09m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ceecf32d24c4052937a42dc0a144f0e : SUCCESS in 59m 57s","accounts_in_message":[],"_revision_number":3},{"id":"b1b0bb6dc5aa1c8db43fda5f29d4ab28273ebc78","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-20 22:30:43.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"0fb49e2d3bb076f9b49e74388177d6e8e28ab0ef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-20 22:32:28.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"1423a10b09c1aaade39721886596cb13705d5457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 00:35:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d0ac721f93274055bb35cc7458bdafeb : SUCCESS in 1h 14m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85d70af626ad4d0a8285a5bc927647c3 : SUCCESS in 1h 20m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccbbb20b36ec4170a3552bc896db52d0 : FAILURE in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e9c77f26f8641b08d93077e8250632e : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b50a063f562408485e7808a5af01f04 : FAILURE in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f6b7e2f044146619a1d84792d35a68d : FAILURE in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb4c98ab5814612a344f92f351e9004 : SUCCESS in 5m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ca8fac11165f482abb561a4cdea6d366 : SUCCESS in 8m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b40d1ddfb9e2490c95b9296cac156137 : FAILURE in 9m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/deb883ebefab405dbac3c0b0ca27c111 : SUCCESS in 4m 52s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/335b9b342f2e4f119c04e527ab72c4e9 : SUCCESS in 1h 03m 00s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/b075f03af2574cb59d71da5a2f974826 : SUCCESS in 1h 05m 26s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/aaced82dca1243dea7514fb504202f6d : SUCCESS in 1h 14m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ca572f49a40453792c4be9e5764c8bd : SUCCESS in 53m 32s","accounts_in_message":[],"_revision_number":4},{"id":"c6fd42dfd098eda970e972125c3209755bda9351","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-21 14:30:36.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d7251526a0cfe1fbac187af0c1ff2cac5b36a4ab","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-21 14:32:26.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"1d93b7062cf3bee1dbfeb0c232e0042a531c1199","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-21 14:36:26.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"d2faa43935886c253abed79256eb79942e1bb6b3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-21 15:18:01.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f03d5937c3a71c7e5c444150c631faf6bc9fd79d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-21 15:19:43.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"1a466d66ef6c7198430f2326b46137f1b98167c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 19:26:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/921f5ada006742c6a004bb9e3f67c7f8 : SUCCESS in 1h 06m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5bc02e7f2dd4558a31b74ab9ed405b9 : SUCCESS in 1h 09m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26950a9afaa442de92e91fa25b080658 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f176e21c34040fbb8177bf5c0d86ce0 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a67a2490e3948618faa3f488a93137c : SUCCESS in 8m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0e3ea11f19648c09642ae15e57a7dfb : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e374f77f5564a71a4470986ed36a195 : SUCCESS in 8m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5f27c304c324453834dcadd0585e7cf : SUCCESS in 10m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/16a65d85829b4392801924b76f3fd21c : FAILURE in 10m 41s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6cc5888636bc4c62a4cd72cdce3f1243 : SUCCESS in 6m 36s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2c7edf74de2d4ae59273b07018dd3f77 : SUCCESS in 1h 00m 24s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/dcbcf6c5ed0847ed9c14e5aea7de094a : SUCCESS in 1h 12m 06s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/f4caa1e8fbba46689a25ee4d75287bcb : SUCCESS in 1h 11m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/370590610d8c47d2b4f381bf3b2ba5e0 : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":6},{"id":"b9bb8fc3b930ba10bf28493be2703c8d1a13ed04","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-21 19:53:47.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"17c08bcbdc1ad3e90bdb29c169c1cb1a56f3f198","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-21 19:56:08.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"d748d6296789afa8572e82328e343db4338e9ccc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 22:50:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f140c08ddad3447bb9831b91dadf88a2 : SUCCESS in 56m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/93bc2d86bafd43fbbe6c5a3b3700480c : SUCCESS in 1h 14m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41e1323486804787934238eb099b2763 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc770eb64064960acaddb5e28b1a19c : FAILURE in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22193e8b3c2f446cb008ded47b33a512 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cef29acad896484190072de190e26c8e : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fda1b0164044de38124f85b313a663a : SUCCESS in 8m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/762ade84d3334bfab2474511b002aac9 : SUCCESS in 9m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/78887698392743f5bf31320673d204be : SUCCESS in 9m 15s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7ee2405b78ee482182ddfa62f97f2073 : SUCCESS in 5m 20s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/81abbd7207a74f80a371abf07edc9953 : SUCCESS in 1h 03m 28s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f46e701486d743b2814e6f04b73b527c : SUCCESS in 1h 06m 19s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/99e6f614f9504604985d80732d55fbb2 : SUCCESS in 1h 11m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec732f44fd2944db89a5af0aef19c2d1 : SUCCESS in 55m 44s\n\nWarning:\n  Comments left for invalid file glance/tests/functional/v2/test_images.py","accounts_in_message":[],"_revision_number":7},{"id":"6907655e73f26fcdef8193951de2d367eb7be55c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-24 13:41:26.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"ebc967709fb83063a351d4c359ae4094a2c548c4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-24 13:43:46.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":8},{"id":"17a04b7c2ede35719b8ef59fc05bc58807fe33ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 17:03:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/abf626c116384b749b757c803c31824b : SUCCESS in 1h 01m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/248a742dbd934b46a0dd9da65b0dd2e5 : SUCCESS in 1h 14m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b72ed749ab984311838d16061933aa36 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b96704ab55b4c1faaa81637f80cfaa8 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe4a4ef2ea054f53b2a824d6e8253c78 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/920c3c0a0f3f4f75a260875a0170f19d : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6378fe7585014953833d8b3d1a8395ef : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e9118b493b1842ecb594ee15f06f3b92 : SUCCESS in 7m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ae38cfe715944caa9d35d604519cf94f : SUCCESS in 9m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8e330083a6984f3e88f54bcc565fffad : SUCCESS in 5m 01s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bcca705d00f04e6493207304014095cc : SUCCESS in 1h 00m 15s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/3f03fd56dc904f4f89021b4ec48cb6ec : SUCCESS in 1h 03m 55s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/17b13a3bedbd473496d0f2524f99669a : SUCCESS in 52m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9e81bc961aa49d4a8aa936d64d974b8 : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":8},{"id":"48846ea078714d9161c2ac0edc40f4e645841322","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-28 15:26:43.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nLooks good to me.","accounts_in_message":[],"_revision_number":8},{"id":"ecf0bfcdcb6d516c21f8c07611b114be94ba9a06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 15:27:09.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"9e0327519b37193a3dfe3054e811ded150fbecf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 17:30:10.000000000","message":"Patch Set 8: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"fadf250e3a938e3d1035b0d9d2f10c959fd9e6c8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-28 17:33:08.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"1fa445388c71b56e8c92c9a08c7086dbe5501b43","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-28 17:33:32.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":8},{"id":"0f803811ebb63d23591a33c6030518aea2b98e2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 21:09:03.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d98f5e25e0e94b8389b8e253f4ee2be8 : SUCCESS in 59m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5fed84b749b04bc7817f808c60e292f5 : SUCCESS in 1h 37m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6538f9e09469404e862753e06b96d961 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06735d9eddf046559d9aa1d556a847a2 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86287460894644d79c5122660ddf5201 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5a4753d8b9c4579b1e6821577a132ee : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0567c54c1c8a49219f1812b610426c28 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec73e38e1c3d47f1aa164b015a74e213 : SUCCESS in 8m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/76d0126cd4ba42c19246fef48fe405ab : SUCCESS in 11m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/467be751dc694fec86037f952cf7ea73 : SUCCESS in 5m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ad4c0bc40810482fbd4fff539d02c3a8 : SUCCESS in 1h 03m 32s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/e18e6855d7a74945abc20e53ff4269a3 : SUCCESS in 1h 08m 28s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/6a4ee519ba484907af7b95cdaf533940 : SUCCESS in 1h 02m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4f442a482234e5db5851024e472d1ee : SUCCESS in 40m 38s","accounts_in_message":[],"_revision_number":8},{"id":"3dd65ff5baf4249cf119fabe5b2374f2a48f8ae8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 21:09:30.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2171275685941f4eea1d5d067c16d4ee1a1b2d7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 22:00:56.000000000","message":"Patch Set 8: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"cdbc9fc99c882dc04af8cf760c62f4ac1bc7da5b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-29 05:04:06.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"8f499241a3660b9342553f0630c2ae95c7eb1560","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-29 05:04:31.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":8},{"id":"470d52f7106d6e8ad498ae0c08c1d2f899212a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 07:12:39.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6f1ea5b7b4a0436b841f6adcdf352f3c : SUCCESS in 58m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/33b8fab1d0b844e7bec500a5f87e2574 : SUCCESS in 1h 05m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2d57ccab1d74d30a6df149d9fbeaa43 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f49ff64acdcc4fec9c96f9063eadd8bb : SUCCESS in 9m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcd1f3e0b62d4a48b599d3229a865185 : SUCCESS in 11m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07a975c84e76491e948525c719fb368b : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1716512398804eb4b4182ccdb81288df : SUCCESS in 12m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b5475f9c135c4b1e8a396cdb852b54db : SUCCESS in 24m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/736def8aac8f47d190a3a552c8d982a7 : SUCCESS in 27m 16s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9fa12db5374a40aca19cadf1ab710f1a : SUCCESS in 10m 33s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6cf601fb1ca945febdfda5a067c13d4b : TIMED_OUT in 2h 05m 56s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/bef6760c940d42a1b927b515b56f3f14 : FAILURE in 2h 01m 27s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/af84e75cc8804c818f687d16bb96787b : FAILURE in 1h 59m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50568c4c21064eb686fd7f2711a617ed : SUCCESS in 1h 43m 24s","accounts_in_message":[],"_revision_number":8},{"id":"08618807fd746be14d21959bd2e66fa7698476b1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-29 09:20:32.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"b11e46d6294b4eef1971f0551d7515d38f8c4851","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-29 09:20:54.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":8},{"id":"5241fa4f967e7df4668c2271d69f7ea841905df9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 11:00:04.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9a974de3ec4a4a95bc3ff8f647df94f7 : SUCCESS in 1h 26m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b55c48862dee46bf826742f996910448 : SUCCESS in 1h 20m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2066e961f9ac4815b841466da1694391 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41937426208e4d0da12a447f7beeede9 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ea4966d15eb418382427087d1bb96e3 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a28679c2b8014439a5be504c193a26d2 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcc2e156beb8427a98e00da5466ae39a : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4cfe4610cc147e7900acb3441b56445 : TIMED_OUT in 30m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/51a7575911e04cb9b9a6eaa6ed972634 : SUCCESS in 17m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2666d51e0edd4cf88447074acacd8e74 : SUCCESS in 7m 33s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e15f4e4385e44e3a9c01f1a333ffd0ff : SUCCESS in 1h 30m 19s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f0e81275414441dc8a3685a87af75bb7 : SUCCESS in 1h 37m 31s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/ced1e41198774896a394424b56bc6e7e : SUCCESS in 1h 35m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d5b178c81104e5195ab0e012e7667a8 : SUCCESS in 1h 23m 13s","accounts_in_message":[],"_revision_number":8},{"id":"aef6adca0fd8a3206cc24c5e51828f43669092c3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-29 16:48:00.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"efe061a88ea67eede06028b7355cb7e614c9555d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 18:00:21.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/69d2d7254e1949a39ee352eafe7d5af1 : SUCCESS in 1h 07m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bda87c8b476144189ac4797eb127e712 : SUCCESS in 48m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a85007e23124011bd3bded0e9b63ae1 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3e54f82c4f943e387ca6b2e5405c27a : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/221a75a1b20347bd84133d5cd3bfe80b : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8eee31390d1a447bb93b9911940cdab1 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a867000a4f1841d5ad70993660d161ef : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5d30cde3218542b6959bacf130989ad0 : SUCCESS in 6m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9f8bc78723ed4671a60a23275e2d9ec8 : SUCCESS in 6m 37s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9186292c6f7c41aea2c740d2b311db22 : SUCCESS in 4m 04s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8488638e20bb4d5d9f15e2b948254c9a : SUCCESS in 1h 00m 13s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/b67baa19f96b40a09651f8e6f08f6f25 : SUCCESS in 46m 13s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/715721b0671140978dae2c842fe8c360 : SUCCESS in 1h 10m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6c9bffc980349289a9f04760173f20d : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":8},{"id":"bf3bd8e45bf32af6a58e5baa123c040f9f3ed8a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 18:00:33.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"17eb9dd7d13e99687ea8ba5b591384a9532c6f9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 19:13:33.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0cfa1e0770ff405b8e141d81fe39e39c : SUCCESS in 54m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec225e41b3f144c193ef45d0d207a148 : SUCCESS in 1h 12m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7144097d5ee54958afb31da31e64e616 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ead1b598bd30416b9d0da406733c0702 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c830d094b73414884279550428c550d : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc652b9edb614f2ba33f5d88393c90e2 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59677b1bdc8a49dd99bb38386b949c19 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/678580695f7a484c8f4bc35c9b1b34d4 : SUCCESS in 8m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/207228c7dcdf4ee0bd24d74c3da2ba4c : SUCCESS in 10m 12s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/edf12fa8d96d4985b2db88f7705a1901 : SUCCESS in 1h 05m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4902c66a57c41e59557d90a0cb5c229 : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":8},{"id":"695fe1ce60263aa35a7b036c407245267307bc01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 19:13:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"f8a932dfceb449abffd75e176a5c3754977852a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 19:14:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f0371a283040efb2d32f438c196534 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"26f0311b29b18589ad134a34d5abd8ff2f82cf47","revisions":{"d64a99feaad9f9bba660b58952f6c335cf2d8080":{"kind":"REWORK","_number":1,"created":"2020-08-17 14:35:50.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/18/746518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/746518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/746518/1"}}},"commit":{"parents":[{"commit":"17c0b9ce86611aabce5707851471a15019fef20d","subject":"Move SynchronousAPIBase to a generalized location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17c0b9ce86611aabce5707851471a15019fef20d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 13:58:00.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 14:35:29.000000000","tz":-420},"subject":"Handle atomic image properties separately","message":"Handle atomic image properties separately\n\nThe image_update() code will clobber, revert, or update lock values\nin keys that we use as atomic properties. This adds an exclusion\nlist of properties that we handle specially and plumbs them down\nto image_update() so that they will be excluded from the\nadd/update/delete logic.\n\nChange-Id: Ib910274472346ce0c336cd1ead8370d5799d0b96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d64a99feaad9f9bba660b58952f6c335cf2d8080"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d64a99feaad9f9bba660b58952f6c335cf2d8080"}]},"branch":"refs/heads/master"},"c5e4170c95d2ec3fad3cac66fd0566b5da9bc27b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-08-18 15:04:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/18/746518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/746518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/746518/2"}}},"commit":{"parents":[{"commit":"a814b79c2edc26af4d9261ac28d647ea0a522094","subject":"Move SynchronousAPIBase to a generalized location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a814b79c2edc26af4d9261ac28d647ea0a522094"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 13:58:00.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 15:03:57.000000000","tz":-420},"subject":"Handle atomic image properties separately","message":"Handle atomic image properties separately\n\nThe image_update() code will clobber, revert, or update lock values\nin keys that we use as atomic properties. This adds an exclusion\nlist of properties that we handle specially and plumbs them down\nto image_update() so that they will be excluded from the\nadd/update/delete logic.\n\nChange-Id: Ib910274472346ce0c336cd1ead8370d5799d0b96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c5e4170c95d2ec3fad3cac66fd0566b5da9bc27b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c5e4170c95d2ec3fad3cac66fd0566b5da9bc27b"}]},"branch":"refs/heads/master"},"5132df94985d3ba15f298ec981002a7a3cf09b26":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-18 16:34:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/18/746518/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/746518/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/746518/3"}}},"commit":{"parents":[{"commit":"9f35d3bdafe8c93a01b8ccdac57f77cd266ee3ac","subject":"Move SynchronousAPIBase to a generalized location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f35d3bdafe8c93a01b8ccdac57f77cd266ee3ac"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 13:58:00.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 16:34:10.000000000","tz":-420},"subject":"Handle atomic image properties separately","message":"Handle atomic image properties separately\n\nThe image_update() code will clobber, revert, or update lock values\nin keys that we use as atomic properties. This adds an exclusion\nlist of properties that we handle specially and plumbs them down\nto image_update() so that they will be excluded from the\nadd/update/delete logic.\n\nChange-Id: Ib910274472346ce0c336cd1ead8370d5799d0b96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5132df94985d3ba15f298ec981002a7a3cf09b26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5132df94985d3ba15f298ec981002a7a3cf09b26"}]},"branch":"refs/heads/master"},"fdc431f0789d995640b11806cfeed4fb6a936647":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-20 22:30:43.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/18/746518/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/746518/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/746518/4"}}},"commit":{"parents":[{"commit":"1909406049211cdd525677f52de5c536a15b6258","subject":"Move SynchronousAPIBase to a generalized location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1909406049211cdd525677f52de5c536a15b6258"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 13:58:00.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-20 22:30:23.000000000","tz":-420},"subject":"Handle atomic image properties separately","message":"Handle atomic image properties separately\n\nThe image_update() code will clobber, revert, or update lock values\nin keys that we use as atomic properties. This adds an exclusion\nlist of properties that we handle specially and plumbs them down\nto image_update() so that they will be excluded from the\nadd/update/delete logic.\n\nChange-Id: Ib910274472346ce0c336cd1ead8370d5799d0b96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fdc431f0789d995640b11806cfeed4fb6a936647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fdc431f0789d995640b11806cfeed4fb6a936647"}]},"branch":"refs/heads/master"},"cbd708a4489c7257dccaa69e868dff7c3678c89d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-08-21 14:30:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/18/746518/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/746518/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/746518/5"}}},"commit":{"parents":[{"commit":"6d2622cd18606c030c17490d81fdca1af9667bd8","subject":"Move SynchronousAPIBase to a generalized location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d2622cd18606c030c17490d81fdca1af9667bd8"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 13:58:00.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 14:27:34.000000000","tz":-420},"subject":"Handle atomic image properties separately","message":"Handle atomic image properties separately\n\nThe image_update() code will clobber, revert, or update lock values\nin keys that we use as atomic properties. This adds an exclusion\nlist of properties that we handle specially and plumbs them down\nto image_update() so that they will be excluded from the\nadd/update/delete logic.\n\nChange-Id: Ib910274472346ce0c336cd1ead8370d5799d0b96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbd708a4489c7257dccaa69e868dff7c3678c89d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbd708a4489c7257dccaa69e868dff7c3678c89d"}]},"branch":"refs/heads/master"},"7790996982e9d672753cd14679a312284904b0af":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-08-21 15:18:01.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/18/746518/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/746518/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/746518/6"}}},"commit":{"parents":[{"commit":"5da78e8c501d5b308346b7582b66805dc8ae7cd6","subject":"Move SynchronousAPIBase to a generalized location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5da78e8c501d5b308346b7582b66805dc8ae7cd6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 13:58:00.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 15:17:55.000000000","tz":-420},"subject":"Handle atomic image properties separately","message":"Handle atomic image properties separately\n\nThe image_update() code will clobber, revert, or update lock values\nin keys that we use as atomic properties. This adds an exclusion\nlist of properties that we handle specially and plumbs them down\nto image_update() so that they will be excluded from the\nadd/update/delete logic.\n\nChange-Id: Ib910274472346ce0c336cd1ead8370d5799d0b96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7790996982e9d672753cd14679a312284904b0af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7790996982e9d672753cd14679a312284904b0af"}]},"branch":"refs/heads/master"},"9c0f0fbf845b24ed4c79e668e149f7fd62e64df5":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-08-21 19:53:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/18/746518/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/746518/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/746518/7"}}},"commit":{"parents":[{"commit":"5193c869af115a977c946853b5644070b4df15b2","subject":"Move SynchronousAPIBase to a generalized location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5193c869af115a977c946853b5644070b4df15b2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 13:58:00.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 19:52:05.000000000","tz":-420},"subject":"Handle atomic image properties separately","message":"Handle atomic image properties separately\n\nThe image_update() code will clobber, revert, or update lock values\nin keys that we use as atomic properties. This adds an exclusion\nlist of properties that we handle specially and plumbs them down\nto image_update() so that they will be excluded from the\nadd/update/delete logic.\n\nChange-Id: Ib910274472346ce0c336cd1ead8370d5799d0b96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c0f0fbf845b24ed4c79e668e149f7fd62e64df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c0f0fbf845b24ed4c79e668e149f7fd62e64df5"}]},"branch":"refs/heads/master"},"26f0311b29b18589ad134a34d5abd8ff2f82cf47":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-08-24 13:41:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/18/746518/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/746518/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/746518/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/746518/8"}}},"commit":{"parents":[{"commit":"36cbc50e7d70699d6ca5dd4fa81844567e39ddf0","subject":"Move SynchronousAPIBase to a generalized location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36cbc50e7d70699d6ca5dd4fa81844567e39ddf0"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-17 13:58:00.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-24 13:41:13.000000000","tz":-420},"subject":"Handle atomic image properties separately","message":"Handle atomic image properties separately\n\nThe image_update() code will clobber, revert, or update lock values\nin keys that we use as atomic properties. This adds an exclusion\nlist of properties that we handle specially and plumbs them down\nto image_update() so that they will be excluded from the\nadd/update/delete logic.\n\nChange-Id: Ib910274472346ce0c336cd1ead8370d5799d0b96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26f0311b29b18589ad134a34d5abd8ff2f82cf47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26f0311b29b18589ad134a34d5abd8ff2f82cf47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
