)]}'
{"id":"openstack%2Fglance~746746","triplet_id":"openstack%2Fglance~master~Iff1fbe2b55f4be12e69c9fd3dec7e3b3e2593e53","project":"openstack/glance","branch":"master","topic":"task-heartbeat","hashtags":[],"change_id":"Iff1fbe2b55f4be12e69c9fd3dec7e3b3e2593e53","subject":"Add FakeData generator test utility","status":"MERGED","created":"2020-08-18 15:04:45.000000000","updated":"2020-08-26 19:25:28.000000000","submitted":"2020-08-26 19:22:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746746-1598469739525-9749b533","meta_rev_id":"dadd0791cd1bc4be36b1b7287af8fa1f40bae48f","_number":746746,"virtual_id_number":746746,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-26 14:01:30.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"date":"2020-08-26 13:52:42.000000000","_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 19:22:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-24 13:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2020-08-25 15:23:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-08-25 15:23:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2020-08-26 19:25:28.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-18 15:15:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-21 16:28:37.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2020-08-25 15:23:29.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-08-26 14:01:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-26 19:22:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abf68195abab51bae4c929d4416f72117e3bec35","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-18 15:04:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f2c3e7b5b0b7de56c06ba75a9cfd882b28b7af6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-18 15:07:42.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"1f90a1216463af3ac34283d2d2289cb16214b9b1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-18 15:12:23.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"93957563d8ef871f3f33d1125ebf62b0d26f0ecc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-18 15:15:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d731844f9be23b9d395409919b1deb49575827d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 16:24:42.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0aa88dafdf940dfbcaa5431524febfe : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e018e3af33b942dcbc0a168818f628fa : FAILURE in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32e3d6209e1c4e04bed1337b6117b2a5 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b11542f8d3c64b6f90cde14cba6947ec : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00bdc2f4a79b4dd6b5526b51f01a17e6 : SUCCESS in 7m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c71fa12e58ea49b49274f244bd9abec6 : SUCCESS in 8m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b4839a6df4d3444383081755a2229c3f : SUCCESS in 10m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/943da3dffa544b7a931b8e0efdc890bd : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"04cf5bd0b0abe03f0b4c8ff2ad7f913d11661ffa","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-18 16:34:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb1f7b1cba3f07c68cfbd667bde5a2516beda38e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-18 16:35:06.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":2},{"id":"6423c727a788d236bbda648f0db5a37efb51615a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 16:51:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a5a2df2219546b199254e7cbe03c7c0 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/549a2f5b52184b93a56db14386ccdcc2 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5b22f680dab402a9f9eee8797c21b46 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abbbfa2242ff4595adba7a75edeba6f4 : SUCCESS in 8m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9817bf95ae4a8da558f79d15ffbb4b : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4c5a130c620340dbaba196371ef5acec : SUCCESS in 7m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/007077da000c4c7b8c44562c88258c70 : SUCCESS in 9m 04s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cea8c7fa27634926b2917e504d83e11d : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"4c4dca0e16cd1be5b6e9122b0f9aa6bf1b6f4b84","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-20 22:30:43.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"8dd8b00e348d7f2ab51f7fefc0b4608284e1e3cb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-20 22:31:22.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":3},{"id":"d868377749e218a595eacf1ed42a782e47c61126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 23:22:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff4eb3029c954cd198ae4e0f205c3302 : FAILURE in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd967a488d54325b9a5c8557a1a0e6b : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45fb1587574b4ffaa363d63953c3a91f : FAILURE in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e5bf930d99f4f4597c29e4ce196000d : FAILURE in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cfa10da135b4e6e99732261dab57df9 : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5f84800013248d19c924b2d9fb00db7 : SUCCESS in 7m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/51f1cc58ac2f4839b62bca289dd265ee : SUCCESS in 11m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4b965ab0a2f647e59a3caba912109fab : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":3},{"id":"ce0ca4d9b1b20c182d32944b2357e7d9ecc6d948","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-21 14:30:36.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8f4539ab081aaab96718abcc01936be3d2065c10","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-21 14:31:15.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":4},{"id":"25241b106b3150c687ef08c5a85e11cb676d86d9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-21 14:35:46.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"84bbd9739be9da1054d7b95bd2b3853e770bf1e9","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-21 14:57:38.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nOne actual concern inline.","accounts_in_message":[],"_revision_number":4},{"id":"059be9d759fafecce570269778b188254aad1aa0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-21 15:05:23.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"dc65eaa6c844b4be3fc3fe7b525f9138d8c5d4e9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-21 15:18:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6b5d857b5c7dae73b52d6d0288a9e10262879798","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-21 15:18:32.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":5},{"id":"de013b0797bb6831ee8ce241b3911824386fdf95","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2020-08-21 16:28:37.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ab38865914a2131c4adbb7096978faac5948a8ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 18:21:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e543af4eaa74602b66dbfa17901b1aa : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/139c4b7fdd114bb2930561e5564c240b : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0348b43d0449451bafc7c25866d9d015 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8473d6b6ed854d908ec0d7c91e71cae0 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db803969242a4a499addedadc6c90995 : SUCCESS in 9m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce70f6c32bfd4bce809d3b330140aa71 : SUCCESS in 10m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b8c8edbedb5241e99f937420f2aeea7d : SUCCESS in 10m 47s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/acda144d463f4188b5e074ce88572079 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":5},{"id":"99ac986a6a62d1a1a2c1102a38b77a7103ff3b78","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-21 19:53:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"26ca759eabdd711d3fe0274d69dcdd99f9a549d5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-21 19:54:40.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":6},{"id":"1533ac357fdcdb9f1921ff1e88bc326264a009bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 21:42:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dc8b052a721438089963693a135864d : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de002dc07f14c3187e977e8407cc612 : FAILURE in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c64d9423c3e41fc8ed1012d7b456b8b : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bd7f5d4611845b692a47487de50cae4 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c2669a87734cb784c80d164653d02c : SUCCESS in 8m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6c6dce4f1a1743b98db17d2bc60fb02d : SUCCESS in 7m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f8bdf0da6c124614a70540fadb33a7b6 : SUCCESS in 9m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/313b90080c144a6e9544f354a387cc19 : SUCCESS in 5m 26s\n\nWarning:\n  Comments left for invalid file glance/tests/functional/v2/test_images.py","accounts_in_message":[],"_revision_number":6},{"id":"11cd3850846b21445e77a91e6acc401676b7c1db","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-24 13:41:26.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"c44774661d3601bfa13eda39f5766bfad5dd20d3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-24 13:42:22.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"5044623f5d9300227e669ace49ccf1c74cd55741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 15:50:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/622b768487974ef8af5bfcaeeb507791 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dfd281fb7a84a71b9872b68a0ebadc6 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/752b9820ec364460a2db85bd97003034 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f938002a136045ceb16c9d2618c8c5cc : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d16ea26a0448409196b8c7bad97d8c : SUCCESS in 6m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aadfc9dc0270482ea00e1c1d8895fedb : SUCCESS in 8m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ab4e3afa8479435bb0e9f6eae24b6e68 : SUCCESS in 9m 41s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0f55a3d3afd44bf7b653f1bd59c4ada9 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":7},{"id":"1a76b562d58a60174ad902b556ba8b572f7cb3ed","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-25 15:23:29.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"4cee896bf1bc900a8314505bdb3d44fc17e611c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 15:23:58.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ee55e7d48eaaf3693e2e6c509ab79983a043e7f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 16:36:52.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a44a8d213e049b284e2fac0731dc19d : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37e86b19ce704f92947246ef1f9b0909 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e8549e5d4e64ab7be5e6760787bb1c4 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dde7b4df9464bc0a71a81e23bb10c02 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b55549097d254796b1816f305c263169 : SUCCESS in 12m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5551a9ea2692432da864d21c31ec292a : FAILURE in 22m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a1076ef416374d9ab6f8219da7291004 : SUCCESS in 11m 04s","accounts_in_message":[],"_revision_number":7},{"id":"a834fecaa951cac74649f18cc772982dd133d66c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-08-26 07:00:32.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"3505c61d4f3be97239c6edc992295af358ae4b59","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-26 07:00:58.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"5b2c5aa60c6e2b758c1aa91078393dce1f36be24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 09:11:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41c299d4143a49bd8327e24fe606a356 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72e87e6fa6aa41bfba2faf529cc914ac : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71d4a734b9d44ef9b49af256e36e3773 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f8e21182c9545068b479b381b75eeb3 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7285acec19da485f8830b6992e54160a : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/82ab0ee3ee4a4520be02ec33936db499 : FAILURE in 17m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/65d2a5fee4504575ad28cee9f2a33e9e : SUCCESS in 8m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7778b37ec73e47968f38a2ddd6cdeb20 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":7},{"id":"22e834783d8896111d4873e8674053b62c3fb4df","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-08-26 13:52:42.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"4db33728cd3bb68123b09c0ff63c4d22cda26e3a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-26 14:01:30.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":7},{"id":"fc5b46bfb02a40b78a61303c40a7b2cca135dd55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:55:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69553cdcff824ff686160f1f28f398be : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc874b53480a43739aac64248baf168b : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/018b6269ce924373a81ed3c38e5b85d2 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/313902966aad465db011224bb6453429 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/874e09322a084a7fb307f582bd01f2d3 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b1b9606e1054eaf8000fec2209074ca : SUCCESS in 9m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4b236054bd5e469db7a28fe8d7bfb1c6 : SUCCESS in 9m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2f89601e208440b0b307b69003afcdf2 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":7},{"id":"f65ff4a775003471bbc8195f5c2f76168c603675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:56:42.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f9023a6cd640b42d4d15e81f6729fa12b871fda7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:22:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"e7e4ba77208bbbd99b73a9c68ffb73c5c2418b29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:22:19.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dbbfe8b0027486d9587b165e2e3e0b0 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfeeea019ba84276b93897f1a7ce3e22 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e7092fffc7e4ccf9a6327735922b027 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ef806c8a95a4db197ce92a286280f70 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b832283477453eb5ba54defc4451a2 : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bba87ce7cf824908bc36016ba455d121 : SUCCESS in 7m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f27596abff364c5eaf412e9ed91791ec : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":7},{"id":"318819dd9101561d28e00c69a0b5d8c5bb6c0ef7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:25:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/069a83084b46404fbb3a307b8139364c : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"dc08127f059a227928400bf2b629ef503dbf28e7","revisions":{"15ed9eb9a5fba0bc73f83a4db916534d5777ef9b":{"kind":"REWORK","_number":1,"created":"2020-08-18 15:04:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/46/746746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/746746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/746746/1"}}},"commit":{"parents":[{"commit":"a309fdcd8ac8973a43dd94d8703006a29dcfba39","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a309fdcd8ac8973a43dd94d8703006a29dcfba39"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 14:51:43.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 15:00:17.000000000","tz":-420},"subject":"Add FakeData generator test utility","message":"Add FakeData generator test utility\n\nThis adds a FakeData file-like generator that is able to generate\narbitrary amounts of data when we need to simulate reading from a\nfile or request pipeline, without having to store all of that in\nmemory.\n\nChange-Id: Iff1fbe2b55f4be12e69c9fd3dec7e3b3e2593e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/15ed9eb9a5fba0bc73f83a4db916534d5777ef9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/15ed9eb9a5fba0bc73f83a4db916534d5777ef9b"}]},"branch":"refs/heads/master"},"c4ded9f657453870dc9e392dbe175d4ab7a0c71a":{"kind":"REWORK","_number":2,"created":"2020-08-18 16:34:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/46/746746/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/746746/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/746746/2"}}},"commit":{"parents":[{"commit":"a309fdcd8ac8973a43dd94d8703006a29dcfba39","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a309fdcd8ac8973a43dd94d8703006a29dcfba39"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 14:51:43.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 16:32:05.000000000","tz":-420},"subject":"Add FakeData generator test utility","message":"Add FakeData generator test utility\n\nThis adds a FakeData file-like generator that is able to generate\narbitrary amounts of data when we need to simulate reading from a\nfile or request pipeline, without having to store all of that in\nmemory.\n\nChange-Id: Iff1fbe2b55f4be12e69c9fd3dec7e3b3e2593e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4ded9f657453870dc9e392dbe175d4ab7a0c71a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4ded9f657453870dc9e392dbe175d4ab7a0c71a"}]},"branch":"refs/heads/master"},"5658a1e3859201d3c3735b1ab39f077806f3329d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-20 22:30:43.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/46/746746/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/746746/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/746746/3"}}},"commit":{"parents":[{"commit":"8a717ae18c71b90ccf2bbe4ba61b28afcc0ce352","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a717ae18c71b90ccf2bbe4ba61b28afcc0ce352"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 14:51:43.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-20 22:30:23.000000000","tz":-420},"subject":"Add FakeData generator test utility","message":"Add FakeData generator test utility\n\nThis adds a FakeData file-like generator that is able to generate\narbitrary amounts of data when we need to simulate reading from a\nfile or request pipeline, without having to store all of that in\nmemory.\n\nChange-Id: Iff1fbe2b55f4be12e69c9fd3dec7e3b3e2593e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5658a1e3859201d3c3735b1ab39f077806f3329d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5658a1e3859201d3c3735b1ab39f077806f3329d"}]},"branch":"refs/heads/master"},"3e8df230a5ea8970c73b390d93824d504cbb788c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-21 14:30:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/46/746746/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/746746/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/746746/4"}}},"commit":{"parents":[{"commit":"123cd9a1475226eb6b0301ef1e7237b9e526da89","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/123cd9a1475226eb6b0301ef1e7237b9e526da89"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 14:51:43.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 14:27:34.000000000","tz":-420},"subject":"Add FakeData generator test utility","message":"Add FakeData generator test utility\n\nThis adds a FakeData file-like generator that is able to generate\narbitrary amounts of data when we need to simulate reading from a\nfile or request pipeline, without having to store all of that in\nmemory.\n\nChange-Id: Iff1fbe2b55f4be12e69c9fd3dec7e3b3e2593e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e8df230a5ea8970c73b390d93824d504cbb788c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e8df230a5ea8970c73b390d93824d504cbb788c"}]},"branch":"refs/heads/master"},"0593907da56ff36fca00feffce7579fc2f786fa3":{"kind":"REWORK","_number":5,"created":"2020-08-21 15:18:01.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/46/746746/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/746746/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/746746/5"}}},"commit":{"parents":[{"commit":"123cd9a1475226eb6b0301ef1e7237b9e526da89","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/123cd9a1475226eb6b0301ef1e7237b9e526da89"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 14:51:43.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 15:17:45.000000000","tz":-420},"subject":"Add FakeData generator test utility","message":"Add FakeData generator test utility\n\nThis adds a FakeData file-like generator that is able to generate\narbitrary amounts of data when we need to simulate reading from a\nfile or request pipeline, without having to store all of that in\nmemory.\n\nChange-Id: Iff1fbe2b55f4be12e69c9fd3dec7e3b3e2593e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0593907da56ff36fca00feffce7579fc2f786fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0593907da56ff36fca00feffce7579fc2f786fa3"}]},"branch":"refs/heads/master"},"3d6fe925a767de904cf975e0ff5bdf25a65aa2d4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-08-21 19:53:47.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/46/746746/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/746746/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/746746/6"}}},"commit":{"parents":[{"commit":"72c713744969865c35d2a14695189e784f566055","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/72c713744969865c35d2a14695189e784f566055"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 14:51:43.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-21 19:52:05.000000000","tz":-420},"subject":"Add FakeData generator test utility","message":"Add FakeData generator test utility\n\nThis adds a FakeData file-like generator that is able to generate\narbitrary amounts of data when we need to simulate reading from a\nfile or request pipeline, without having to store all of that in\nmemory.\n\nChange-Id: Iff1fbe2b55f4be12e69c9fd3dec7e3b3e2593e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d6fe925a767de904cf975e0ff5bdf25a65aa2d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d6fe925a767de904cf975e0ff5bdf25a65aa2d4"}]},"branch":"refs/heads/master"},"dc08127f059a227928400bf2b629ef503dbf28e7":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-08-24 13:41:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/46/746746/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/746746/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/746746/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/746746/7"}}},"commit":{"parents":[{"commit":"3f6e349d0853a9746d0d744bc3eb0b2baa1ddff9","subject":"Implement time-limited import locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f6e349d0853a9746d0d744bc3eb0b2baa1ddff9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-18 14:51:43.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2020-08-24 13:41:13.000000000","tz":-420},"subject":"Add FakeData generator test utility","message":"Add FakeData generator test utility\n\nThis adds a FakeData file-like generator that is able to generate\narbitrary amounts of data when we need to simulate reading from a\nfile or request pipeline, without having to store all of that in\nmemory.\n\nChange-Id: Iff1fbe2b55f4be12e69c9fd3dec7e3b3e2593e53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc08127f059a227928400bf2b629ef503dbf28e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc08127f059a227928400bf2b629ef503dbf28e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
